)]}'
{"id":"openstack%2Fpuppet-neutron~666406","triplet_id":"openstack%2Fpuppet-neutron~stable%2Frocky~I8248ad80e6ff88aec252919fa687018bda78a60b","project":"openstack/puppet-neutron","branch":"stable/rocky","topic":"bug/1832251","hashtags":[],"change_id":"I8248ad80e6ff88aec252919fa687018bda78a60b","subject":"Ensure tunnel types are absent if not configured","status":"MERGED","created":"2019-06-19 17:10:08.000000000","updated":"2019-07-25 16:59:47.000000000","submitted":"2019-07-25 16:59:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"666406-1564073987885-62a04a63","meta_rev_id":"84651544413a86126cc5257a27ac46b33a1ae404","_number":666406,"virtual_id_number":666406,"owner":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-07-25 16:59:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2019-07-25 03:49:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-06-24 21:56:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2019-07-25 03:49:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-24 21:56:22.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-07-25 03:49:49.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2019-07-25 16:59:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"31d2ffdfbd96c7686e120f65c8c0de4219d4d7b6","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-06-19 17:10:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5a17552ac2a3ad7e80c69fce76d7693730322ac0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 18:53:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint http://logs.openstack.org/06/666406/1/check/puppet-openstack-lint/43cb427/ : SUCCESS in 4m 37s\n- puppet-openstack-syntax-4 http://logs.openstack.org/06/666406/1/check/puppet-openstack-syntax-4/b007165/ : SUCCESS in 5m 17s\n- puppet-openstack-syntax-5 http://logs.openstack.org/06/666406/1/check/puppet-openstack-syntax-5/5dc6ca9/ : SUCCESS in 5m 28s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/06/666406/1/check/puppet-openstack-unit-4.8-centos-7/c27d9e0/ : SUCCESS in 26m 03s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/9fbea29/ : SUCCESS in 55m 34s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/db4bef5/ : SUCCESS in 1h 00m 56s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/e706dcf/ : SUCCESS in 59m 45s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic/795006f/ : SUCCESS in 51m 16s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/5426d6c/ : SUCCESS in 51m 13s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/43f2b2c/ : FAILURE in 4m 47s\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/cea166f/ : FAILURE in 4m 12s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/61e0f1b/ : FAILURE in 4m 44s (non-voting)\n- puppet-openstack-integration-4-scenario001-tempest-debian-stable-luminous http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-4-scenario001-tempest-debian-stable-luminous/45c5c87/ : FAILURE in 4m 31s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-debian-stable http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-4-scenario002-tempest-debian-stable/83b87dc/ : FAILURE in 3m 48s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-debian-stable http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-4-scenario003-tempest-debian-stable/d288462/ : FAILURE in 4m 00s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-debian-stable-luminous http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-4-scenario004-tempest-debian-stable-luminous/164d6f1/ : FAILURE in 3m 47s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/06/666406/1/check/puppet-openstack-beaker-centos-7/fc99bca/ : SUCCESS in 16m 48s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/06/666406/1/check/puppet-openstack-beaker-ubuntu-bionic/4b2acca/ : SUCCESS in 14m 05s\n- tripleo-puppet-ci-centos-7-undercloud-oooq http://logs.openstack.org/06/666406/1/check/tripleo-puppet-ci-centos-7-undercloud-oooq/014d763/ : SUCCESS in 1h 05m 48s\n- tripleo-puppet-ci-centos-7-undercloud-containers http://logs.openstack.org/06/666406/1/check/tripleo-puppet-ci-centos-7-undercloud-containers/95ec3e9/ : SUCCESS in 1h 28m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"40b7bef9457bef4db088f46ca1d32996346a253f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-24 21:56:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1d882650b666ef1797b471c4ec58ed6cd3878037","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-24 21:56:22.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5c7643e32ba4ead3905e406f2010cf5c22557050","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 02:35:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint http://logs.openstack.org/06/666406/1/check/puppet-openstack-lint/53dbd1f/ : RETRY_LIMIT in 1m 47s\n- puppet-openstack-syntax-4 http://logs.openstack.org/06/666406/1/check/puppet-openstack-syntax-4/0d5f568/ : RETRY_LIMIT in 1m 48s\n- puppet-openstack-syntax-5 http://logs.openstack.org/06/666406/1/check/puppet-openstack-syntax-5/cc5c2c6/ : RETRY_LIMIT in 1m 46s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/06/666406/1/check/puppet-openstack-unit-4.8-centos-7/deb5b0d/ : RETRY_LIMIT in 5m 10s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/2a603f7/ : RETRY_LIMIT in 4m 14s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/b8bdff9/ : RETRY_LIMIT in 4m 37s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/846ecb0/ : RETRY_LIMIT in 4m 05s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic/7343e4a/ : RETRY_LIMIT in 3m 53s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/0e71e14/ : RETRY_LIMIT in 3m 26s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/241c40d/ : RETRY_LIMIT in 3m 34s\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/5198e19/ : RETRY_LIMIT in 3m 36s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/1d867f9/ : RETRY_LIMIT in 4m 58s (non-voting)\n- puppet-openstack-integration-4-scenario001-tempest-debian-stable-luminous http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-4-scenario001-tempest-debian-stable-luminous/0d3700a/ : RETRY_LIMIT in 4m 43s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-debian-stable http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-4-scenario002-tempest-debian-stable/963ee63/ : RETRY_LIMIT in 3m 49s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-debian-stable http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-4-scenario003-tempest-debian-stable/f08f77a/ : RETRY_LIMIT in 3m 36s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-debian-stable-luminous http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-4-scenario004-tempest-debian-stable-luminous/64648e5/ : RETRY_LIMIT in 2m 39s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/06/666406/1/check/puppet-openstack-beaker-centos-7/ff28885/ : RETRY_LIMIT in 5m 16s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/06/666406/1/check/puppet-openstack-beaker-ubuntu-bionic/2bf98e0/ : RETRY_LIMIT in 5m 01s\n- tripleo-puppet-ci-centos-7-undercloud-oooq http://logs.openstack.org/06/666406/1/check/tripleo-puppet-ci-centos-7-undercloud-oooq/a3a0db2/ : SUCCESS in 1h 05m 07s\n- tripleo-puppet-ci-centos-7-undercloud-containers http://logs.openstack.org/06/666406/1/check/tripleo-puppet-ci-centos-7-undercloud-containers/f821ce2/ : RETRY_LIMIT in 1h 49m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a4b134baa3fa516f0598970eaa80c0f3119cd496","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-25 13:59:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0ade8c22893af380b2034093a3f73201da73a028","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 19:00:00.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint http://logs.openstack.org/06/666406/1/check/puppet-openstack-lint/9f6cff0/ : RETRY_LIMIT in 1m 33s\n- puppet-openstack-syntax-4 http://logs.openstack.org/06/666406/1/check/puppet-openstack-syntax-4/d38c655/ : RETRY_LIMIT in 1m 39s\n- puppet-openstack-syntax-5 http://logs.openstack.org/06/666406/1/check/puppet-openstack-syntax-5/0873dea/ : RETRY_LIMIT in 1m 32s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/06/666406/1/check/puppet-openstack-unit-4.8-centos-7/996a47c/ : RETRY_LIMIT in 5m 31s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/19ebf8c/ : RETRY_LIMIT in 6m 37s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/112da21/ : RETRY_LIMIT in 4m 43s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/bcc9bbb/ : RETRY_LIMIT in 5m 24s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic/6e8e9e7/ : RETRY_LIMIT in 6m 28s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/b78fc1c/ : RETRY_LIMIT in 6m 33s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/12db07e/ : RETRY_LIMIT in 3m 32s\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/a0fe00c/ : RETRY_LIMIT in 3m 47s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/1298ff6/ : RETRY_LIMIT in 3m 39s (non-voting)\n- puppet-openstack-integration-4-scenario001-tempest-debian-stable-luminous http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-4-scenario001-tempest-debian-stable-luminous/c7dfc28/ : RETRY_LIMIT in 3m 32s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-debian-stable http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-4-scenario002-tempest-debian-stable/daf4ca5/ : RETRY_LIMIT in 4m 34s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-debian-stable http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-4-scenario003-tempest-debian-stable/b4c24f8/ : RETRY_LIMIT in 3m 34s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-debian-stable-luminous http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-4-scenario004-tempest-debian-stable-luminous/c6adda0/ : RETRY_LIMIT in 4m 14s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/06/666406/1/check/puppet-openstack-beaker-centos-7/c227737/ : RETRY_LIMIT in 4m 57s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/06/666406/1/check/puppet-openstack-beaker-ubuntu-bionic/06ae3e5/ : RETRY_LIMIT in 3m 25s\n- tripleo-puppet-ci-centos-7-undercloud-oooq http://logs.openstack.org/06/666406/1/check/tripleo-puppet-ci-centos-7-undercloud-oooq/5382c99/ : SUCCESS in 1h 04m 23s\n- tripleo-puppet-ci-centos-7-undercloud-containers http://logs.openstack.org/06/666406/1/check/tripleo-puppet-ci-centos-7-undercloud-containers/8b27424/ : SUCCESS in 1h 34m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3ce2287fe85a333559b782725e8b6df2b02897e5","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-07-25 01:05:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b8aed136c6eb4b38d1641e3e9f3bbd3096135d80","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 02:42:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint http://logs.openstack.org/06/666406/1/check/puppet-openstack-lint/baef384/ : SUCCESS in 4m 12s\n- puppet-openstack-syntax-4 http://logs.openstack.org/06/666406/1/check/puppet-openstack-syntax-4/b3b8f2a/ : SUCCESS in 4m 25s\n- puppet-openstack-syntax-5 http://logs.openstack.org/06/666406/1/check/puppet-openstack-syntax-5/2b0a2b0/ : SUCCESS in 4m 31s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/06/666406/1/check/puppet-openstack-unit-4.8-centos-7/2bc2c5f/ : SUCCESS in 30m 26s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/c49e75f/ : SUCCESS in 1h 02m 41s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/9486be3/ : SUCCESS in 1h 01m 12s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/5cad8ec/ : SUCCESS in 53m 16s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic/b183015/ : SUCCESS in 46m 34s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/28a48ac/ : FAILURE in 4m 32s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/8438361/ : TIMED_OUT in 1h 30m 50s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/83f05e9/ : FAILURE in 4m 49s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/2cfb92a/ : TIMED_OUT in 1h 30m 46s (non-voting)\n- puppet-openstack-integration-4-scenario001-tempest-debian-stable-luminous http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-4-scenario001-tempest-debian-stable-luminous/58a0ab1/ : FAILURE in 4m 43s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-debian-stable http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-4-scenario002-tempest-debian-stable/b4da114/ : FAILURE in 46m 24s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-debian-stable http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-4-scenario003-tempest-debian-stable/10816cc/ : FAILURE in 4m 53s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-debian-stable-luminous http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-4-scenario004-tempest-debian-stable-luminous/558474c/ : TIMED_OUT in 1h 30m 51s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/06/666406/1/check/puppet-openstack-beaker-centos-7/c9200a4/ : SUCCESS in 16m 39s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/06/666406/1/check/puppet-openstack-beaker-ubuntu-bionic/21530ef/ : SUCCESS in 14m 16s\n- tripleo-puppet-ci-centos-7-undercloud-oooq http://logs.openstack.org/06/666406/1/check/tripleo-puppet-ci-centos-7-undercloud-oooq/209eb18/ : SUCCESS in 1h 09m 57s\n- tripleo-puppet-ci-centos-7-undercloud-containers http://logs.openstack.org/06/666406/1/check/tripleo-puppet-ci-centos-7-undercloud-containers/b510a0e/ : SUCCESS in 1h 25m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e5fc15e655c18d308e249888b71cf99565b07a2d","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-07-25 03:49:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8f435103b1df82dde9604c2108c73509136a262a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 03:50:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"690c7b355ac6da6ce7d9b28076aac3690a53dac2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 05:27:34.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint http://logs.openstack.org/06/666406/1/gate/puppet-openstack-lint/725482e/ : SUCCESS in 7m 05s\n- puppet-openstack-syntax-4 http://logs.openstack.org/06/666406/1/gate/puppet-openstack-syntax-4/c3fe3b4/ : SUCCESS in 6m 58s\n- puppet-openstack-syntax-5 http://logs.openstack.org/06/666406/1/gate/puppet-openstack-syntax-5/d15c45d/ : SUCCESS in 5m 52s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/06/666406/1/gate/puppet-openstack-unit-4.8-centos-7/e9f5e13/ : SUCCESS in 25m 53s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/06/666406/1/gate/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/b184dc3/ : SUCCESS in 1h 01m 51s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/06/666406/1/gate/puppet-openstack-integration-5-scenario002-tempest-centos-7/24ac01e/ : SUCCESS in 1h 05m 30s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/06/666406/1/gate/puppet-openstack-integration-5-scenario003-tempest-centos-7/cef1509/ : SUCCESS in 58m 56s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic http://logs.openstack.org/06/666406/1/gate/puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic/526c7c0/ : SUCCESS in 56m 21s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/06/666406/1/gate/puppet-openstack-beaker-centos-7/896ff0b/ : SUCCESS in 25m 38s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/06/666406/1/gate/puppet-openstack-beaker-ubuntu-bionic/1553768/ : SUCCESS in 23m 20s\n- tripleo-puppet-ci-centos-7-undercloud-oooq http://logs.openstack.org/06/666406/1/gate/tripleo-puppet-ci-centos-7-undercloud-oooq/0a16ad7/ : FAILURE in 52m 31s\n- tripleo-puppet-ci-centos-7-undercloud-containers http://logs.openstack.org/06/666406/1/gate/tripleo-puppet-ci-centos-7-undercloud-containers/c783798/ : SUCCESS in 1h 34m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1785beebd74d6921a8a97297baf8e41c98daaf49","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-25 13:22:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d20a31a9f257639e73387aaf1dd2145fb29e2988","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 15:06:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint http://logs.openstack.org/06/666406/1/check/puppet-openstack-lint/ffa8833/ : SUCCESS in 5m 23s\n- puppet-openstack-syntax-4 http://logs.openstack.org/06/666406/1/check/puppet-openstack-syntax-4/ecf58c0/ : SUCCESS in 4m 30s\n- puppet-openstack-syntax-5 http://logs.openstack.org/06/666406/1/check/puppet-openstack-syntax-5/afa1639/ : SUCCESS in 6m 52s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/06/666406/1/check/puppet-openstack-unit-4.8-centos-7/a9f330e/ : SUCCESS in 29m 51s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/087d0ae/ : SUCCESS in 57m 38s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/afb0645/ : SUCCESS in 59m 30s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/cb33363/ : SUCCESS in 1h 06m 52s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic/ab30add/ : SUCCESS in 54m 40s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/15bb073/ : TIMED_OUT in 1h 31m 33s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/362accd/ : FAILURE in 5m 51s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/fdacac6/ : TIMED_OUT in 1h 30m 55s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/d073bc6/ : TIMED_OUT in 1h 30m 47s (non-voting)\n- puppet-openstack-integration-4-scenario001-tempest-debian-stable-luminous http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-4-scenario001-tempest-debian-stable-luminous/97d0583/ : TIMED_OUT in 1h 30m 46s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-debian-stable http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-4-scenario002-tempest-debian-stable/6727e7e/ : FAILURE in 6m 28s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-debian-stable http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-4-scenario003-tempest-debian-stable/81dc6fc/ : TIMED_OUT in 1h 31m 00s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-debian-stable-luminous http://logs.openstack.org/06/666406/1/check/puppet-openstack-integration-4-scenario004-tempest-debian-stable-luminous/c563351/ : TIMED_OUT in 1h 31m 31s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/06/666406/1/check/puppet-openstack-beaker-centos-7/2da5f43/ : SUCCESS in 21m 32s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/06/666406/1/check/puppet-openstack-beaker-ubuntu-bionic/af601b3/ : SUCCESS in 16m 01s\n- tripleo-puppet-ci-centos-7-undercloud-oooq http://logs.openstack.org/06/666406/1/check/tripleo-puppet-ci-centos-7-undercloud-oooq/87ffc37/ : SUCCESS in 1h 09m 48s\n- tripleo-puppet-ci-centos-7-undercloud-containers http://logs.openstack.org/06/666406/1/check/tripleo-puppet-ci-centos-7-undercloud-containers/801fd1b/ : SUCCESS in 1h 38m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"730dcc505a3930cd076b561a0af774661982472c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 15:06:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1788329c535739acca7b629602996e7ab457e37b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 16:59:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint http://logs.openstack.org/06/666406/1/gate/puppet-openstack-lint/49c3a22/ : SUCCESS in 5m 33s\n- puppet-openstack-syntax-4 http://logs.openstack.org/06/666406/1/gate/puppet-openstack-syntax-4/fd0fcfb/ : SUCCESS in 6m 55s\n- puppet-openstack-syntax-5 http://logs.openstack.org/06/666406/1/gate/puppet-openstack-syntax-5/ca5ac01/ : SUCCESS in 6m 14s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/06/666406/1/gate/puppet-openstack-unit-4.8-centos-7/9102ba6/ : SUCCESS in 31m 15s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/06/666406/1/gate/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/ea237a4/ : SUCCESS in 56m 50s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/06/666406/1/gate/puppet-openstack-integration-5-scenario002-tempest-centos-7/e6cb2ad/ : SUCCESS in 1h 09m 04s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/06/666406/1/gate/puppet-openstack-integration-5-scenario003-tempest-centos-7/3edd90e/ : SUCCESS in 1h 08m 33s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic http://logs.openstack.org/06/666406/1/gate/puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic/1dd94f0/ : SUCCESS in 53m 40s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/06/666406/1/gate/puppet-openstack-beaker-centos-7/b361836/ : SUCCESS in 28m 19s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/06/666406/1/gate/puppet-openstack-beaker-ubuntu-bionic/0241483/ : SUCCESS in 15m 55s\n- tripleo-puppet-ci-centos-7-undercloud-oooq http://logs.openstack.org/06/666406/1/gate/tripleo-puppet-ci-centos-7-undercloud-oooq/dd0872c/ : SUCCESS in 1h 11m 01s\n- tripleo-puppet-ci-centos-7-undercloud-containers http://logs.openstack.org/06/666406/1/gate/tripleo-puppet-ci-centos-7-undercloud-containers/32d24a6/ : SUCCESS in 1h 46m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"61989b7270bdcbf9ef79dc4151f2aa7303c8f7c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 16:59:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f55b869b508494328b3cbee39c4a2451d912f170","revisions":{"f55b869b508494328b3cbee39c4a2451d912f170":{"kind":"REWORK","_number":1,"created":"2019-06-19 17:10:08.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/06/666406/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/06/666406/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/06/666406/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/06/666406/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/06/666406/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/06/666406/1"}}},"commit":{"parents":[{"commit":"594ebd3c5634ea2f8b5a218650154a6c7a000747","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/594ebd3c5634ea2f8b5a218650154a6c7a000747"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-06-10 18:24:23.000000000","tz":-150},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-06-19 17:09:50.000000000","tz":-150},"subject":"Ensure tunnel types are absent if not configured","message":"Ensure tunnel types are absent if not configured\n\nThis patch fixes a bug where updating a configuration wouldn\u0027t remove\nagent/tunnel_types configuration when required.\n\nConflicts:\n  spec/classes/neutron_agents_ml2_ovs_spec.rb\n\nChange-Id: I8248ad80e6ff88aec252919fa687018bda78a60b\nCloses-Bug: #1832251\n(cherry picked from commit 6c34ce2a42ae8c445a1fbe14ab18d4ee59e09c9b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/f55b869b508494328b3cbee39c4a2451d912f170"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/f55b869b508494328b3cbee39c4a2451d912f170"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
