)]}'
{"id":"openstack%2Fpuppet-neutron~666407","triplet_id":"openstack%2Fpuppet-neutron~stable%2Fqueens~I8248ad80e6ff88aec252919fa687018bda78a60b","project":"openstack/puppet-neutron","branch":"stable/queens","topic":"bug/1832251","hashtags":[],"change_id":"I8248ad80e6ff88aec252919fa687018bda78a60b","subject":"Ensure tunnel types are absent if not configured","status":"MERGED","created":"2019-06-19 17:11:38.000000000","updated":"2019-07-09 16:03:24.000000000","submitted":"2019-07-09 16:03:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"666407-1562688204525-eaa21e4c","meta_rev_id":"ebeb34849a7fda1cd3d6224297646650300e6129","_number":666407,"virtual_id_number":666407,"owner":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-07-09 16:03:24.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-06-20 15:48:33.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-07-08 19:23:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-07-08 19:23:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-20 15:48:33.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-07-08 19:23:26.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-07-09 16:03:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"224a8f90f71509002b81d63658f3475cfc7c7bd1","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-06-19 17:11:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"898ab17be2a8505c212bf7506a1b24d596ffcec5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 18:46:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint http://logs.openstack.org/07/666407/1/check/puppet-openstack-lint/0b29317/ : SUCCESS in 4m 38s\n- puppet-openstack-syntax-4 http://logs.openstack.org/07/666407/1/check/puppet-openstack-syntax-4/348b947/ : SUCCESS in 4m 44s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/07/666407/1/check/puppet-openstack-unit-4.8-centos-7/2d5ff3c/ : SUCCESS in 43m 55s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/07/666407/1/check/puppet-openstack-unit-5.5-centos-7/e17ad2a/ : SUCCESS in 21m 44s\n- puppet-openstack-integration-4-scenario001-tempest-centos-7-luminous http://logs.openstack.org/07/666407/1/check/puppet-openstack-integration-4-scenario001-tempest-centos-7-luminous/3fd752a/ : SUCCESS in 57m 57s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/07/666407/1/check/puppet-openstack-integration-4-scenario002-tempest-centos-7/479a07e/ : SUCCESS in 1h 03m 40s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/07/666407/1/check/puppet-openstack-integration-4-scenario003-tempest-centos-7/08ebe7c/ : SUCCESS in 1h 06m 37s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7-luminous http://logs.openstack.org/07/666407/1/check/puppet-openstack-integration-4-scenario004-tempest-centos-7-luminous/f46691d/ : SUCCESS in 47m 57s\n- puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial-luminous http://logs.openstack.org/07/666407/1/check/puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial-luminous/b8c702f/ : SUCCESS in 43m 55s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/07/666407/1/check/puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/a6a2e08/ : SUCCESS in 40m 37s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/07/666407/1/check/puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/f7fc56c/ : SUCCESS in 53m 19s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial-luminous http://logs.openstack.org/07/666407/1/check/puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial-luminous/87a8088/ : SUCCESS in 50m 18s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/07/666407/1/check/puppet-openstack-beaker-centos-7/8e476db/ : SUCCESS in 15m 23s\n- puppet-openstack-beaker-xenial http://logs.openstack.org/07/666407/1/check/puppet-openstack-beaker-xenial/82d69b3/ : SUCCESS in 16m 54s","accounts_in_message":[],"_revision_number":1},{"id":"5b1890ab47a4d27ccdbb1029721481c46f8a0909","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-20 15:48:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1b178445208313b1392927fdc03e2f1310a316fc","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-08 19:23:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d489e7b34625db5c118af2e4125b7719d3b6eb76","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 19:23:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"68920b665e0bb351c4cfc49a59994aacb5044f4b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 20:38:47.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint http://logs.openstack.org/07/666407/1/gate/puppet-openstack-lint/5f5c0d1/ : SUCCESS in 6m 23s\n- puppet-openstack-syntax-4 http://logs.openstack.org/07/666407/1/gate/puppet-openstack-syntax-4/f060bc4/ : SUCCESS in 7m 19s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/07/666407/1/gate/puppet-openstack-unit-4.8-centos-7/87ab707/ : SUCCESS in 30m 55s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/07/666407/1/gate/puppet-openstack-unit-5.5-centos-7/b91bf00/ : SUCCESS in 23m 33s\n- puppet-openstack-integration-4-scenario001-tempest-centos-7-luminous http://logs.openstack.org/07/666407/1/gate/puppet-openstack-integration-4-scenario001-tempest-centos-7-luminous/280aeb6/ : FAILURE in 26m 47s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/07/666407/1/gate/puppet-openstack-integration-4-scenario002-tempest-centos-7/8d6e12d/ : SUCCESS in 1h 01m 38s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/07/666407/1/gate/puppet-openstack-integration-4-scenario003-tempest-centos-7/472f260/ : SUCCESS in 1h 08m 47s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7-luminous http://logs.openstack.org/07/666407/1/gate/puppet-openstack-integration-4-scenario004-tempest-centos-7-luminous/08a24e2/ : SUCCESS in 52m 29s\n- puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial-luminous http://logs.openstack.org/07/666407/1/gate/puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial-luminous/7a2b48a/ : FAILURE in 5m 19s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/07/666407/1/gate/puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/9f27463/ : FAILURE in 5m 44s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/07/666407/1/gate/puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/a31954a/ : FAILURE in 6m 10s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial-luminous http://logs.openstack.org/07/666407/1/gate/puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial-luminous/e56675d/ : FAILURE in 5m 38s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/07/666407/1/gate/puppet-openstack-beaker-centos-7/b7182c4/ : SUCCESS in 24m 53s\n- puppet-openstack-beaker-xenial http://logs.openstack.org/07/666407/1/gate/puppet-openstack-beaker-xenial/bb9320e/ : SUCCESS in 17m 41s","accounts_in_message":[],"_revision_number":1},{"id":"b36abf6c1b5e8cd489240a8af8aa8e2b9860e549","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-08 20:54:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"41ff5d8700101cce07e5bba757e1092762624479","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 22:07:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint http://logs.openstack.org/07/666407/1/check/puppet-openstack-lint/2627e82/ : SUCCESS in 6m 40s\n- puppet-openstack-syntax-4 http://logs.openstack.org/07/666407/1/check/puppet-openstack-syntax-4/a08d299/ : SUCCESS in 5m 00s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/07/666407/1/check/puppet-openstack-unit-4.8-centos-7/8d5b31d/ : SUCCESS in 38m 27s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/07/666407/1/check/puppet-openstack-unit-5.5-centos-7/d810fda/ : SUCCESS in 22m 57s\n- puppet-openstack-integration-4-scenario001-tempest-centos-7-luminous http://logs.openstack.org/07/666407/1/check/puppet-openstack-integration-4-scenario001-tempest-centos-7-luminous/b862fe5/ : SUCCESS in 1h 08m 08s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/07/666407/1/check/puppet-openstack-integration-4-scenario002-tempest-centos-7/5cdd4de/ : SUCCESS in 1h 05m 33s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/07/666407/1/check/puppet-openstack-integration-4-scenario003-tempest-centos-7/06668d4/ : SUCCESS in 1h 02m 16s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7-luminous http://logs.openstack.org/07/666407/1/check/puppet-openstack-integration-4-scenario004-tempest-centos-7-luminous/c3cfe9e/ : SUCCESS in 48m 46s\n- puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial-luminous http://logs.openstack.org/07/666407/1/check/puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial-luminous/dd9cb75/ : FAILURE in 4m 49s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/07/666407/1/check/puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/e4ceb31/ : SUCCESS in 43m 17s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/07/666407/1/check/puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/7678c80/ : FAILURE in 5m 08s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial-luminous http://logs.openstack.org/07/666407/1/check/puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial-luminous/fa7f8f2/ : FAILURE in 5m 26s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/07/666407/1/check/puppet-openstack-beaker-centos-7/87b689e/ : SUCCESS in 33m 45s\n- puppet-openstack-beaker-xenial http://logs.openstack.org/07/666407/1/check/puppet-openstack-beaker-xenial/e587c85/ : SUCCESS in 14m 08s","accounts_in_message":[],"_revision_number":1},{"id":"af03ce1d29b47c9b44added8da79fdc853c50508","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 22:07:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5f3241d7cbb47e15f7548d8a3428f7c86f008d86","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 23:13:09.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint http://logs.openstack.org/07/666407/1/gate/puppet-openstack-lint/a3c334f/ : SUCCESS in 4m 57s\n- puppet-openstack-syntax-4 http://logs.openstack.org/07/666407/1/gate/puppet-openstack-syntax-4/3175860/ : SUCCESS in 4m 29s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/07/666407/1/gate/puppet-openstack-unit-4.8-centos-7/1ec757a/ : SUCCESS in 32m 39s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/07/666407/1/gate/puppet-openstack-unit-5.5-centos-7/662cf60/ : SUCCESS in 25m 05s\n- puppet-openstack-integration-4-scenario001-tempest-centos-7-luminous http://logs.openstack.org/07/666407/1/gate/puppet-openstack-integration-4-scenario001-tempest-centos-7-luminous/ac7c7d5/ : SUCCESS in 59m 28s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/07/666407/1/gate/puppet-openstack-integration-4-scenario002-tempest-centos-7/de6028c/ : SUCCESS in 1h 01m 25s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/07/666407/1/gate/puppet-openstack-integration-4-scenario003-tempest-centos-7/fa9f9bd/ : FAILURE in 25m 15s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7-luminous http://logs.openstack.org/07/666407/1/gate/puppet-openstack-integration-4-scenario004-tempest-centos-7-luminous/2556f69/ : SUCCESS in 47m 07s\n- puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial-luminous http://logs.openstack.org/07/666407/1/gate/puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial-luminous/e478a03/ : SUCCESS in 52m 55s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/07/666407/1/gate/puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/2635bb1/ : SUCCESS in 48m 38s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/07/666407/1/gate/puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/403ce73/ : SUCCESS in 42m 33s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial-luminous http://logs.openstack.org/07/666407/1/gate/puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial-luminous/a17d09f/ : SUCCESS in 39m 11s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/07/666407/1/gate/puppet-openstack-beaker-centos-7/10c3ea3/ : SUCCESS in 19m 20s\n- puppet-openstack-beaker-xenial http://logs.openstack.org/07/666407/1/gate/puppet-openstack-beaker-xenial/58e5186/ : SUCCESS in 17m 11s","accounts_in_message":[],"_revision_number":1},{"id":"ac9347c1d65a0c93557d50f677d323338e1e50e9","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-09 13:23:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ab6af0de74cc1c685ede9e3185b76da1cd20c33e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 14:35:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint http://logs.openstack.org/07/666407/1/check/puppet-openstack-lint/b0d6f83/ : SUCCESS in 4m 52s\n- puppet-openstack-syntax-4 http://logs.openstack.org/07/666407/1/check/puppet-openstack-syntax-4/c420ac1/ : SUCCESS in 4m 47s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/07/666407/1/check/puppet-openstack-unit-4.8-centos-7/9e8be4f/ : SUCCESS in 31m 53s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/07/666407/1/check/puppet-openstack-unit-5.5-centos-7/08a251b/ : SUCCESS in 30m 39s\n- puppet-openstack-integration-4-scenario001-tempest-centos-7-luminous http://logs.openstack.org/07/666407/1/check/puppet-openstack-integration-4-scenario001-tempest-centos-7-luminous/47be23e/ : SUCCESS in 55m 15s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/07/666407/1/check/puppet-openstack-integration-4-scenario002-tempest-centos-7/42ff2db/ : SUCCESS in 54m 17s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/07/666407/1/check/puppet-openstack-integration-4-scenario003-tempest-centos-7/0625b50/ : SUCCESS in 1h 04m 56s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7-luminous http://logs.openstack.org/07/666407/1/check/puppet-openstack-integration-4-scenario004-tempest-centos-7-luminous/1ddb74c/ : SUCCESS in 51m 04s\n- puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial-luminous http://logs.openstack.org/07/666407/1/check/puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial-luminous/0c1d242/ : SUCCESS in 47m 41s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/07/666407/1/check/puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/f4ddffe/ : FAILURE in 4m 37s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/07/666407/1/check/puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/a5a52ed/ : FAILURE in 4m 09s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial-luminous http://logs.openstack.org/07/666407/1/check/puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial-luminous/a82dc17/ : SUCCESS in 28m 42s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/07/666407/1/check/puppet-openstack-beaker-centos-7/682114c/ : SUCCESS in 19m 08s\n- puppet-openstack-beaker-xenial http://logs.openstack.org/07/666407/1/check/puppet-openstack-beaker-xenial/a88770c/ : SUCCESS in 12m 42s","accounts_in_message":[],"_revision_number":1},{"id":"1035e3a24a019a574256a6fb1581828145e17978","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 14:35:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"38998de3a3034798cf084101a668a4134b433a1b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 16:03:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f5cc088e3bb8631b8964c90207595a361c1732f4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 16:03:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint http://logs.openstack.org/07/666407/1/gate/puppet-openstack-lint/bd78179/ : SUCCESS in 5m 03s\n- puppet-openstack-syntax-4 http://logs.openstack.org/07/666407/1/gate/puppet-openstack-syntax-4/454ae5d/ : SUCCESS in 5m 09s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/07/666407/1/gate/puppet-openstack-unit-4.8-centos-7/975b5d7/ : SUCCESS in 32m 32s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/07/666407/1/gate/puppet-openstack-unit-5.5-centos-7/1b5c926/ : SUCCESS in 26m 33s\n- puppet-openstack-integration-4-scenario001-tempest-centos-7-luminous http://logs.openstack.org/07/666407/1/gate/puppet-openstack-integration-4-scenario001-tempest-centos-7-luminous/18d270c/ : SUCCESS in 1h 01m 12s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/07/666407/1/gate/puppet-openstack-integration-4-scenario002-tempest-centos-7/5a5f708/ : SUCCESS in 59m 46s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/07/666407/1/gate/puppet-openstack-integration-4-scenario003-tempest-centos-7/d4edada/ : SUCCESS in 1h 13m 43s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7-luminous http://logs.openstack.org/07/666407/1/gate/puppet-openstack-integration-4-scenario004-tempest-centos-7-luminous/62fcc28/ : SUCCESS in 51m 05s\n- puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial-luminous http://logs.openstack.org/07/666407/1/gate/puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial-luminous/b8b045a/ : SUCCESS in 45m 22s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/07/666407/1/gate/puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/4886858/ : SUCCESS in 42m 42s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/07/666407/1/gate/puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/ed3f721/ : SUCCESS in 40m 05s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial-luminous http://logs.openstack.org/07/666407/1/gate/puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial-luminous/197e72c/ : SUCCESS in 35m 57s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/07/666407/1/gate/puppet-openstack-beaker-centos-7/6564052/ : SUCCESS in 16m 27s\n- puppet-openstack-beaker-xenial http://logs.openstack.org/07/666407/1/gate/puppet-openstack-beaker-xenial/dc2e320/ : SUCCESS in 13m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fedd77e759f42dd086975e285f2c7bcac419d939","revisions":{"fedd77e759f42dd086975e285f2c7bcac419d939":{"kind":"REWORK","_number":1,"created":"2019-06-19 17:11:38.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/07/666407/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/07/666407/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/07/666407/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/07/666407/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/07/666407/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/07/666407/1"}}},"commit":{"parents":[{"commit":"66fff82237a4f3d56b6a506b1906ed48312b1899","subject":"[placement] Replace port 35357 with 5000 for \"auth_url\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/66fff82237a4f3d56b6a506b1906ed48312b1899"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-06-10 18:24:23.000000000","tz":-150},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-06-19 17:11:25.000000000","tz":-150},"subject":"Ensure tunnel types are absent if not configured","message":"Ensure tunnel types are absent if not configured\n\nThis patch fixes a bug where updating a configuration wouldn\u0027t remove\nagent/tunnel_types configuration when required.\n\nConflicts:\n  spec/classes/neutron_agents_ml2_ovs_spec.rb\n\nChange-Id: I8248ad80e6ff88aec252919fa687018bda78a60b\nCloses-Bug: #1832251\n(cherry picked from commit 6c34ce2a42ae8c445a1fbe14ab18d4ee59e09c9b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/fedd77e759f42dd086975e285f2c7bcac419d939"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/fedd77e759f42dd086975e285f2c7bcac419d939"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
