)]}'
{"id":"openstack%2Fpuppet-neutron~668922","triplet_id":"openstack%2Fpuppet-neutron~master~I32fbaac96ad0476b5cf1f9f3b93e1cc8119fc59e","project":"openstack/puppet-neutron","branch":"master","hashtags":[],"change_id":"I32fbaac96ad0476b5cf1f9f3b93e1cc8119fc59e","subject":"Add support for neutron-mlnx-agent","status":"MERGED","created":"2019-07-03 14:34:37.000000000","updated":"2019-07-16 15:48:32.000000000","submitted":"2019-07-16 15:48:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668922-1563292112066-5870c8eb","meta_rev_id":"3d653b6e28f68e29da722a54588169f6804642a2","_number":668922,"virtual_id_number":668922,"owner":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"value":2,"date":"2019-07-16 15:48:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-14 12:13:48.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-07-16 12:43:14.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},{"value":2,"date":"2019-07-15 16:38:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-07-16 12:43:14.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-03 18:11:59.000000000","updated_by":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"reviewer":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"state":"REVIEWER"},{"updated":"2019-07-14 12:13:48.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-15 16:38:14.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-07-16 12:43:14.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-07-16 15:48:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a152cf0d624a4e59e94e2136d458c47df7448b03","author":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"date":"2019-07-03 14:34:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5362844bc02ff4ea5ce291faf36c26baa4664b74","author":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"date":"2019-07-03 14:45:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"883dcf8847d5ecbe8061477ad894771dfa397413","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-03 15:31:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/22/668922/2/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/1e6cf22/ : SUCCESS in 32m 04s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/22/668922/2/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/47b8d4f/ : SUCCESS in 43m 19s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/22/668922/2/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/c61875c/ : SUCCESS in 36m 20s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/22/668922/2/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/8fe60ee/ : SUCCESS in 26m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"977ed5485677f38e6a84e8b3b1551b95cdba1ae6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 17:47:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/22/668922/2/check/puppet-openstack-lint-ubuntu-bionic/6eb6756/ : SUCCESS in 4m 59s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/22/668922/2/check/puppet-openstack-syntax-5-ubuntu-bionic/7cb312e/ : SUCCESS in 4m 19s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/22/668922/2/check/puppet-openstack-syntax-6-ubuntu-bionic/fe8e735/ : SUCCESS in 4m 43s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/22/668922/2/check/puppet-openstack-unit-5.5-centos-7/e601314/ : SUCCESS in 24m 56s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/22/668922/2/check/puppet-openstack-unit-6.4-ubuntu-bionic/af78414/ : SUCCESS in 16m 37s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/22/668922/2/check/puppet-openstack-unit-latest-ubuntu-bionic/17e46cd/ : FAILURE in 18m 08s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/22/668922/2/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/827325c/ : SUCCESS in 56m 25s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/22/668922/2/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/816d383/ : FAILURE in 6m 47s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/22/668922/2/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/66caeb8/ : SUCCESS in 1h 09m 28s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus http://logs.openstack.org/22/668922/2/check/puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus/1dd30dc/ : FAILURE in 19m 11s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/22/668922/2/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/429dae3/ : FAILURE in 3m 48s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/22/668922/2/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/bf0a2ce/ : FAILURE in 5m 32s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/22/668922/2/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/8e7dae3/ : FAILURE in 5m 05s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic http://logs.openstack.org/22/668922/2/check/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/bd1045a/ : FAILURE in 6m 13s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/22/668922/2/check/puppet-openstack-beaker-centos-7/fa34125/ : SUCCESS in 22m 41s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/22/668922/2/check/puppet-openstack-beaker-ubuntu-bionic/07586b6/ : FAILURE in 12m 55s (non-voting)\n- puppet-neutron-tripleo-standalone http://logs.openstack.org/22/668922/2/check/puppet-neutron-tripleo-standalone/529a95e/ : SUCCESS in 2h 57m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5c3b633ee3410994a06f5c6f3126b5ca73e48838","author":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"date":"2019-07-03 18:11:59.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ea33abc114d24a3bc98072147e428b08a595c15f","author":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"date":"2019-07-09 11:27:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6aa6a2424f6b9f145fdf28fa6dc058dcce511784","author":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"date":"2019-07-09 11:29:30.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"107791e4fb8d4404b71df13f4c1e364a53506df7","author":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"date":"2019-07-09 11:29:37.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"ef92958a9c5c97887f2e2e0742257328edae63f9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-09 12:15:54.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/22/668922/4/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/c3d0cc1/ : SUCCESS in 33m 24s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/22/668922/4/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/f2a7d55/ : SUCCESS in 44m 45s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/22/668922/4/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/5aa5076/ : SUCCESS in 43m 59s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/22/668922/4/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/8588e6e/ : SUCCESS in 34m 33s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"dcc517fb4d24c77666f9f21ad8ebc8f709fee4e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 14:37:38.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/22/668922/4/check/puppet-openstack-lint-ubuntu-bionic/17c1cbe/ : SUCCESS in 5m 53s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/22/668922/4/check/puppet-openstack-syntax-5-ubuntu-bionic/c311e50/ : SUCCESS in 6m 29s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/22/668922/4/check/puppet-openstack-syntax-6-ubuntu-bionic/abab65a/ : SUCCESS in 7m 14s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/22/668922/4/check/puppet-openstack-unit-5.5-centos-7/ba22930/ : SUCCESS in 32m 51s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/22/668922/4/check/puppet-openstack-unit-6.4-ubuntu-bionic/e1674f8/ : SUCCESS in 17m 14s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/22/668922/4/check/puppet-openstack-unit-latest-ubuntu-bionic/cf8dfb1/ : FAILURE in 14m 08s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/22/668922/4/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/c31b686/ : FAILURE in 1h 03m 28s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/22/668922/4/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/81b39de/ : SUCCESS in 1h 05m 03s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/22/668922/4/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/3b4b537/ : FAILURE in 30m 13s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus http://logs.openstack.org/22/668922/4/check/puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus/0b1cdee/ : SUCCESS in 43m 12s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/22/668922/4/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/5bf9daf/ : FAILURE in 32m 16s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/22/668922/4/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/a22dbc6/ : FAILURE in 35m 11s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/22/668922/4/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/b8e32cf/ : FAILURE in 30m 38s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic http://logs.openstack.org/22/668922/4/check/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/ac61274/ : FAILURE in 26m 53s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/22/668922/4/check/puppet-openstack-beaker-centos-7/8ec6101/ : SUCCESS in 17m 03s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/22/668922/4/check/puppet-openstack-beaker-ubuntu-bionic/03d3f25/ : FAILURE in 8m 13s (non-voting)\n- puppet-neutron-tripleo-standalone http://logs.openstack.org/22/668922/4/check/puppet-neutron-tripleo-standalone/ce05c85/ : SUCCESS in 2h 56m 01s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8396f87061dac1b95ac7d07c8ff8f2d8bd319f46","author":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"date":"2019-07-14 11:33:27.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"152f0bdb88a6dc4a9343825d55f8517f1a13ed6f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-14 12:13:48.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/22/668922/5/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/cfb51b4/ : SUCCESS in 33m 50s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/22/668922/5/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/bdb8e1d/ : SUCCESS in 38m 00s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/22/668922/5/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/2d894d8/ : SUCCESS in 37m 23s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/22/668922/5/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/cb1f62d/ : SUCCESS in 29m 57s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"17ad02c8b1f78640d3be7b51d6ded4a7f599fa44","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-14 13:54:54.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/22/668922/5/check/puppet-openstack-lint-ubuntu-bionic/de627c0/ : SUCCESS in 5m 44s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/22/668922/5/check/puppet-openstack-syntax-5-ubuntu-bionic/9000bdf/ : SUCCESS in 4m 29s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/22/668922/5/check/puppet-openstack-syntax-6-ubuntu-bionic/7ca9b14/ : SUCCESS in 4m 23s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/22/668922/5/check/puppet-openstack-unit-5.5-centos-7/52ec2b3/ : SUCCESS in 23m 30s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/22/668922/5/check/puppet-openstack-unit-6.4-ubuntu-bionic/3e397e4/ : SUCCESS in 13m 52s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/22/668922/5/check/puppet-openstack-unit-latest-ubuntu-bionic/fba04cc/ : FAILURE in 14m 12s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/22/668922/5/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/9bcb50e/ : SUCCESS in 52m 09s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/22/668922/5/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/5c9d64a/ : SUCCESS in 53m 41s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/22/668922/5/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/b238942/ : SUCCESS in 55m 49s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus http://logs.openstack.org/22/668922/5/check/puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus/556dd12/ : SUCCESS in 42m 19s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/22/668922/5/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/3c778b2/ : FAILURE in 33m 04s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/22/668922/5/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/dfe5905/ : FAILURE in 29m 58s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/22/668922/5/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/a827f01/ : FAILURE in 28m 40s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic http://logs.openstack.org/22/668922/5/check/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/d053132/ : FAILURE in 25m 28s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/22/668922/5/check/puppet-openstack-beaker-centos-7/181b7de/ : SUCCESS in 24m 52s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/22/668922/5/check/puppet-openstack-beaker-ubuntu-bionic/e78b6c9/ : FAILURE in 9m 29s (non-voting)\n- puppet-neutron-tripleo-standalone http://logs.openstack.org/22/668922/5/check/puppet-neutron-tripleo-standalone/a249915/ : SUCCESS in 2h 19m 39s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a9654a02970b0ddc3465acbb6e66f0c115af6d98","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-15 16:38:14.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"42e7cccb1a9ea9e66b2dfb59dba740ac54af4170","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-07-16 12:43:14.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"bfcf0186e4c6a342e05cdf14078c51547bbde308","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 12:43:26.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"40a79a7abc2c4324960fbdedec196f9f61bf6d29","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 15:48:31.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/22/668922/5/gate/puppet-openstack-lint-ubuntu-bionic/529be8d/ : SUCCESS in 4m 45s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/22/668922/5/gate/puppet-openstack-syntax-5-ubuntu-bionic/51f0dc8/ : SUCCESS in 6m 58s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/22/668922/5/gate/puppet-openstack-syntax-6-ubuntu-bionic/e17e2c9/ : SUCCESS in 7m 06s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/22/668922/5/gate/puppet-openstack-unit-5.5-centos-7/e4e93dd/ : SUCCESS in 27m 29s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/22/668922/5/gate/puppet-openstack-unit-6.4-ubuntu-bionic/c55f990/ : SUCCESS in 16m 00s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/22/668922/5/gate/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/bf4e7e9/ : SUCCESS in 54m 31s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/22/668922/5/gate/puppet-openstack-integration-5-scenario002-tempest-centos-7/66bff02/ : SUCCESS in 59m 21s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/22/668922/5/gate/puppet-openstack-integration-5-scenario003-tempest-centos-7/0d4c271/ : SUCCESS in 1h 04m 28s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus http://logs.openstack.org/22/668922/5/gate/puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus/a8afc6c/ : SUCCESS in 47m 52s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/22/668922/5/gate/puppet-openstack-beaker-centos-7/26d992c/ : SUCCESS in 24m 58s\n- puppet-neutron-tripleo-standalone http://logs.openstack.org/22/668922/5/gate/puppet-neutron-tripleo-standalone/75c4687/ : SUCCESS in 3h 02m 20s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3d653b6e28f68e29da722a54588169f6804642a2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 15:48:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"e4a0b96628e09170ac775b98ce42503b5857a576","revisions":{"968c903aa45b607c3e566804f040868369df6109":{"kind":"REWORK","_number":1,"created":"2019-07-03 14:34:37.000000000","uploader":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"ref":"refs/changes/22/668922/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/22/668922/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/22/668922/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/22/668922/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/22/668922/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/22/668922/1"}}},"commit":{"parents":[{"commit":"b5769833436962a76f46ebe181859dc6db3973a3","subject":"Merge \"Ensure tunnel types are absent if not configured\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b5769833436962a76f46ebe181859dc6db3973a3"}]}],"author":{"name":"Hamdy Khader","email":"hamdyk@mellanox.com","date":"2019-07-03 14:31:39.000000000","tz":180},"committer":{"name":"Hamdy Khader","email":"hamdyk@mellanox.com","date":"2019-07-03 14:35:34.000000000","tz":180},"subject":"Add support for neutron-agent-mlnx","message":"Add support for neutron-agent-mlnx\n\nConfigure mlnx neutron agent and start the required services.\n\nChange-Id: I32fbaac96ad0476b5cf1f9f3b93e1cc8119fc59e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/968c903aa45b607c3e566804f040868369df6109"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/968c903aa45b607c3e566804f040868369df6109"}]},"branch":"refs/heads/master"},"5427b9561199b499fafa4c8050752de8b13a67d2":{"kind":"REWORK","_number":2,"created":"2019-07-03 14:45:51.000000000","uploader":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"ref":"refs/changes/22/668922/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/22/668922/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/22/668922/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/22/668922/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/22/668922/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/22/668922/2"}}},"commit":{"parents":[{"commit":"b5769833436962a76f46ebe181859dc6db3973a3","subject":"Merge \"Ensure tunnel types are absent if not configured\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b5769833436962a76f46ebe181859dc6db3973a3"}]}],"author":{"name":"Hamdy Khader","email":"hamdyk@mellanox.com","date":"2019-07-03 14:31:39.000000000","tz":180},"committer":{"name":"Hamdy Khader","email":"hamdyk@mellanox.com","date":"2019-07-03 14:44:58.000000000","tz":180},"subject":"Add support for neutron-mlnx-agent","message":"Add support for neutron-mlnx-agent\n\nConfigure mlnx neutron agent and start the required services.\n\nChange-Id: I32fbaac96ad0476b5cf1f9f3b93e1cc8119fc59e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/5427b9561199b499fafa4c8050752de8b13a67d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/5427b9561199b499fafa4c8050752de8b13a67d2"}]},"branch":"refs/heads/master"},"2c9a5bd2bd2822e875c8468d2585b6eff8a2e718":{"kind":"REWORK","_number":3,"created":"2019-07-09 11:27:24.000000000","uploader":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"ref":"refs/changes/22/668922/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/22/668922/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/22/668922/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/22/668922/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/22/668922/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/22/668922/3"}}},"commit":{"parents":[{"commit":"b5769833436962a76f46ebe181859dc6db3973a3","subject":"Merge \"Ensure tunnel types are absent if not configured\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b5769833436962a76f46ebe181859dc6db3973a3"}]}],"author":{"name":"Hamdy Khader","email":"hamdyk@mellanox.com","date":"2019-07-03 14:31:39.000000000","tz":180},"committer":{"name":"Hamdy Khader","email":"hamdyk@mellanox.com","date":"2019-07-09 11:28:28.000000000","tz":180},"subject":"Add support for neutron-mlnx-agent","message":"Add support for neutron-mlnx-agent\n\nConfigure mlnx neutron agent and start the required services.\n\nChange-Id: I32fbaac96ad0476b5cf1f9f3b93e1cc8119fc59e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/2c9a5bd2bd2822e875c8468d2585b6eff8a2e718"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/2c9a5bd2bd2822e875c8468d2585b6eff8a2e718"}]},"branch":"refs/heads/master"},"2c203a2370fdd0d8eab11a75bb6473d63d1aee51":{"kind":"REWORK","_number":4,"created":"2019-07-09 11:29:30.000000000","uploader":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"ref":"refs/changes/22/668922/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/22/668922/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/22/668922/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/22/668922/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/22/668922/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/22/668922/4"}}},"commit":{"parents":[{"commit":"b5769833436962a76f46ebe181859dc6db3973a3","subject":"Merge \"Ensure tunnel types are absent if not configured\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b5769833436962a76f46ebe181859dc6db3973a3"}]}],"author":{"name":"Hamdy Khader","email":"hamdyk@mellanox.com","date":"2019-07-03 14:31:39.000000000","tz":180},"committer":{"name":"Hamdy Khader","email":"hamdyk@mellanox.com","date":"2019-07-09 11:30:34.000000000","tz":180},"subject":"Add support for neutron-mlnx-agent","message":"Add support for neutron-mlnx-agent\n\nConfigure mlnx neutron agent and start the required services.\n\nChange-Id: I32fbaac96ad0476b5cf1f9f3b93e1cc8119fc59e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/2c203a2370fdd0d8eab11a75bb6473d63d1aee51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/2c203a2370fdd0d8eab11a75bb6473d63d1aee51"}]},"branch":"refs/heads/master"},"e4a0b96628e09170ac775b98ce42503b5857a576":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-07-14 11:33:27.000000000","uploader":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"ref":"refs/changes/22/668922/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/22/668922/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/22/668922/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/22/668922/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/22/668922/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/22/668922/5"}}},"commit":{"parents":[{"commit":"def07975c604ecc3215fa17dc10fc0ac8fa40545","subject":"Add support for configuring permitted_ethertypes on OVS agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/def07975c604ecc3215fa17dc10fc0ac8fa40545"}]}],"author":{"name":"Hamdy Khader","email":"hamdyk@mellanox.com","date":"2019-07-03 14:31:39.000000000","tz":180},"committer":{"name":"Hamdy Khader","email":"hamdyk@mellanox.com","date":"2019-07-14 11:33:27.000000000","tz":0},"subject":"Add support for neutron-mlnx-agent","message":"Add support for neutron-mlnx-agent\n\nConfigure mlnx neutron agent and start the required services.\n\nChange-Id: I32fbaac96ad0476b5cf1f9f3b93e1cc8119fc59e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/e4a0b96628e09170ac775b98ce42503b5857a576"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/e4a0b96628e09170ac775b98ce42503b5857a576"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
