)]}'
{"id":"openstack%2Fpuppet-neutron~669042","triplet_id":"openstack%2Fpuppet-neutron~master~I0834578df444671bbf86b4d08e6b719ece16c661","project":"openstack/puppet-neutron","branch":"master","topic":"bug/1835161","hashtags":[],"change_id":"I0834578df444671bbf86b4d08e6b719ece16c661","subject":"Manage oslo_middleware/max_request_body_size","status":"MERGED","created":"2019-07-04 03:44:21.000000000","updated":"2019-07-05 04:00:44.000000000","submitted":"2019-07-05 04:00:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669042-1562299244562-49c2a3a2","meta_rev_id":"26cd5899763d3afd1d8195b83ea87293096251c3","_number":669042,"virtual_id_number":669042,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-07-05 04:00:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-04 05:06:22.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-07-04 13:59:57.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-07-05 00:51:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-04 05:06:22.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-04 13:59:57.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-07-05 04:00:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"19e504681860dc3cb6f252695bec249e2bcde387","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-07-04 03:44:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"16342702a2710b0ce3dddb6c54599905a7165723","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-04 05:06:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/42/669042/1/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/c532b21/ : FAILURE in 1h 18m 35s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/42/669042/1/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/e125c33/ : SUCCESS in 56m 26s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/42/669042/1/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/dd13e35/ : SUCCESS in 57m 20s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/42/669042/1/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/bd30260/ : SUCCESS in 59m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8a8d2a7a8672b917cd6092a2544b2e2816449596","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 07:09:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/42/669042/1/check/puppet-openstack-lint-ubuntu-bionic/0b4d41e/ : SUCCESS in 4m 19s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/42/669042/1/check/puppet-openstack-syntax-5-ubuntu-bionic/3620e0b/ : SUCCESS in 5m 38s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/42/669042/1/check/puppet-openstack-syntax-6-ubuntu-bionic/69bbd5f/ : SUCCESS in 5m 24s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/42/669042/1/check/puppet-openstack-unit-5.5-centos-7/2da16bf/ : SUCCESS in 25m 27s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/42/669042/1/check/puppet-openstack-unit-6.4-ubuntu-bionic/c4f6618/ : SUCCESS in 17m 51s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/42/669042/1/check/puppet-openstack-unit-latest-ubuntu-bionic/2ac9f7d/ : FAILURE in 19m 49s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/42/669042/1/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/6e5fe3c/ : SUCCESS in 56m 02s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/42/669042/1/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/bf62053/ : SUCCESS in 56m 51s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/42/669042/1/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/3a86d40/ : SUCCESS in 1h 11m 54s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus http://logs.openstack.org/42/669042/1/check/puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus/121b065/ : SUCCESS in 44m 17s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/42/669042/1/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/cb1be56/ : FAILURE in 7m 12s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/42/669042/1/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/f59fcb2/ : FAILURE in 6m 10s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/42/669042/1/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/9a956f8/ : FAILURE in 5m 06s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic http://logs.openstack.org/42/669042/1/check/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/edd441b/ : FAILURE in 5m 50s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/42/669042/1/check/puppet-openstack-beaker-centos-7/2e433dd/ : SUCCESS in 18m 51s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/42/669042/1/check/puppet-openstack-beaker-ubuntu-bionic/04cd8be/ : FAILURE in 9m 07s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/42/669042/1/check/build-openstack-releasenotes/f5ad7d7/html/ : SUCCESS in 4m 43s\n- puppet-neutron-tripleo-standalone http://logs.openstack.org/42/669042/1/check/puppet-neutron-tripleo-standalone/adfa64c/ : TIMED_OUT in 3h 13m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3d7e061a3302a16a3441f1c13a50414e63f30648","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-07-04 13:59:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e41d6ccfae2057647b10dca8dc4b767f1e180a70","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-07-05 00:51:26.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ce319dfa7e0e9b3ab74ab9d9d17f46b2430f0049","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 00:51:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c53e6d8eb71c79d8cf0c820f83ac04b1c06cce40","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 04:00:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c4785826105fca68267ff840760d66dc2e65465e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 04:00:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/42/669042/1/gate/puppet-openstack-lint-ubuntu-bionic/30d7bac/ : SUCCESS in 5m 31s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/42/669042/1/gate/puppet-openstack-syntax-5-ubuntu-bionic/3de0f54/ : SUCCESS in 5m 16s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/42/669042/1/gate/puppet-openstack-syntax-6-ubuntu-bionic/b216492/ : SUCCESS in 5m 31s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/42/669042/1/gate/puppet-openstack-unit-5.5-centos-7/c2bdac5/ : SUCCESS in 24m 10s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/42/669042/1/gate/puppet-openstack-unit-6.4-ubuntu-bionic/1965dc3/ : SUCCESS in 15m 30s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/42/669042/1/gate/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/e5256f7/ : SUCCESS in 49m 19s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/42/669042/1/gate/puppet-openstack-integration-5-scenario002-tempest-centos-7/14bdc70/ : SUCCESS in 1h 04m 01s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/42/669042/1/gate/puppet-openstack-integration-5-scenario003-tempest-centos-7/07dbacc/ : SUCCESS in 1h 06m 28s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus http://logs.openstack.org/42/669042/1/gate/puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus/97d884b/ : SUCCESS in 48m 06s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/42/669042/1/gate/puppet-openstack-beaker-centos-7/9e5fe2f/ : SUCCESS in 16m 39s\n- build-openstack-releasenotes http://logs.openstack.org/42/669042/1/gate/build-openstack-releasenotes/79ffa49/html/ : SUCCESS in 3m 56s\n- puppet-neutron-tripleo-standalone http://logs.openstack.org/42/669042/1/gate/puppet-neutron-tripleo-standalone/c667788/ : SUCCESS in 3h 03m 27s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"380ed979d4a4c9fa82b902ff7697158be524aaf6","revisions":{"380ed979d4a4c9fa82b902ff7697158be524aaf6":{"kind":"REWORK","_number":1,"created":"2019-07-04 03:44:21.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/42/669042/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/42/669042/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/42/669042/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/42/669042/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/42/669042/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/42/669042/1"}}},"commit":{"parents":[{"commit":"b5769833436962a76f46ebe181859dc6db3973a3","subject":"Merge \"Ensure tunnel types are absent if not configured\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b5769833436962a76f46ebe181859dc6db3973a3"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-07-04 03:34:43.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-07-04 03:37:02.000000000","tz":480},"subject":"Manage oslo_middleware/max_request_body_size","message":"Manage oslo_middleware/max_request_body_size\n\nSo that we can increase it from the default 114688\n\nUseful in case for example the OS-Federation mapping is too large.\n\nIf this limit is breached neutron will return a 413 Entity Too Large\nand not log anything to neutron.log.\n\nChange-Id: I0834578df444671bbf86b4d08e6b719ece16c661\nCloses-Bug: #1835161\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/380ed979d4a4c9fa82b902ff7697158be524aaf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/380ed979d4a4c9fa82b902ff7697158be524aaf6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
