)]}'
{"id":"openstack%2Fpuppet-neutron~670429","triplet_id":"openstack%2Fpuppet-neutron~master~I3075b760149a3510fe93bc8bce062b53465a7f1e","project":"openstack/puppet-neutron","branch":"master","topic":"service-token-roles","hashtags":[],"change_id":"I3075b760149a3510fe93bc8bce062b53465a7f1e","subject":"Configure keystone_authtoken/service_token_roles","status":"MERGED","created":"2019-07-12 02:10:59.000000000","updated":"2019-07-24 01:33:58.000000000","submitted":"2019-07-24 01:33:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670429-1563932038923-d13e3f1b","meta_rev_id":"0e6f0835a4ac5d8844424ca6b21417262b70b173","_number":670429,"virtual_id_number":670429,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-07-24 01:33:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-16 07:59:53.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2019-07-17 00:56:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-07-23 22:45:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":1,"date":"2019-07-23 22:45:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-16 07:59:53.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-17 00:56:52.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2019-07-23 22:45:35.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-07-24 01:33:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d0515fee86f44d2584ae0a96e1ee304bbd3b90e4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2019-07-12 02:10:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b35d2531f17a9696de8059bd9446698606f36f13","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2019-07-12 02:18:31.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7cf1aae69181047876d4307154b81daf03a81227","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2019-07-12 02:19:29.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"ddbf259bdeb733f69a96cfb4dde5f2519f3da6a3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-12 03:03:05.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/29/670429/3/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/b89eeae/ : SUCCESS in 30m 22s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/29/670429/3/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/422cc48/ : SUCCESS in 35m 20s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/29/670429/3/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/dcaae80/ : SUCCESS in 41m 54s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/29/670429/3/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/5f6f05a/ : SUCCESS in 32m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"decac83b270f940cb8be62837e38a9ec06680491","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 05:36:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/29/670429/3/check/puppet-openstack-lint-ubuntu-bionic/cdfa8c4/ : SUCCESS in 4m 20s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/29/670429/3/check/puppet-openstack-syntax-5-ubuntu-bionic/c3d9a49/ : SUCCESS in 6m 00s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/29/670429/3/check/puppet-openstack-syntax-6-ubuntu-bionic/5bfc9a7/ : SUCCESS in 5m 29s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/29/670429/3/check/puppet-openstack-unit-5.5-centos-7/6649370/ : SUCCESS in 31m 53s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/29/670429/3/check/puppet-openstack-unit-6.4-ubuntu-bionic/7f321c9/ : SUCCESS in 14m 10s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/29/670429/3/check/puppet-openstack-unit-latest-ubuntu-bionic/8983e3b/ : FAILURE in 18m 42s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/29/670429/3/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/ddae023/ : SUCCESS in 54m 34s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/29/670429/3/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/d433d33/ : SUCCESS in 59m 50s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/29/670429/3/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/4cd5e5f/ : SUCCESS in 1h 01m 26s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus http://logs.openstack.org/29/670429/3/check/puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus/8fc43ff/ : SUCCESS in 48m 24s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/29/670429/3/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/f138495/ : FAILURE in 37m 33s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/29/670429/3/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/7286210/ : FAILURE in 33m 05s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/29/670429/3/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/1a9f07c/ : FAILURE in 28m 19s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic http://logs.openstack.org/29/670429/3/check/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/4627b71/ : FAILURE in 30m 24s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/29/670429/3/check/puppet-openstack-beaker-centos-7/78c05d3/ : SUCCESS in 18m 49s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/29/670429/3/check/puppet-openstack-beaker-ubuntu-bionic/9fc69be/ : FAILURE in 10m 00s (non-voting)\n- puppet-neutron-tripleo-standalone http://logs.openstack.org/29/670429/3/check/puppet-neutron-tripleo-standalone/350c8da/ : SUCCESS in 2h 47m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"dccf5eed8b6d9af06e2d709093698087b075be1f","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-07-15 01:21:49.000000000","message":"Patch Set 3: Code-Review-1\n\nit\u0027s better to add release note:)","accounts_in_message":[],"_revision_number":3},{"id":"846e2c1db5b55a2d700281a12d4a67009fcdf096","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2019-07-16 07:13:25.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0d00fa947303d07f6ad552d565d00df0586205d6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-16 07:59:53.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/29/670429/4/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/4c5eb87/ : SUCCESS in 38m 49s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/29/670429/4/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/cf43c4f/ : SUCCESS in 36m 41s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/29/670429/4/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/247156c/ : SUCCESS in 44m 31s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/29/670429/4/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/468163f/ : SUCCESS in 35m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b8b40996f7d95fb1c7d788edcdb973e52c8ab2e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 09:50:14.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/29/670429/4/check/puppet-openstack-lint-ubuntu-bionic/068496c/ : SUCCESS in 4m 16s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/29/670429/4/check/puppet-openstack-syntax-5-ubuntu-bionic/67017e6/ : SUCCESS in 6m 23s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/29/670429/4/check/puppet-openstack-syntax-6-ubuntu-bionic/248f37b/ : SUCCESS in 5m 03s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/29/670429/4/check/puppet-openstack-unit-5.5-centos-7/b41f46e/ : SUCCESS in 23m 49s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/29/670429/4/check/puppet-openstack-unit-6.4-ubuntu-bionic/3adc9d3/ : SUCCESS in 13m 49s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/29/670429/4/check/puppet-openstack-unit-latest-ubuntu-bionic/632554a/ : FAILURE in 14m 37s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/29/670429/4/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/c83c1da/ : SUCCESS in 51m 39s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/29/670429/4/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/614681d/ : SUCCESS in 57m 00s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/29/670429/4/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/a1dfcc8/ : SUCCESS in 1h 04m 55s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus http://logs.openstack.org/29/670429/4/check/puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus/ac331e9/ : SUCCESS in 44m 21s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/29/670429/4/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/b35fd5b/ : FAILURE in 36m 30s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/29/670429/4/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/b245025/ : FAILURE in 39m 34s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/29/670429/4/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/1dae5d4/ : FAILURE in 28m 16s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic http://logs.openstack.org/29/670429/4/check/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/6b506ab/ : FAILURE in 23m 32s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/29/670429/4/check/puppet-openstack-beaker-centos-7/5075231/ : SUCCESS in 20m 48s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/29/670429/4/check/puppet-openstack-beaker-ubuntu-bionic/36b9464/ : FAILURE in 10m 24s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/29/670429/4/check/build-openstack-releasenotes/608505f/html/ : SUCCESS in 4m 01s\n- puppet-neutron-tripleo-standalone http://logs.openstack.org/29/670429/4/check/puppet-neutron-tripleo-standalone/842b3b4/ : SUCCESS in 2h 34m 02s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9bb67a4cdd40de72daa54423c6bb68b32d8a2054","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-07-17 00:56:52.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"916d744d1ca0a965af022cd9b838424065bab57d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-23 22:45:35.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"c841d4c6b82e4ede557879208d058fd7afb9df53","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 22:45:47.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"45f7d681f32533dca967c7af36786d5cce728ac6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 01:33:58.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/29/670429/4/gate/puppet-openstack-lint-ubuntu-bionic/57c9323/ : SUCCESS in 5m 37s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/29/670429/4/gate/puppet-openstack-syntax-5-ubuntu-bionic/c650a1d/ : SUCCESS in 6m 06s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/29/670429/4/gate/puppet-openstack-syntax-6-ubuntu-bionic/bbac993/ : SUCCESS in 6m 54s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/29/670429/4/gate/puppet-openstack-unit-5.5-centos-7/1ead97a/ : SUCCESS in 39m 42s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/29/670429/4/gate/puppet-openstack-unit-6.4-ubuntu-bionic/9d4fa8e/ : SUCCESS in 18m 38s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/29/670429/4/gate/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/60ee144/ : SUCCESS in 1h 11m 30s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/29/670429/4/gate/puppet-openstack-integration-5-scenario002-tempest-centos-7/2732bd9/ : SUCCESS in 1h 12m 23s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/29/670429/4/gate/puppet-openstack-integration-5-scenario003-tempest-centos-7/eeab7dd/ : SUCCESS in 1h 03m 29s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus http://logs.openstack.org/29/670429/4/gate/puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus/7d679ec/ : SUCCESS in 45m 02s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/29/670429/4/gate/puppet-openstack-beaker-centos-7/3b94f24/ : SUCCESS in 17m 15s\n- build-openstack-releasenotes http://logs.openstack.org/29/670429/4/gate/build-openstack-releasenotes/1124b86/html/ : SUCCESS in 4m 25s\n- puppet-neutron-tripleo-standalone http://logs.openstack.org/29/670429/4/gate/puppet-neutron-tripleo-standalone/b56a42a/ : SUCCESS in 2h 43m 01s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"bd386656d868e0bf65f92c9f1fc8912ab14865c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 01:33:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c460fd0bf6779b14653ed394ad06809e165bfa07","revisions":{"54a6878e3a8dcb8cd121f35536ac94859cfd82ab":{"kind":"REWORK","_number":1,"created":"2019-07-12 02:10:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/29/670429/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/29/670429/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/29/670429/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/29/670429/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/29/670429/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/29/670429/1"}}},"commit":{"parents":[{"commit":"def07975c604ecc3215fa17dc10fc0ac8fa40545","subject":"Add support for configuring permitted_ethertypes on OVS agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/def07975c604ecc3215fa17dc10fc0ac8fa40545"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-07-12 02:09:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-07-12 02:10:57.000000000","tz":540},"subject":"Configure authtoken/service_user_roles","message":"Configure authtoken/service_user_roles\n\nThis patch introduces a new hieradata to configure service_user_roles\nin keystone authtoken middleware configuration, so that we can use\na customized role for user who uses service token feature.\n\nChange-Id: I3075b760149a3510fe93bc8bce062b53465a7f1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/54a6878e3a8dcb8cd121f35536ac94859cfd82ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/54a6878e3a8dcb8cd121f35536ac94859cfd82ab"}]},"branch":"refs/heads/master"},"f95adda9c5003ec038731dc293b10ac743037b78":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-07-12 02:18:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/29/670429/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/29/670429/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/29/670429/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/29/670429/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/29/670429/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/29/670429/2"}}},"commit":{"parents":[{"commit":"def07975c604ecc3215fa17dc10fc0ac8fa40545","subject":"Add support for configuring permitted_ethertypes on OVS agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/def07975c604ecc3215fa17dc10fc0ac8fa40545"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-07-12 02:09:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-07-12 02:18:20.000000000","tz":540},"subject":"Configure keystone_authtoken/service_user_roles","message":"Configure keystone_authtoken/service_user_roles\n\nThis patch introduces a new hieradata to configure service_user_roles\nin keystone authtoken middleware configuration, so that we can use\na customized role for user who uses service token feature.\n\nChange-Id: I3075b760149a3510fe93bc8bce062b53465a7f1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/f95adda9c5003ec038731dc293b10ac743037b78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/f95adda9c5003ec038731dc293b10ac743037b78"}]},"branch":"refs/heads/master"},"85dbb6a81a3a002ebe4bf70bccf861eb1462922c":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-07-12 02:19:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/29/670429/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/29/670429/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/29/670429/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/29/670429/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/29/670429/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/29/670429/3"}}},"commit":{"parents":[{"commit":"def07975c604ecc3215fa17dc10fc0ac8fa40545","subject":"Add support for configuring permitted_ethertypes on OVS agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/def07975c604ecc3215fa17dc10fc0ac8fa40545"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-07-12 02:09:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-07-12 02:19:04.000000000","tz":540},"subject":"Configure keystone_authtoken/service_token_roles","message":"Configure keystone_authtoken/service_token_roles\n\nThis patch introduces a new hieradata to configure service_token_roles\nin keystone authtoken middleware configuration, so that we can use\na customized role for user who uses service token feature.\n\nChange-Id: I3075b760149a3510fe93bc8bce062b53465a7f1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/85dbb6a81a3a002ebe4bf70bccf861eb1462922c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/85dbb6a81a3a002ebe4bf70bccf861eb1462922c"}]},"branch":"refs/heads/master"},"c460fd0bf6779b14653ed394ad06809e165bfa07":{"kind":"REWORK","_number":4,"created":"2019-07-16 07:13:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/29/670429/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/29/670429/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/29/670429/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/29/670429/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/29/670429/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/29/670429/4"}}},"commit":{"parents":[{"commit":"def07975c604ecc3215fa17dc10fc0ac8fa40545","subject":"Add support for configuring permitted_ethertypes on OVS agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/def07975c604ecc3215fa17dc10fc0ac8fa40545"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-07-12 02:09:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-07-16 07:12:52.000000000","tz":540},"subject":"Configure keystone_authtoken/service_token_roles","message":"Configure keystone_authtoken/service_token_roles\n\nThis patch introduces a new hieradata to configure service_token_roles\nin keystone authtoken middleware configuration, so that we can use\na customized role for user who uses service token feature.\n\nChange-Id: I3075b760149a3510fe93bc8bce062b53465a7f1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c460fd0bf6779b14653ed394ad06809e165bfa07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c460fd0bf6779b14653ed394ad06809e165bfa07"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
