)]}'
{"id":"openstack%2Fpuppet-neutron~674471","triplet_id":"openstack%2Fpuppet-neutron~master~I31fd5223ce528c4fe0ee0fe1b8f7ed8cdf38a796","project":"openstack/puppet-neutron","branch":"master","hashtags":[],"change_id":"I31fd5223ce528c4fe0ee0fe1b8f7ed8cdf38a796","subject":"Fix config name for neutron-mlnx-agent","status":"MERGED","created":"2019-08-03 19:12:53.000000000","updated":"2019-08-27 16:44:27.000000000","submitted":"2019-08-27 16:44:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674471-1566924267183-91f773fe","meta_rev_id":"984688a596128482dbba4a2bfb3c85c6a7e94d40","_number":674471,"virtual_id_number":674471,"owner":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2019-08-27 13:23:41.000000000","_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},{"value":0,"date":"2019-08-06 18:59:47.000000000","_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"value":2,"date":"2019-08-27 16:44:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-06 09:22:12.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-08-20 18:05:48.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},{"value":2,"date":"2019-08-27 13:33:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2019-08-11 23:09:21.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},{"value":0,"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},{"value":1,"date":"2019-08-27 13:33:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"date":"2019-08-27 07:14:27.000000000","_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},{"value":0,"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-06 09:22:12.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-06 18:59:47.000000000","updated_by":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"reviewer":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"state":"REVIEWER"},{"updated":"2019-08-11 23:09:21.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2019-08-20 18:05:48.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-08-27 07:14:27.000000000","updated_by":{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},"reviewer":{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},"state":"REVIEWER"},{"updated":"2019-08-27 13:33:45.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-27 16:44:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"26ecb1d4344fbde03a17d2868407d9671e14c663","author":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"date":"2019-08-03 19:12:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5be8b8e46b9bd7ece275cabfd197b44c88e39144","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-03 20:06:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/71/674471/1/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/cf36a62/ : SUCCESS in 49m 02s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/71/674471/1/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/e1405f3/ : SUCCESS in 51m 17s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/71/674471/1/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/590c9cc/ : SUCCESS in 38m 01s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/71/674471/1/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/1111775/ : SUCCESS in 30m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d2ed13e3142e9017cdf49cb7b53779e542e76708","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-03 20:19:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://logs.opendev.org/71/674471/1/check/puppet-openstack-lint-ubuntu-bionic/5f62351/ : SUCCESS in 4m 39s\n- puppet-openstack-syntax-5-ubuntu-bionic https://logs.opendev.org/71/674471/1/check/puppet-openstack-syntax-5-ubuntu-bionic/5987e26/ : SUCCESS in 4m 21s\n- puppet-openstack-syntax-6-ubuntu-bionic https://logs.opendev.org/71/674471/1/check/puppet-openstack-syntax-6-ubuntu-bionic/6e8628a/ : SUCCESS in 4m 07s\n- puppet-openstack-unit-5.5-centos-7 https://logs.opendev.org/71/674471/1/check/puppet-openstack-unit-5.5-centos-7/21cf386/ : SUCCESS in 26m 27s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://logs.opendev.org/71/674471/1/check/puppet-openstack-unit-6.4-ubuntu-bionic/7b09b08/ : SUCCESS in 17m 21s\n- puppet-openstack-unit-latest-ubuntu-bionic https://logs.opendev.org/71/674471/1/check/puppet-openstack-unit-latest-ubuntu-bionic/a94ac51/ : FAILURE in 13m 46s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://logs.opendev.org/71/674471/1/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/33e5db6/ : SUCCESS in 46m 35s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://logs.opendev.org/71/674471/1/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/c1a9c41/ : SUCCESS in 59m 00s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://logs.opendev.org/71/674471/1/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/5571d00/ : SUCCESS in 1h 04m 28s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://logs.opendev.org/71/674471/1/check/puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus/9902d88/ : SUCCESS in 47m 10s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://logs.opendev.org/71/674471/1/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/b5867c9/ : FAILURE in 32m 19s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://logs.opendev.org/71/674471/1/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/f29dc91/ : FAILURE in 33m 28s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://logs.opendev.org/71/674471/1/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/6813967/ : FAILURE in 24m 31s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://logs.opendev.org/71/674471/1/check/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/1dcf8bd/ : FAILURE in 24m 58s (non-voting)\n- puppet-openstack-beaker-centos-7 https://logs.opendev.org/71/674471/1/check/puppet-openstack-beaker-centos-7/1dfac70/ : SUCCESS in 20m 40s\n- puppet-openstack-beaker-ubuntu-bionic https://logs.opendev.org/71/674471/1/check/puppet-openstack-beaker-ubuntu-bionic/bb9a066/ : FAILURE in 7m 47s (non-voting)\n- puppet-neutron-tripleo-standalone https://logs.opendev.org/71/674471/1/check/puppet-neutron-tripleo-standalone/298695b/ : FAILURE in 9m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fc5bf51b527dd20e74e2e28d0f95e2782a22d1bf","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-05 13:43:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"74d1d53a133f7306a41643b241c9702f86ef845a","author":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"date":"2019-08-05 14:16:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2d4f5ed11fd9851ab2fd779a0f54ccd508f054c2","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-05 14:22:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fc6a72de913a1372e0e70b568467e26f6c9b98f0","author":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"date":"2019-08-06 07:58:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8f305ea1b8a4765060a9782c51c8b9f091a7843c","author":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"date":"2019-08-06 08:34:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e04a4b5b78d4397e714da5fd6c3d1aafc2168a0b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-06 09:22:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/71/674471/2/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/7951747/ : SUCCESS in 32m 40s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/71/674471/2/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/8ca5b8a/ : SUCCESS in 45m 25s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/71/674471/2/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/3e54f4d/ : SUCCESS in 45m 03s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/71/674471/2/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/d59513a/ : SUCCESS in 28m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c0864dc5bcd56cd49921bd2b59072f66026005f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 11:46:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://logs.opendev.org/71/674471/2/check/puppet-openstack-lint-ubuntu-bionic/67bde5b/ : SUCCESS in 4m 38s\n- puppet-openstack-syntax-5-ubuntu-bionic https://logs.opendev.org/71/674471/2/check/puppet-openstack-syntax-5-ubuntu-bionic/169c107/ : SUCCESS in 4m 34s\n- puppet-openstack-syntax-6-ubuntu-bionic https://logs.opendev.org/71/674471/2/check/puppet-openstack-syntax-6-ubuntu-bionic/1d62e26/ : SUCCESS in 4m 37s\n- puppet-openstack-unit-5.5-centos-7 https://logs.opendev.org/71/674471/2/check/puppet-openstack-unit-5.5-centos-7/15ab15f/ : SUCCESS in 26m 10s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://logs.opendev.org/71/674471/2/check/puppet-openstack-unit-6.4-ubuntu-bionic/6cc61a0/ : SUCCESS in 16m 31s\n- puppet-openstack-unit-latest-ubuntu-bionic https://logs.opendev.org/71/674471/2/check/puppet-openstack-unit-latest-ubuntu-bionic/965b184/ : FAILURE in 16m 39s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://logs.opendev.org/71/674471/2/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/2fa0474/ : SUCCESS in 51m 49s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://logs.opendev.org/71/674471/2/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/09f36d2/ : SUCCESS in 59m 39s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://logs.opendev.org/71/674471/2/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/53a73c6/ : SUCCESS in 1h 13m 15s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://logs.opendev.org/71/674471/2/check/puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus/1390c05/ : SUCCESS in 50m 30s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://logs.opendev.org/71/674471/2/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/f00cb00/ : FAILURE in 40m 04s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://logs.opendev.org/71/674471/2/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/f365414/ : FAILURE in 34m 10s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://logs.opendev.org/71/674471/2/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/c0e19f6/ : FAILURE in 25m 16s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://logs.opendev.org/71/674471/2/check/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/c2374a0/ : FAILURE in 22m 33s (non-voting)\n- puppet-openstack-beaker-centos-7 https://logs.opendev.org/71/674471/2/check/puppet-openstack-beaker-centos-7/9ad5429/ : SUCCESS in 17m 12s\n- puppet-openstack-beaker-ubuntu-bionic https://logs.opendev.org/71/674471/2/check/puppet-openstack-beaker-ubuntu-bionic/c0b418a/ : FAILURE in 12m 30s (non-voting)\n- puppet-neutron-tripleo-standalone https://logs.opendev.org/71/674471/2/check/puppet-neutron-tripleo-standalone/12fecd1/ : SUCCESS in 2h 55m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dc96dea104ffa0012ce5dc76d036f4b9245999fb","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-06 15:11:31.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e4b4e94969b09736d437b8ca97ffa70ec7eac074","author":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"date":"2019-08-06 18:59:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6ecb6f2d43acef2d7b98c033bf9ce2649fb46579","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-08-11 23:09:21.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)\n\nI don\u0027t see an issue with changing this since the feature was added in master, we don\u0027t claim compatibility for changes on master specifically.\n\nI would however like to see a follow up patch that adds a release note about this new feature (which was not added in the initial patch you mentioned), also highlighted a small formatting fix which doesn\u0027t really matter.","accounts_in_message":[],"_revision_number":2},{"id":"5fa6b39037004b7152d09387d0cde305a3b079b4","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-20 18:05:48.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"021b879eae9f8c2f82bdfe2a0a64d4f9ee589d99","author":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"date":"2019-08-26 12:09:17.000000000","message":"Patch Set 2:\n\n\u003e (1 comment)\n \u003e \n \u003e I don\u0027t see an issue with changing this since the feature was added\n \u003e in master, we don\u0027t claim compatibility for changes on master\n \u003e specifically.\n \u003e \n \u003e I would however like to see a follow up patch that adds a release\n \u003e note about this new feature (which was not added in the initial\n \u003e patch you mentioned), also highlighted a small formatting fix which\n \u003e doesn\u0027t really matter.\n\nWill fix the newline and add release note in a follow-up patch.","accounts_in_message":[],"_revision_number":2},{"id":"0ca56de4a3c3055644317d5ae6ccdc276626b1a0","author":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"date":"2019-08-27 13:23:41.000000000","message":"Patch Set 2:\n\nfollow-up patch:\nhttps://review.opendev.org/#/c/678831/","accounts_in_message":[],"_revision_number":2},{"id":"eb483a65f606f81747bbee22304906bd040606f0","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-27 13:33:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"30defe559676db5c66c1a737485c723ad9d45888","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 13:33:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8c2ab79c91644b12acb2db7163183590b329172f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 16:44:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8c9e19e9ea9e4aad9f0e0c4bf858afea : SUCCESS in 5m 21s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7a3ad142b1c14ceca7cc83cb5aff1571 : SUCCESS in 5m 04s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7301939059c646ee8f8e0591c7f5820b : SUCCESS in 5m 27s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/a3ce8219f57b4d4784fb95bdfb82fbc0 : SUCCESS in 27m 46s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f1e1e616eb474d708fda786a5843cc2e : SUCCESS in 16m 05s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/22b9fcd292384f2388c4056722fdb111 : SUCCESS in 55m 02s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/26bc334e10cc4264876d551e458def30 : SUCCESS in 57m 56s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/fef5f86afe6840d6a4a2510c73264f97 : SUCCESS in 1h 00m 44s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/0a9a73a2926949eeb50b11bb1db5c603 : SUCCESS in 47m 09s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/9366d96329324854a278f65e0dcba602 : SUCCESS in 19m 14s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0a01ca75c3f7424aac9d785dafbbda11 : SUCCESS in 3h 05m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"984688a596128482dbba4a2bfb3c85c6a7e94d40","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 16:44:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"28d9cf41075775d4aa4b668f1c94472908d23f35","revisions":{"c4855e151689e163caba8bf0e574245de72769e3":{"kind":"REWORK","_number":1,"created":"2019-08-03 19:12:53.000000000","uploader":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"ref":"refs/changes/71/674471/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/71/674471/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/71/674471/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/71/674471/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/71/674471/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/71/674471/1"}}},"commit":{"parents":[{"commit":"db5c3194582eab56dab50be0f8a0c6a796a9bae4","subject":"Prepare Train M3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/db5c3194582eab56dab50be0f8a0c6a796a9bae4"}]}],"author":{"name":"Hamdy Khader","email":"hamdyk@mellanox.com","date":"2019-08-03 19:10:05.000000000","tz":180},"committer":{"name":"Hamdy Khader","email":"hamdyk@mellanox.com","date":"2019-08-03 19:10:05.000000000","tz":180},"subject":"Fix config name for neutron-mlnx-agent","message":"Fix config name for neutron-mlnx-agent\n\n  - use config name \u0027physical_interface_mappings\u0027\n  - create/ensure init file \u0027/etc/neutron/plugins/mlnx/mlnx_conf.ini\u0027\n\nChange-Id: I31fd5223ce528c4fe0ee0fe1b8f7ed8cdf38a796\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c4855e151689e163caba8bf0e574245de72769e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c4855e151689e163caba8bf0e574245de72769e3"}]},"branch":"refs/heads/master"},"28d9cf41075775d4aa4b668f1c94472908d23f35":{"kind":"REWORK","_number":2,"created":"2019-08-06 08:34:47.000000000","uploader":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"ref":"refs/changes/71/674471/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/71/674471/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/71/674471/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/71/674471/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/71/674471/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/71/674471/2"}}},"commit":{"parents":[{"commit":"db5c3194582eab56dab50be0f8a0c6a796a9bae4","subject":"Prepare Train M3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/db5c3194582eab56dab50be0f8a0c6a796a9bae4"}]}],"author":{"name":"Hamdy Khader","email":"hamdyk@mellanox.com","date":"2019-08-03 19:10:05.000000000","tz":180},"committer":{"name":"Hamdy Khader","email":"hamdyk@mellanox.com","date":"2019-08-06 08:34:45.000000000","tz":180},"subject":"Fix config name for neutron-mlnx-agent","message":"Fix config name for neutron-mlnx-agent\n\n  - use config name \u0027physical_interface_mappings\u0027\n\nChange-Id: I31fd5223ce528c4fe0ee0fe1b8f7ed8cdf38a796\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/28d9cf41075775d4aa4b668f1c94472908d23f35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/28d9cf41075775d4aa4b668f1c94472908d23f35"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
