)]}'
{"id":"openstack%2Fpuppet-neutron~674603","triplet_id":"openstack%2Fpuppet-neutron~master~I5a03a74e68d2b8b7364341ba2ed77dfb419ebc58","project":"openstack/puppet-neutron","branch":"master","hashtags":[],"change_id":"I5a03a74e68d2b8b7364341ba2ed77dfb419ebc58","subject":"Support connecting OVN DB over SSL","status":"MERGED","created":"2019-08-05 13:01:48.000000000","updated":"2019-08-08 00:46:27.000000000","submitted":"2019-08-08 00:46:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674603-1565225187812-e657b839","meta_rev_id":"441a3d2ef59414098e8ea6bcea2981c27c917d47","_number":674603,"virtual_id_number":674603,"owner":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-08-08 00:46:27.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-06 16:21:05.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-08-07 13:53:18.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-08-07 21:58:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-08-07 21:58:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-06 16:21:05.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-07 13:53:18.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-08-07 21:58:40.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-08 00:46:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c00d4a32e8619497450b908912e0798e44e8d158","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2019-08-05 13:01:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a5259ae33ac2cbed17834bccb32f2624e27904c7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-05 13:47:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/03/674603/1/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/4c5e523/ : SUCCESS in 41m 33s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/03/674603/1/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/0685be4/ : SUCCESS in 43m 38s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/03/674603/1/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/23b49b0/ : SUCCESS in 38m 59s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/03/674603/1/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/a5502b1/ : SUCCESS in 29m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ca39c3b720319aec5cbe2412ef30f2f31e92d3d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 15:49:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://logs.opendev.org/03/674603/1/check/puppet-openstack-lint-ubuntu-bionic/5e65d91/ : SUCCESS in 5m 24s\n- puppet-openstack-syntax-5-ubuntu-bionic https://logs.opendev.org/03/674603/1/check/puppet-openstack-syntax-5-ubuntu-bionic/d557c4e/ : SUCCESS in 5m 35s\n- puppet-openstack-syntax-6-ubuntu-bionic https://logs.opendev.org/03/674603/1/check/puppet-openstack-syntax-6-ubuntu-bionic/dc599a8/ : SUCCESS in 4m 14s\n- puppet-openstack-unit-5.5-centos-7 https://logs.opendev.org/03/674603/1/check/puppet-openstack-unit-5.5-centos-7/5ff19b5/ : SUCCESS in 28m 05s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://logs.opendev.org/03/674603/1/check/puppet-openstack-unit-6.4-ubuntu-bionic/6c4183f/ : SUCCESS in 16m 59s\n- puppet-openstack-unit-latest-ubuntu-bionic https://logs.opendev.org/03/674603/1/check/puppet-openstack-unit-latest-ubuntu-bionic/95c21b3/ : FAILURE in 17m 43s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://logs.opendev.org/03/674603/1/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/fee5415/ : SUCCESS in 47m 26s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://logs.opendev.org/03/674603/1/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/95a6c7f/ : SUCCESS in 54m 58s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://logs.opendev.org/03/674603/1/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/524f5ce/ : SUCCESS in 1h 01m 17s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://logs.opendev.org/03/674603/1/check/puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus/826652d/ : SUCCESS in 40m 41s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://logs.opendev.org/03/674603/1/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/8ce0da1/ : FAILURE in 38m 44s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://logs.opendev.org/03/674603/1/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/6a40bf4/ : FAILURE in 28m 54s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://logs.opendev.org/03/674603/1/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/7e93215/ : FAILURE in 27m 50s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://logs.opendev.org/03/674603/1/check/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/747d232/ : FAILURE in 32m 01s (non-voting)\n- puppet-openstack-beaker-centos-7 https://logs.opendev.org/03/674603/1/check/puppet-openstack-beaker-centos-7/04391c3/ : SUCCESS in 22m 32s\n- puppet-openstack-beaker-ubuntu-bionic https://logs.opendev.org/03/674603/1/check/puppet-openstack-beaker-ubuntu-bionic/6989e6b/ : FAILURE in 11m 59s (non-voting)\n- build-openstack-releasenotes https://logs.opendev.org/03/674603/1/check/build-openstack-releasenotes/86434e9/ : FAILURE in 3m 45s\n- puppet-neutron-tripleo-standalone https://logs.opendev.org/03/674603/1/check/puppet-neutron-tripleo-standalone/b74a559/ : SUCCESS in 2h 41m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dfdd98cbf4ba4a34dc88d39a5ebcc787023a192c","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2019-08-06 15:32:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5d54240dc68661c8e5407a860e4cf47fac6f0a91","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-06 16:21:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/03/674603/2/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/a99ea83/ : SUCCESS in 42m 17s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/03/674603/2/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/90d3580/ : SUCCESS in 39m 41s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/03/674603/2/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/a05ae60/ : SUCCESS in 46m 13s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/03/674603/2/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/6d84119/ : SUCCESS in 34m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c0d19982cbc6d5efa1ffff81db984654e54b89e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 18:51:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://logs.opendev.org/03/674603/2/check/puppet-openstack-lint-ubuntu-bionic/7fc2ab0/ : SUCCESS in 5m 06s\n- puppet-openstack-syntax-5-ubuntu-bionic https://logs.opendev.org/03/674603/2/check/puppet-openstack-syntax-5-ubuntu-bionic/e398bc6/ : SUCCESS in 4m 33s\n- puppet-openstack-syntax-6-ubuntu-bionic https://logs.opendev.org/03/674603/2/check/puppet-openstack-syntax-6-ubuntu-bionic/dcf7104/ : SUCCESS in 5m 19s\n- puppet-openstack-unit-5.5-centos-7 https://logs.opendev.org/03/674603/2/check/puppet-openstack-unit-5.5-centos-7/80b9dd2/ : SUCCESS in 32m 32s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://logs.opendev.org/03/674603/2/check/puppet-openstack-unit-6.4-ubuntu-bionic/287ddbc/ : SUCCESS in 21m 20s\n- puppet-openstack-unit-latest-ubuntu-bionic https://logs.opendev.org/03/674603/2/check/puppet-openstack-unit-latest-ubuntu-bionic/8a3d904/ : FAILURE in 17m 02s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://logs.opendev.org/03/674603/2/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/dd799a6/ : SUCCESS in 57m 17s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://logs.opendev.org/03/674603/2/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/3bc96d5/ : SUCCESS in 58m 53s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://logs.opendev.org/03/674603/2/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/6a6c181/ : SUCCESS in 57m 00s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://logs.opendev.org/03/674603/2/check/puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus/4cedbd0/ : SUCCESS in 48m 10s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://logs.opendev.org/03/674603/2/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/607705c/ : FAILURE in 41m 00s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://logs.opendev.org/03/674603/2/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/23f9afa/ : FAILURE in 38m 09s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://logs.opendev.org/03/674603/2/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/42a2629/ : FAILURE in 41m 13s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://logs.opendev.org/03/674603/2/check/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/3250c36/ : FAILURE in 24m 13s (non-voting)\n- puppet-openstack-beaker-centos-7 https://logs.opendev.org/03/674603/2/check/puppet-openstack-beaker-centos-7/a5d1f2f/ : SUCCESS in 19m 51s\n- puppet-openstack-beaker-ubuntu-bionic https://logs.opendev.org/03/674603/2/check/puppet-openstack-beaker-ubuntu-bionic/c31803e/ : FAILURE in 9m 03s (non-voting)\n- build-openstack-releasenotes https://logs.opendev.org/03/674603/2/check/build-openstack-releasenotes/5b20496/html/ : SUCCESS in 4m 25s\n- puppet-neutron-tripleo-standalone https://logs.opendev.org/03/674603/2/check/puppet-neutron-tripleo-standalone/4ce423a/ : SUCCESS in 3h 11m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2f9ab45c77a1eeaf78cecaea55b6dbfc4e82e69f","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-07 13:53:18.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1ba47924b4a7d92382cfec1c57175bb9e496d8d8","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-07 21:58:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"df8a8b90782f49c1a827d9dc0a98f729e8d07f7c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 21:58:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"37ce1c50b4b1a10c320716d1ec46bb0cd13e1df8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 00:46:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"441a3d2ef59414098e8ea6bcea2981c27c917d47","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 00:46:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://logs.opendev.org/03/674603/2/gate/puppet-openstack-lint-ubuntu-bionic/c3b110f/ : SUCCESS in 4m 20s\n- puppet-openstack-syntax-5-ubuntu-bionic https://logs.opendev.org/03/674603/2/gate/puppet-openstack-syntax-5-ubuntu-bionic/861a235/ : SUCCESS in 4m 26s\n- puppet-openstack-syntax-6-ubuntu-bionic https://logs.opendev.org/03/674603/2/gate/puppet-openstack-syntax-6-ubuntu-bionic/dc05bf0/ : SUCCESS in 4m 40s\n- puppet-openstack-unit-5.5-centos-7 https://logs.opendev.org/03/674603/2/gate/puppet-openstack-unit-5.5-centos-7/62f9b9a/ : SUCCESS in 28m 39s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://logs.opendev.org/03/674603/2/gate/puppet-openstack-unit-6.4-ubuntu-bionic/0d0f205/ : SUCCESS in 22m 01s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://logs.opendev.org/03/674603/2/gate/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/d7329a5/ : SUCCESS in 54m 23s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://logs.opendev.org/03/674603/2/gate/puppet-openstack-integration-5-scenario002-tempest-centos-7/c9994a9/ : SUCCESS in 1h 04m 21s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://logs.opendev.org/03/674603/2/gate/puppet-openstack-integration-5-scenario003-tempest-centos-7/890d596/ : SUCCESS in 1h 05m 25s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://logs.opendev.org/03/674603/2/gate/puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus/35445c2/ : SUCCESS in 45m 17s\n- puppet-openstack-beaker-centos-7 https://logs.opendev.org/03/674603/2/gate/puppet-openstack-beaker-centos-7/4ab062b/ : SUCCESS in 18m 33s\n- build-openstack-releasenotes https://logs.opendev.org/03/674603/2/gate/build-openstack-releasenotes/90fdf9f/html/ : SUCCESS in 4m 16s\n- puppet-neutron-tripleo-standalone https://logs.opendev.org/03/674603/2/gate/puppet-neutron-tripleo-standalone/e3b467b/ : SUCCESS in 2h 42m 35s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c6881ea3febb1d8ccf2b1ca48d22cb89dd11f5ca","revisions":{"03869b638d4339b5258d1d67869bd0d9c5bf1143":{"kind":"REWORK","_number":1,"created":"2019-08-05 13:01:48.000000000","uploader":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"ref":"refs/changes/03/674603/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/03/674603/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/03/674603/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/03/674603/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/03/674603/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/03/674603/1"}}},"commit":{"parents":[{"commit":"db5c3194582eab56dab50be0f8a0c6a796a9bae4","subject":"Prepare Train M3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/db5c3194582eab56dab50be0f8a0c6a796a9bae4"}]}],"author":{"name":"Kamil Sambor","email":"ksambor@redhat.com","date":"2019-08-05 12:52:43.000000000","tz":120},"committer":{"name":"Kamil Sambor","email":"ksambor@redhat.com","date":"2019-08-05 13:01:45.000000000","tz":120},"subject":"Support connecting OVN DB over SSL","message":"Support connecting OVN DB over SSL\n\nThis patch introduce parameters which support using SSL to connect to\nOVN_Northbound DB and OVN_Southbound DB.\n\nChange-Id: I5a03a74e68d2b8b7364341ba2ed77dfb419ebc58\nSigned-off-by: Kamil Sambor \u003cksambor@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/03869b638d4339b5258d1d67869bd0d9c5bf1143"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/03869b638d4339b5258d1d67869bd0d9c5bf1143"}]},"branch":"refs/heads/master"},"c6881ea3febb1d8ccf2b1ca48d22cb89dd11f5ca":{"kind":"REWORK","_number":2,"created":"2019-08-06 15:32:57.000000000","uploader":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"ref":"refs/changes/03/674603/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/03/674603/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/03/674603/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/03/674603/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/03/674603/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/03/674603/2"}}},"commit":{"parents":[{"commit":"db5c3194582eab56dab50be0f8a0c6a796a9bae4","subject":"Prepare Train M3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/db5c3194582eab56dab50be0f8a0c6a796a9bae4"}]}],"author":{"name":"Kamil Sambor","email":"ksambor@redhat.com","date":"2019-08-05 12:52:43.000000000","tz":120},"committer":{"name":"Kamil Sambor","email":"ksambor@redhat.com","date":"2019-08-06 15:32:39.000000000","tz":120},"subject":"Support connecting OVN DB over SSL","message":"Support connecting OVN DB over SSL\n\nThis patch introduce parameters which support using SSL to connect to\nOVN_Northbound DB and OVN_Southbound DB.\n\nChange-Id: I5a03a74e68d2b8b7364341ba2ed77dfb419ebc58\nSigned-off-by: Kamil Sambor \u003cksambor@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c6881ea3febb1d8ccf2b1ca48d22cb89dd11f5ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c6881ea3febb1d8ccf2b1ca48d22cb89dd11f5ca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
