)]}'
{"id":"openstack%2Fpuppet-neutron~676570","triplet_id":"openstack%2Fpuppet-neutron~master~I6d4d6627804bd0718c3183c6cd419c12212c74ea","project":"openstack/puppet-neutron","branch":"master","topic":"remove_deprecate_pki_related_options","hashtags":[],"change_id":"I6d4d6627804bd0718c3183c6cd419c12212c74ea","subject":"Remove deprecated pki related options","status":"MERGED","created":"2019-08-15 03:52:35.000000000","updated":"2019-08-16 11:06:54.000000000","submitted":"2019-08-16 11:06:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676570-1565953614239-3f7c88c0","meta_rev_id":"6d37626056fe5bf2e22597eae19f41ca80e693b7","_number":676570,"virtual_id_number":676570,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-08-15 06:55:21.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2019-08-16 11:06:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-15 07:43:25.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-08-15 16:23:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-08-16 08:14:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2019-08-16 08:14:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-15 07:43:25.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-15 16:23:30.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-16 08:14:47.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2019-08-16 11:06:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"472c9c4aa04542f593a7b6688314f223b9d494a5","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-08-15 03:52:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d4c292d01d9092588c112b88fda22b66fef93156","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-15 05:17:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/70/676570/1/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/9b09105/ : FAILURE in 1h 20m 12s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/70/676570/1/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/04cffdc/ : SUCCESS in 39m 14s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/70/676570/1/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/84bd912/ : SUCCESS in 36m 21s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/70/676570/1/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/6bc840a/ : SUCCESS in 28m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"da7dde8c6ad4709e3dc26df60758ce3c1312280d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 06:50:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/21956891691a439aa869e9864f52bfb9 : SUCCESS in 4m 15s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b30b15bceb9d4977b75f039212dd39eb : SUCCESS in 3m 58s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dffc526d1aa44932b4b14523bd0c2e33 : SUCCESS in 4m 41s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/1759322afea94b41a5819c56a32f8611 : SUCCESS in 28m 17s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e390de149a224afabea6b1a1b446b8a3 : SUCCESS in 16m 19s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/998762fcaf744761bea03cd87008cddf : FAILURE in 21m 26s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/cd88518161114e4da799b71571da2f08 : SUCCESS in 46m 15s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/a8dd66b997c84c2d9133c774e4610a0c : SUCCESS in 1h 05m 00s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/6871c1666b4a46bb8f67e067fc6298fb : FAILURE in 1h 11m 43s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/6b42bd4033974849b3958295c785e369 : SUCCESS in 42m 13s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/34bbc84fff874c569f71d84e9a9308df : FAILURE in 33m 22s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/92566f9575d542e4ace188e22969c341 : FAILURE in 30m 25s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8746ad09f9e44f37985a68f79a296966 : FAILURE in 24m 35s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/ea94d51d8e3647619ae295c676f42330 : FAILURE in 22m 48s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/84138bbf3eb54ceaa091a372d02b39e6 : SUCCESS in 15m 43s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f27be91f7a53459c903c14b29315b451 : FAILURE in 8m 55s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/146731056dd24c39b8c350b14858afe0 : SUCCESS in 3m 58s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/72eb52d279c640c387178bbf0ca009a9 : SUCCESS in 2h 51m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f88654f579d86a481ca6d1ee12f2f4878181a8c2","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-08-15 06:55:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3b475ea014b0c89285b4dcf66ffd264646b6a172","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-15 07:43:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/70/676570/1/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/c39c6cf/ : SUCCESS in 39m 51s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/70/676570/1/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/855d0cd/ : SUCCESS in 46m 12s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/70/676570/1/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/4831af5/ : SUCCESS in 36m 49s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/70/676570/1/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/cbbbad9/ : SUCCESS in 33m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6c3944bb694b7a385cda0621304ae941d11925ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 10:01:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e832aaee8b45409caff07516c06d65e4 : SUCCESS in 3m 54s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a5e1ec70220c4b89889d2c4832de7b13 : SUCCESS in 3m 49s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/91b0a44e9fed47d8b90a2c0792506248 : SUCCESS in 3m 53s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/7630001c03924a938aa6e0a0eae4304a : SUCCESS in 25m 01s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fe3b89b629dd4df4a5223d3ee91c5faa : SUCCESS in 20m 03s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/190fbbf7318c4ea9a978e09e5d1fbd21 : FAILURE in 18m 59s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/f0c66177fe1545e0a5cdb84811089dbb : SUCCESS in 47m 30s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/fa4807e321694f7e9778e502d5387ddc : SUCCESS in 1h 07m 43s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/b1ff7ede38f74c6a888d2ef7707904c9 : SUCCESS in 1h 01m 35s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/fe27a362d5b342788dafbf2578631eeb : SUCCESS in 43m 29s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/684b462dd2a54362bb04390a1377803a : FAILURE in 29m 53s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c5b9e48d526f4d33b5dcbb700881cd92 : FAILURE in 30m 01s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8b16d2ab270343beae7acac54d92a6f9 : FAILURE in 26m 37s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/68e82a69370f41eda9b0df626e8d4fef : FAILURE in 22m 56s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/5be5a9a01f85450f9d3cb593f4a57acd : SUCCESS in 16m 35s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0c228ad1c1164ebab711eac01a0af15d : FAILURE in 7m 31s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31ee41ff17a84079a66cc7b18da9ceb7 : SUCCESS in 3m 19s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d7512372e5be4f9c903bb2810cea6332 : SUCCESS in 2h 58m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"60ebcfcb9e647f68638a1d4328deb25ee96ecd0d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-15 16:23:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"10792311294c4c87d9e7b589666e4fb1c7387012","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-08-16 08:14:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c711e82d3825a261772bea5e7e3500495cb7b716","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 08:14:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"068686a4fabae13e38d983482c0bedb61b6d3982","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 11:06:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5b7c99540d11469096f34d5dcab34afe : SUCCESS in 4m 29s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a489bb2f8de142c29fe405bd262d0501 : SUCCESS in 4m 32s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eaded8a8c8f8408d8138a3e30c7aa216 : SUCCESS in 4m 20s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/2d56f0d3ba74409dbd66e9cea395f235 : SUCCESS in 26m 03s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ef48cd663b7d48e1ac438b470f0eb2b4 : SUCCESS in 17m 36s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/77d0afd1bab34269a648df6272d82303 : SUCCESS in 52m 25s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/2c3ef3ad02374d2a997caccc6bcca2bb : SUCCESS in 55m 07s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/30fa9a6f0c774f90878b9a7bf2eda0be : SUCCESS in 56m 15s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/6dca952db08146f7977c3b9172c82981 : SUCCESS in 43m 05s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/9a234a3ea57d4de594d7cfbb5affabb5 : SUCCESS in 17m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ecb2a8dc44dc4066bdc16c89b5abc2c5 : SUCCESS in 3m 53s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f152bb8490bf4020b69c72bcbf5aef13 : SUCCESS in 2h 50m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5de08c40522f558988cba66fe6b687dcbd0c6832","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 11:06:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"56a1080ee081da51cd3382be5061b27cf0ca6034","revisions":{"56a1080ee081da51cd3382be5061b27cf0ca6034":{"kind":"REWORK","_number":1,"created":"2019-08-15 03:52:35.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/70/676570/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/70/676570/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/70/676570/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/70/676570/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/70/676570/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/70/676570/1"}}},"commit":{"parents":[{"commit":"add8873b699f99f1ca9fdeb8094d44a829755217","subject":"Bump the openstackdocstheme extension to 1.20","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/add8873b699f99f1ca9fdeb8094d44a829755217"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-08-15 03:51:36.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-08-15 03:51:36.000000000","tz":480},"subject":"Remove deprecated pki related options","message":"Remove deprecated pki related options\n\nThe deprecated pki related options check_revocations_for_cached and\nhash_algorithms option has been removed.\n\nChange-Id: I6d4d6627804bd0718c3183c6cd419c12212c74ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/56a1080ee081da51cd3382be5061b27cf0ca6034"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/56a1080ee081da51cd3382be5061b27cf0ca6034"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
