)]}'
{"id":"openstack%2Fpuppet-neutron~679556","triplet_id":"openstack%2Fpuppet-neutron~master~I79a9bf7c07ca7c60e982a30b23308576daffc017","project":"openstack/puppet-neutron","branch":"master","topic":"multi","hashtags":[],"change_id":"I79a9bf7c07ca7c60e982a30b23308576daffc017","subject":"Add support for multi interface driver","status":"MERGED","created":"2019-09-01 07:11:50.000000000","updated":"2019-09-04 13:05:23.000000000","submitted":"2019-09-04 13:05:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679556-1567602323537-c2cdff65","meta_rev_id":"260e6db8bbf8e063653d45a094ddb9d4d43dcb0c","_number":679556,"virtual_id_number":679556,"owner":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2019-09-04 08:08:55.000000000","_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},{"value":0,"date":"2019-09-04 06:52:53.000000000","_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"value":2,"date":"2019-09-04 13:05:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-04 08:49:15.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-09-03 19:07:43.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},{"value":2,"date":"2019-09-03 15:45:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},{"value":0,"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-09-03 19:07:43.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},{"value":0,"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-01 07:12:38.000000000","updated_by":{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},"reviewer":{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},"state":"REVIEWER"},{"updated":"2019-09-03 15:45:18.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-09-03 19:07:43.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-09-04 06:52:53.000000000","updated_by":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"reviewer":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"state":"REVIEWER"},{"updated":"2019-09-04 08:49:15.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-04 13:05:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"36b487625731ce841bce89db9674124361c36ec1","author":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"date":"2019-09-01 07:11:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0b36cc706a7916575d9889b3f72d7067316c85e9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-01 07:20:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/56/679556/1/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/36cbbff/ : FAILURE in 7m 12s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/56/679556/1/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/e5ac530/ : FAILURE in 6m 59s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/56/679556/1/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/5fddda7/ : FAILURE in 6m 54s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/56/679556/1/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/cf10264/ : FAILURE in 7m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e9ce8a66b137aac73d078abcc7ac879cbeb6fd9b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-01 08:48:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6aa423a973fe41e68253773fd330b31c : FAILURE in 7m 44s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/29d3ddcedeb94d38926ab215826f070b : SUCCESS in 4m 23s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a9f68a0791674157ace10d053d97c854 : SUCCESS in 4m 53s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/247331e204b54db1b8b81a22efa25887 : SUCCESS in 32m 26s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d4fccdf521cd40dfb163b9284dcea85e : SUCCESS in 19m 31s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dd152b96159148e192f6b57242300b87 : FAILURE in 21m 20s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/ba03bfeb0dfa42e788e35cac47fa6cfd : FAILURE in 27m 24s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/100fea75972e483fbe1b1ced61621e2d : FAILURE in 27m 56s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/d3f72368f5fd44dbbf552a72260935c2 : FAILURE in 7m 30s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/0fc2bdc511424c0c91a92ccb1f8740df : FAILURE in 8m 56s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/e6999176bbd34f9985773a473491dc10 : FAILURE in 7m 40s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/54401d6a125b40ae88b8fca9d48e4554 : FAILURE in 7m 22s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/48e40746b6ee4ade86a95f19792335ed : FAILURE in 7m 39s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/96624cdc95234691befa8061f8311c83 : FAILURE in 7m 17s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/44a06a6f3be04032bf584b1b98efec26 : TIMED_OUT in 1h 31m 26s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3dc8759dcba446ffbd0a1a1a67a419a4 : FAILURE in 10m 49s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/efdfd9124a254acd8dcfd9a7d0ea6be0 : FAILURE in 46m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2a8c6322d8141d6ddf02710c05a8a619ddb10f9b","author":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"date":"2019-09-01 09:05:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b9cccb2c73c25e070e0936665789d6dec878e9a2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-01 09:15:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/56/679556/2/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/736c59d/ : FAILURE in 7m 18s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/56/679556/2/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/72391a6/ : FAILURE in 7m 14s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/56/679556/2/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/dec988d/ : FAILURE in 7m 10s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/56/679556/2/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/bda962d/ : FAILURE in 7m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bff6226d469dd0af3a812e30f6a2a429c1a47143","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-01 10:39:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bcc89d9cf317413b9fc275b77d4c74f7 : SUCCESS in 5m 00s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7366ed3d1a31477e9ea17211128de1f6 : SUCCESS in 5m 13s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/57a28af1eca740e5b930d9cadb366342 : SUCCESS in 5m 19s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/5d140226a90c4d15b9271ddeb04fa5bb : SUCCESS in 30m 55s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/30695a4767e548c595fb80d6683699e8 : SUCCESS in 16m 43s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/889b3a2ec59947c4954d2a844df59285 : FAILURE in 16m 17s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/78fd2b853ee644e29e6492effdd3c999 : FAILURE in 26m 30s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/cd8caf6daffa4c4398ab1a362cc2a996 : FAILURE in 22m 39s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/97bb54f185f84216adad3fe598783767 : FAILURE in 24m 42s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/63ed428dabbc4a0cbd1edbf5b258b5f3 : FAILURE in 24m 38s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/b3fea38d8caa4f039403bcdd605af8db : FAILURE in 42m 10s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/87d7f7375185495196ca63601544c918 : FAILURE in 7m 42s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/69550909562a43e68310751a6239a681 : FAILURE in 39m 18s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/58b9d21a47794c44ba0e0fc0199f2246 : FAILURE in 7m 25s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/25ca2b2a398743479cb2e61018a739dc : TIMED_OUT in 1h 31m 38s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/467e529674614d3991d4976e9d956d79 : FAILURE in 11m 08s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/372642d7c9b64e94a8280cd1db9b6b63 : FAILURE in 19m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"820492af9e8c6b857a3afb957710de35aceca285","author":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"date":"2019-09-02 06:23:09.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8a013edd04597e3a9b186ec5109c73a4158ed802","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-02 06:32:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/56/679556/2/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/92bf6ad/ : FAILURE in 7m 17s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/56/679556/2/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/b24279b/ : FAILURE in 7m 01s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/56/679556/2/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/663854b/ : FAILURE in 7m 09s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/56/679556/2/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/e2555fb/ : FAILURE in 7m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d566fcb7932c15701f258b2bd9dad44001e8ecbc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 07:07:13.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8dded7d76fb04217a4e9aaae5141b4bb : SUCCESS in 4m 58s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b2805a94517c4325be8d4037fb23cebf : SUCCESS in 4m 01s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/20c76e2ff54544538b9c786d4ed987dc : SUCCESS in 4m 57s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/2ee51051ece24c2182ff249b94bb3d1f : SUCCESS in 31m 12s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/047ca931a1c64ca6a39aa1d0f27e416d : SUCCESS in 19m 44s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/116c30e592214d8a8207256f538903c4 : FAILURE in 21m 06s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/4a30ab67063d490aae058519c1b5bb07 : FAILURE in 9m 06s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/1e7edc24728b4a8f891446b8017cb7ad : FAILURE in 8m 06s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/2b44f18f414d49d08d93f1185bf44bfb : FAILURE in 9m 51s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/f0f775f9118e4f509e986c2969662f1c : FAILURE in 8m 05s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/75fef16db233406e938872a202efe310 : FAILURE in 41m 00s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c80728e65e0e430b923e6bd359028260 : FAILURE in 8m 38s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/64a7f35215b94101841c2533201b14de : FAILURE in 7m 44s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/2c17e64fbc7a480ca4b83f18896e2d4e : FAILURE in 8m 33s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/ceab632914344801a364e93b411c4977 : FAILURE in 12m 30s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/852e889508874e3e826a9200aedd1128 : FAILURE in 9m 21s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/238d75c56db242a79c2bacb8c5d4e252 : FAILURE in 20m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5ed13a944881f9ab23bd97b044f5a774d55a920e","author":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"date":"2019-09-03 06:27:01.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ac4c8d5f7988c28e08732c2704f1dbf0c71fbca2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-03 06:32:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/56/679556/2/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/fa1d24c/ : FAILURE in 3m 26s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/56/679556/2/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/8fda8ce/ : FAILURE in 3m 57s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/56/679556/2/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/5f66963/ : FAILURE in 3m 57s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/56/679556/2/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/0dbdf2f/ : FAILURE in 3m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"33db397a649971756ab74ef8cd41fcaf57dec73a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 09:49:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/08b8f7e8b8f9411782eb929ba900ce51 : SUCCESS in 4m 42s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1ddbc789eede4efcbabcbdf111b24f75 : SUCCESS in 5m 34s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/756ce8b108124aa9ac48af0abda18364 : SUCCESS in 5m 46s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/443fc522e6ed4152abdba8ca2fd44cb2 : SUCCESS in 31m 11s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2985f630cbc843f28e0cc36b3ab9b3dc : SUCCESS in 20m 28s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/57c3336d489a4101bfc027d422ce4edc : FAILURE in 21m 34s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/bc8257db2ac6430a99c7f0536b03da89 : SUCCESS in 50m 07s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/21be02a851c44531898c087e67efdb7e : SUCCESS in 1h 00m 43s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/98e14406e0874703bb1247cb8325e6e0 : SUCCESS in 1h 07m 24s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/ed3192003ec540debd7c7acd92add464 : SUCCESS in 46m 38s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/883c23901b78448b8775acbece23f37a : FAILURE in 30m 23s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/958e095bd8294e189eb5a71c9c0a157b : FAILURE in 32m 22s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3338de518cf947cabedae91e53a74b11 : FAILURE in 24m 36s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/7fdfa4d3c5d544639216f508d2569314 : FAILURE in 23m 13s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/b0d81d1f869144cd8797e663aecffc09 : SUCCESS in 18m 43s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e7f0a54beb284555ac40b2fbe6398a33 : FAILURE in 9m 36s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4992e672d937479e8dda5ceaf272987b : TIMED_OUT in 3h 15m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ab52c4b7dc29e281f47a67cc949795d6bb1da407","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-09-03 15:45:18.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3b583424e8f3a63be6dcd826df4cd6cb0fa35264","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-03 19:07:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c136209a63baac0a45bd49bf322f668a91bf2b53","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 19:07:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3080afb17675451ce946b2afcdec2a1dccd343d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 22:51:04.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d9d2dcd62fe047d9971a4d3fbd6b0528 : SUCCESS in 5m 00s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/be187d20ebd6483e8abadf6dbd22c718 : SUCCESS in 6m 01s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d6c31da6f55a4389ab952cffdcd48d14 : SUCCESS in 4m 36s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/ee06cbf4d8074f879b50dd611f553024 : SUCCESS in 33m 18s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c7687c352bdc4083aaaafa23b03bd8d1 : SUCCESS in 20m 23s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/dd613f6f9dc94f22bb4afe62c2cc3add : SUCCESS in 51m 32s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/2b06f7d1b37942bd9a35dcd63b8623d5 : SUCCESS in 1h 01m 51s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/3e9a0cbd01a14598ac31e868c68d856d : FAILURE in 11m 26s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/612a3cfbf3c94dc6ae7bd1e9879aac37 : SUCCESS in 1h 09m 37s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/83df4b3aa2d841b48c9cd95d0f0be932 : SUCCESS in 17m 48s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d11660495cbb4b7f926dfd594a9a843f : TIMED_OUT in 3h 19m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a60ccfb47e9391e6b2355a1b7f41ddbc6dcce1d3","author":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"date":"2019-09-04 06:52:53.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"fd3ecdf443e929708847ecaf5351abc3cb879194","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-04 07:39:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/56/679556/2/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/ceb5814/ : SUCCESS in 32m 48s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/56/679556/2/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/53dae44/ : SUCCESS in 45m 35s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/56/679556/2/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/bda6848/ : SUCCESS in 35m 39s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/56/679556/2/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/2f507d3/ : SUCCESS in 38m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fe0440cc5560f3e42a3af8d2dcbcd86027e827a0","author":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"date":"2019-09-04 08:08:55.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a34f12604d010e5a2ddf331cbc3f04256c6f112f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-04 08:49:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/56/679556/2/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/4e942ca/ : SUCCESS in 38m 49s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/56/679556/2/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/b9a2886/ : SUCCESS in 34m 50s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/56/679556/2/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/e24656c/ : FAILURE in 34m 51s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/56/679556/2/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/7e8dc23/ : SUCCESS in 29m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6b6443bec22f1916324f9f3123e4603842e7a79e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 10:05:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/829614551b9445e19634ee64342a3436 : SUCCESS in 6m 41s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/81cdd52dd5dc41f4840acd8f9f6fa717 : SUCCESS in 5m 02s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9e53cc82f3d14750b548c0f62b933952 : SUCCESS in 4m 35s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/319d8ea8974f41ecbd86d43ead93c511 : SUCCESS in 26m 30s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6cf4b3c5a1224b95b095feadbda6cabd : SUCCESS in 17m 06s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c90e7c6dc5984f09830369849b0849d0 : FAILURE in 17m 14s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/f8ebff251a8047cdba18d525501aac70 : SUCCESS in 54m 19s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/ae59fa27ca3a4e7c97541dd2fafaf8e6 : SUCCESS in 1h 03m 34s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/19c4331425bb4888a35dcd648af0d74b : SUCCESS in 1h 11m 09s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/388e41b7d011457ebc9d8edcf70f043a : SUCCESS in 48m 46s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/f7e2feba578e48f7b800eca5fb03c323 : FAILURE in 29m 29s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/14fadd8f126d4446a881871d78ed088e : FAILURE in 31m 37s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b06920c2758a4f118f2669b5c0202e21 : FAILURE in 32m 04s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/c157fbd2630d41a89447d93490aceb90 : FAILURE in 23m 07s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/2e8eff35713e46e2b7017066dc5cc224 : SUCCESS in 21m 08s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b9435c391aa94110b705bca4cd14e344 : FAILURE in 10m 12s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5fcd7e72dea14369bea752836aef6c0d : SUCCESS in 3h 07m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4884426d52b085054530eee4b28197afb459ee18","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 10:05:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d9e476dcc418cdd3c4ad50f79c33534caf28e54f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 13:05:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a089e858553556203c22f81089e526d2d3f3fee5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 13:05:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a2102ce1188747749ddef2ee1e80866a : SUCCESS in 4m 39s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0655f1d19d264ad685817f2737d50631 : SUCCESS in 4m 35s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dac6b8e4d2784effba93615df924be3e : SUCCESS in 4m 33s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/9f644669217842938637360e46c7064b : SUCCESS in 32m 12s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/59e2386d7b1549f6b939fcd2ee54e394 : SUCCESS in 16m 37s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/c2cc4a8269bb44fcbb4da2dfc4ee9718 : SUCCESS in 49m 36s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/7d1915c9f5494f32ad05a45b1a595993 : SUCCESS in 58m 21s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/214cd8cac0c443bda677925d3271a01d : SUCCESS in 1h 06m 46s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/e686235add9e4cceb1f8e6423aae68b2 : SUCCESS in 44m 19s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/2aac55ddab044337be504c37c21179be : SUCCESS in 18m 42s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/116fa72891244a578305b981806140ea : SUCCESS in 2h 57m 06s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b2e7de79df98a986768aaef203ddbe02bee41072","revisions":{"ef8444403e512663698ac0ea5e4bae77811dddf0":{"kind":"REWORK","_number":1,"created":"2019-09-01 07:11:50.000000000","uploader":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"ref":"refs/changes/56/679556/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/56/679556/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/56/679556/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/56/679556/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/56/679556/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/56/679556/1"}}},"commit":{"parents":[{"commit":"640df7320c165d1a6366e043c437ef20b630217d","subject":"Merge \"[Follow Up] Add support for neutron-mlnx-agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/640df7320c165d1a6366e043c437ef20b630217d"}]}],"author":{"name":"Hamdy Khader","email":"hamdyk@mellanox.com","date":"2019-09-01 07:05:01.000000000","tz":180},"committer":{"name":"Hamdy Khader","email":"hamdyk@mellanox.com","date":"2019-09-01 07:11:48.000000000","tz":180},"subject":"Add support for multi interface driver","message":"Add support for multi interface driver\n\nChange-Id: I79a9bf7c07ca7c60e982a30b23308576daffc017\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/ef8444403e512663698ac0ea5e4bae77811dddf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/ef8444403e512663698ac0ea5e4bae77811dddf0"}]},"branch":"refs/heads/master"},"b2e7de79df98a986768aaef203ddbe02bee41072":{"kind":"REWORK","_number":2,"created":"2019-09-01 09:05:50.000000000","uploader":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"ref":"refs/changes/56/679556/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/56/679556/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/56/679556/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/56/679556/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/56/679556/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/56/679556/2"}}},"commit":{"parents":[{"commit":"640df7320c165d1a6366e043c437ef20b630217d","subject":"Merge \"[Follow Up] Add support for neutron-mlnx-agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/640df7320c165d1a6366e043c437ef20b630217d"}]}],"author":{"name":"Hamdy Khader","email":"hamdyk@mellanox.com","date":"2019-09-01 07:05:01.000000000","tz":180},"committer":{"name":"Hamdy Khader","email":"hamdyk@mellanox.com","date":"2019-09-01 09:06:37.000000000","tz":180},"subject":"Add support for multi interface driver","message":"Add support for multi interface driver\n\nChange-Id: I79a9bf7c07ca7c60e982a30b23308576daffc017\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b2e7de79df98a986768aaef203ddbe02bee41072"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b2e7de79df98a986768aaef203ddbe02bee41072"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
