)]}'
{"id":"openstack%2Fpuppet-neutron~686240","triplet_id":"openstack%2Fpuppet-neutron~stable%2Fqueens~I9ed539745a705936d9a5110a9cfb05c2f28b0bbb","project":"openstack/puppet-neutron","branch":"stable/queens","topic":"bug/1832758-stable/queens","hashtags":[],"change_id":"I9ed539745a705936d9a5110a9cfb05c2f28b0bbb","subject":"Add support for configuring permitted_ethertypes on OVS agent","status":"MERGED","created":"2019-10-02 18:36:11.000000000","updated":"2019-10-11 15:03:47.000000000","submitted":"2019-10-11 15:02:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686240-1570806127319-29eb8155","meta_rev_id":"a565658e7fa6bc73c14fb9420cc5fad6a94eb34b","_number":686240,"virtual_id_number":686240,"owner":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-11 15:02:07.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":2,"date":"2019-10-08 13:53:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-10-11 13:45:09.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2019-10-02 18:36:11.000000000","_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-02 18:36:11.000000000","updated_by":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"reviewer":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"state":"REVIEWER"},{"updated":"2019-10-08 13:53:50.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-10-11 13:45:09.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-10-11 15:02:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ce7183094b3369896fa47a6aab841c60bb091ea8","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-10-02 18:36:11.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"0acf5a360270ce26e54390d348bca57bc9ea6936","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 20:14:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/e5ecb7b04be64c2a89cd4e941d1723da : SUCCESS in 3m 52s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/2e800e66b1ba47d0bbb31c069c8c9e8e : SUCCESS in 4m 17s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/7aa642db62434884a71bdb959122da7e : FAILURE in 4m 56s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/71bb88f278c440639dafd21d8a82b657 : FAILURE in 9m 33s\n- puppet-openstack-integration-4-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/a78df5d5f7b34127b4c9c403ec42070a : SUCCESS in 58m 54s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/df24eb2b004f4bb2805537444c074cd0 : SUCCESS in 1h 14m 37s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/cd263dda790f43589e08b5d82e3b5158 : SUCCESS in 1h 12m 26s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/b58ced1c4f2a4b52b0a45cd41fa004d4 : SUCCESS in 44m 32s\n- puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial-luminous https://zuul.opendev.org/t/openstack/build/6ed3bd11905e4b0e9b112bb5250e4679 : TIMED_OUT in 1h 30m 55s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/368efb41d4a34d478aa24d712638e5b9 : TIMED_OUT in 1h 31m 01s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/85319208be2e440dada7a6b01711a62c : TIMED_OUT in 1h 30m 52s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial-luminous https://zuul.opendev.org/t/openstack/build/b6ad25584d86499895109028e527020e : TIMED_OUT in 1h 30m 53s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/089bfc7fa1ef44869b8cd749853e055e : SUCCESS in 19m 27s\n- puppet-openstack-beaker-xenial https://zuul.opendev.org/t/openstack/build/5d758cd42a7f4671abcb470b2b93a9af : SUCCESS in 15m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5e584624c904a898c91e433b37c0348 : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":1},{"id":"36d7c582bb640e0a9868a398015cb8f48307a1c1","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-10-07 21:03:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6d07ed1db7c164ab46e8267c599053103dd070ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 22:42:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/747056a13b6444c4a8e4e9b21f3a3046 : SUCCESS in 4m 45s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/b9bed85a72194ae79f6a5c2ae597f78f : SUCCESS in 4m 05s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/b241acb685f240f29059b80bc647e5d9 : SUCCESS in 26m 40s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/45efc55baedd4cfdbe19e29ab0a0a5e3 : SUCCESS in 20m 31s\n- puppet-openstack-integration-4-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/19b5ff9c0b564c94a08713a79a4e88a5 : SUCCESS in 1h 02m 02s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/94fc820bad59487c800ef7a25c4dfcc1 : SUCCESS in 1h 04m 41s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/402092d798574fcd87f8b3f538deae3b : SUCCESS in 1h 00m 39s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/3183beedf8654b189c9de0a7c4bc40e2 : SUCCESS in 47m 04s\n- puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial-luminous https://zuul.opendev.org/t/openstack/build/f94bc269fe4a4a4b8414454894f27e08 : TIMED_OUT in 1h 30m 50s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/97ac00c2b56c4540b64d9707b30a70e6 : TIMED_OUT in 1h 31m 13s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/02bf30e711c74d91b0cee10c2753fd48 : TIMED_OUT in 1h 31m 04s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial-luminous https://zuul.opendev.org/t/openstack/build/8cdba7a85765496baa50440302d26231 : TIMED_OUT in 1h 30m 53s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/3c0eaee41ac44a099be6b54838558e43 : SUCCESS in 19m 16s\n- puppet-openstack-beaker-xenial https://zuul.opendev.org/t/openstack/build/011faba5348f40ceb45832275127888d : SUCCESS in 16m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70b7fac513e6489e8d4e14c6d4761b8c : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":1},{"id":"d00f272cb6e4663143019597881974518b520d3b","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-10-08 13:53:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7f92b8b7af1e43d5ba0b30c6a05a13bcba7537ee","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-10-11 13:45:09.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"799a4a6bc8084e0b08e222d26ff03e1cfb60d47a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 13:45:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c23e8a1bb443fb23b936ffac84f32d83e6354b65","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 15:02:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"67d73a0b22b1ac97cc16d60607472b30af77f319","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 15:02:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/40bd5ffe9e0b4773803f558b306b8fd5 : SUCCESS in 4m 14s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/f7ee6a73345e426e9e9d172ba7be6931 : SUCCESS in 4m 16s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/263236e5d7904e64bc18a7a815ca86d8 : SUCCESS in 29m 53s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/3fdf039af1d841a5b3cb44d2b51e7805 : SUCCESS in 25m 47s\n- puppet-openstack-integration-4-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/adb1d9bc47874319a9639b4f551882b5 : SUCCESS in 1h 02m 46s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/8e447a72393049b4aa6819985562462c : SUCCESS in 58m 47s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/561fb59c617e4f2fa47d34e18db943e2 : SUCCESS in 1h 03m 56s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/78158a935e1c432ea05c1b22ba9b3fe3 : SUCCESS in 47m 36s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/378cb9bd3a6048a0983f2635edb58e4e : SUCCESS in 18m 08s\n- puppet-openstack-beaker-xenial https://zuul.opendev.org/t/openstack/build/f461b9c5d1694fc9ae5a9d08b673d039 : SUCCESS in 15m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6b38a3714164187b2682b732f3a7f9a : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":1},{"id":"999e171a6ccc7d26b48b07ccc5666b15c0d2a60d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 15:03:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f32f75980f447029a0d25af64f654e1 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bedb9ca987d0678c8ae271c9d54c3e2829394f6c","revisions":{"bedb9ca987d0678c8ae271c9d54c3e2829394f6c":{"kind":"REWORK","_number":1,"created":"2019-10-02 18:36:11.000000000","uploader":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"ref":"refs/changes/40/686240/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/40/686240/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/40/686240/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/40/686240/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/40/686240/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/40/686240/1"}}},"commit":{"parents":[{"commit":"fedd77e759f42dd086975e285f2c7bcac419d939","subject":"Ensure tunnel types are absent if not configured","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/fedd77e759f42dd086975e285f2c7bcac419d939"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-07-09 19:14:21.000000000","tz":0},"committer":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2019-10-02 18:36:11.000000000","tz":0},"subject":"Add support for configuring permitted_ethertypes on OVS agent","message":"Add support for configuring permitted_ethertypes on OVS agent\n\nNeutron recently added support for configuring non-IP ethernet types to\nbe allowed through firewalls through SECURITYGROUP/permitted_ethertypes.\nThis adds support for configuring this feature on the ML2/OVS agent.\nSee https://review.opendev.org/#/c/668224 for related neutron change.\n\nConflicts:\n  spec/classes/neutron_agents_ml2_ovs_spec.rb\n\nRelated-Bug: #1832758\nChange-Id: I9ed539745a705936d9a5110a9cfb05c2f28b0bbb\n(cherry picked from commit def07975c604ecc3215fa17dc10fc0ac8fa40545)\n(cherry picked from commit 853c5a1568face1b12b115ece4bfbc4253cd67e6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/bedb9ca987d0678c8ae271c9d54c3e2829394f6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/bedb9ca987d0678c8ae271c9d54c3e2829394f6c"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
