)]}'
{"id":"openstack%2Fpuppet-neutron~687717","triplet_id":"openstack%2Fpuppet-neutron~stable%2Ftrain~I930582694aabc9f6c800cca9cf528259e1d29198","project":"openstack/puppet-neutron","branch":"stable/train","hashtags":[],"change_id":"I930582694aabc9f6c800cca9cf528259e1d29198","subject":"Switch to Train","status":"MERGED","created":"2019-10-09 23:10:13.000000000","updated":"2019-10-16 16:00:50.000000000","submitted":"2019-10-16 16:00:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"687717-1571241650081-4f81ad49","meta_rev_id":"23fc1b603a231c1cff97b70255d75078754eb3a3","_number":687717,"virtual_id_number":687717,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"date":"2019-10-11 22:16:46.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-16 16:00:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-10-10 19:15:46.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-10-11 05:46:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":1,"date":"2019-10-11 12:32:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-10 19:15:46.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-10-11 05:46:01.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2019-10-11 22:16:46.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-10-16 16:00:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0edcf7c35524853f07906f1b7a796634b214c237","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-10-09 23:10:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d91238062b4336512378d46b5ac22451580eccdf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 01:52:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7d626c39140543f69fc90a748d8845f6 : SUCCESS in 4m 55s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/132b2d0525e34b1a9db25eaf23a7bc27 : SUCCESS in 4m 30s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/c8f3443a0a1549a9be4657f92be602e8 : SUCCESS in 32m 15s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/9cd2d6403103475dbc454feeb64202d1 : SUCCESS in 51m 42s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/1a2d52e29e4f4d5992d7e2f5cc0c8160 : SUCCESS in 1h 00m 24s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/4d68d7f762f649cc88da7c0d0c916662 : SUCCESS in 1h 06m 22s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/e7cab3e887d14bb7a5e9aa902dd11fc7 : SUCCESS in 48m 23s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/7fa3223bc58e4c9aa588c6bfe69e8bd9 : FAILURE in 34m 15s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ab052905424d463fbfa6d3e6eba3158f : FAILURE in 30m 25s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fcb440899dd14d66a5989729b99f6eea : FAILURE in 20m 53s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/a3b7fec9a39b4319a276b6375913a145 : FAILURE in 35m 36s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/8fbb6a5bb4aa4d1b83f5c56cb6314ae0 : FAILURE in 8m 05s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ef84b6c3162a432b892b81a0198be74b : FAILURE in 13m 18s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4ebcac525be74af4830aac9113ec57e1 : FAILURE in 2h 40m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b35dae424d03c30030d3f4640767beda570ce25f","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-10-10 14:47:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a0006a0248a45e7da71b9205f202495d07f27a7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 18:19:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d665391a29b840ceb5592ed55069b152 : SUCCESS in 4m 23s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/459e59d6dca94ebda4060fedcc37bdec : SUCCESS in 4m 49s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/e495a7fc48c9452eab95a74331c47b98 : SUCCESS in 25m 38s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/062101fe89cf49a888de7da7952290eb : SUCCESS in 1h 01m 41s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/6cd0df16d5eb4d999e799e3d896c05d8 : FAILURE in 1h 11m 22s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/c7601427aa4041b9abd861eb5538f094 : SUCCESS in 1h 10m 55s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/dada50c5403e439bb77ee6bfb938142d : SUCCESS in 44m 04s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/26323e1ca78c4491b29ac17ec3fb0d1e : FAILURE in 52m 04s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/200319a472ee44229a68707f3480013b : FAILURE in 32m 37s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/795ffda0ecd34470bda746e1de6d8424 : FAILURE in 30m 05s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/f06799ad8d3f4e2cbc6f0e57b0a31582 : FAILURE in 33m 19s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/fe2f2e05830a48f184254b909613e4c5 : SUCCESS in 21m 52s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7ac90f63c4a1495cb1dc0f8a401f485e : FAILURE in 9m 46s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/eca260ca8d2048098ada0b380ac7a720 : TIMED_OUT in 3h 24m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"357e1c57ea692a4bb95f8a17df13955e69be1a31","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-10-10 19:15:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fe8a95df7fee3eb1a1ca781b17251c94c37539d0","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-10-10 23:20:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"16f15a42b2704df36746d3cb3b6af7fea48f60a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 02:30:53.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/10d6dab5f3364023a8c91bdfbfab6457 : SUCCESS in 4m 32s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dcf9bc8d357046f8aee848533e3a0a3a : SUCCESS in 4m 14s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/9cd7310ef791463cae51853032eedfe5 : SUCCESS in 31m 53s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/53904b74c2864583b6950bd87598ae04 : SUCCESS in 52m 34s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/0879e9c2e554423591c947af521f92d9 : FAILURE in 1h 05m 15s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/d354d30105a645449dbc7c5a706e40c5 : SUCCESS in 55m 28s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/c8767618e83b4a97891cd2225d25f23b : SUCCESS in 52m 03s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/6f7c75b1f9aa4d6498ecf224a22fd3ae : FAILURE in 29m 25s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2493c1f4abc549e4a35d2a619ecf7075 : FAILURE in 29m 07s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0f7a144fad6f43c9b4eec069bbf62e39 : FAILURE in 20m 05s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/15fe1a0cd81a4951a6422e3092be275f : FAILURE in 21m 46s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/f61f25e98b674e74ac672dd35124bb2c : SUCCESS in 18m 52s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/487e6734a38d44a296edad3bad827030 : FAILURE in 9m 12s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8ca4b33006174ae2bc1e3b4775da4736 : SUCCESS in 3h 05m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"40eb9dd3328a00377c21e0d18a288fdc6d7e733a","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-10-11 05:46:01.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e9bbb09d9a1fcb42914c50a967414c122e18a4dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 09:20:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7968f8fbdf1048e78fb2b8085d1a5763 : SUCCESS in 4m 36s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8d0cf093478043ea9a8486c36bc107d1 : SUCCESS in 4m 46s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/a361f0eddddc477a9181f4dcfc9013f9 : SUCCESS in 28m 09s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/689b9e9b9a944a7bb7791557d96a171e : FAILURE in 1h 12m 05s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/c29d7ac87e2646a7a12c6ca592bf99a3 : SUCCESS in 1h 11m 26s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/3ef8c0dca8214e61bcaf0c357f04083b : SUCCESS in 1h 05m 04s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/26a9c039418a4ad2969c37722d61fa8b : SUCCESS in 59m 05s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/17e53c1ab46c4121bcbfc7eb7cd1bd1d : FAILURE in 31m 50s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2ead4454d6294a05ab6ff10497884be8 : FAILURE in 40m 22s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/353b2765dd604287baa634f1fe334b98 : FAILURE in 23m 33s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/302c1fbc2c524b298c8f47a0b114fcf8 : FAILURE in 31m 59s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/fe6fa72830ad4371a0aa318b3ec5e766 : SUCCESS in 18m 31s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8167268a44ee4238be07eeeb61ba8faf : FAILURE in 13m 30s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ccd975adf9ae43c285b2d6a51bbb8b5a : SUCCESS in 3h 19m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9b5f99ac3b8fdea0eea95f71ac3aaf66132cdad4","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-10-11 09:23:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b5d0bbf5fdef578f53f4475479a0bd51625b0db1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 12:31:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f78ff1af6b424ef6a298c2d5c09838fa : SUCCESS in 4m 15s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d66f03fd99d74fde9139134d1bf3b8f3 : SUCCESS in 3m 44s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/ffc239797906481d94c30bd601195ea1 : SUCCESS in 32m 15s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/3d4660866a1044df8614d553383fcebb : SUCCESS in 50m 41s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/6ee041104da24eb5b57ad212858aa8e2 : SUCCESS in 1h 06m 38s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/804c837f19a54f3eba27f3709a5f436b : SUCCESS in 1h 00m 39s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/353e9846b63d4b3ba433c494f7b37e9c : SUCCESS in 50m 20s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/ac47f16425c644aca4925f72138068e6 : FAILURE in 29m 20s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/324f3ac19fe24ebe8dbeb88ea6405dbc : FAILURE in 32m 04s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7af57a9266724a1a911d55cec3eef7f3 : FAILURE in 21m 16s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/1fda59dfee0a4bb18d745aa502dbb44a : FAILURE in 21m 19s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/5fd970616b1049dd8b496bc79a26fe9f : SUCCESS in 16m 36s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/01562b69b7da421d8889524d2edd00c9 : FAILURE in 9m 30s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e699b1af31d3442f9f554a4b5b9a879e : SUCCESS in 3h 03m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3cf6092631b11beadefef7cabdadfe61ecaab4fb","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-10-11 12:32:12.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e1d6c7ec287a2bd7b87cc1b61acb55defd7babd9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 12:32:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cee07e2a09e2db23c026f596cbc391342e8c3052","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 15:49:07.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e746d8debd954efdbdbca82273f37770 : SUCCESS in 4m 52s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a313d49aef944fcf92a87acbaf70b993 : SUCCESS in 5m 16s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/d5f981dda7914852bb2b2ed415051579 : SUCCESS in 29m 53s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/c5aa796d1752440294e3fc8b3441b5ca : SUCCESS in 1h 11m 19s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/0ba1d24343da45649e62ceb22ce3dc5a : FAILURE in 1h 20m 15s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/b2991ce0fb6943e7ad00a0b552ebf997 : FAILURE in 1h 26m 03s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/3248d7a175494d468a94a5b00c0ab662 : SUCCESS in 47m 26s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/37bda7cb6ae64fee9700a4dcac1e62ac : SUCCESS in 27m 54s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e063b7e219174d86b06689e867103428 : SUCCESS in 3h 14m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f79c5f06ca912809cdd25395b847e1a8a167d65a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-10-11 22:16:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"51e38490022aeea7652a1dfe8521b43db927ee7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-12 01:42:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4d144a73f1214bbdb348c0fb30cd511b : SUCCESS in 8m 27s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3596135801f749c08a9b91584b4694c4 : SUCCESS in 6m 06s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/880c43aa5cf3419aacda6fcd454c3ed2 : SUCCESS in 26m 09s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/4e8eb2e103c449cfbc724c75059413d4 : SUCCESS in 1h 27m 58s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/f9d5f02b79db4621b2ec6c379e20232d : SUCCESS in 1h 05m 44s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/9c341c5f8295439ab07ed40342493bbb : FAILURE in 1h 06m 19s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/818aa41d4b6e45edaca0b0ecae2a9346 : SUCCESS in 45m 08s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/402fa16c360641b0ab1a89e4ab345ba6 : FAILURE in 30m 56s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cabfa147b14c462f977da885de3a11cf : FAILURE in 37m 09s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bd03de6150254da08d3f30e82ac45f7f : FAILURE in 23m 58s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/f65a4938c0eb489f8f4a857c82a45b6d : FAILURE in 27m 12s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/fe31bb783b87451183478aa9ce8ee238 : SUCCESS in 23m 47s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a1d8969e421c4b49b25104beab8469bf : FAILURE in 12m 45s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4f86a6db2c1e483ab536f8f49c9c960a : TIMED_OUT in 3h 24m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ef3a051684d9aeb0730737b79704adfbe425cfe2","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-10-13 15:32:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6ebdb93045f792421c8e6c2f1d1806bb9c3de57c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-13 18:20:37.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3b8594b8da0d4b0c9aae2e02984a512a : SUCCESS in 4m 43s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fb5b9d4df9984ca99f0b4ed7a5fd4712 : SUCCESS in 3m 33s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/e429a50a1ed94a1f8b54c25ac5e681e6 : FAILURE in 10m 22s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/a6c63ac54864485a9a9ab1a28fe7c02c : SUCCESS in 1h 01m 50s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/0779ad1f01cb4969a61e670f22abc9c8 : SUCCESS in 56m 10s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/e7d70061f2794aa1be8cd14675b9ea48 : SUCCESS in 59m 43s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/289d718be1e645479246a04e08ff55d9 : SUCCESS in 38m 54s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/a761b5238edb47a3b573b7d60cb849fb : FAILURE in 46m 31s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a270dd0a934d45cba9ac2a6eca6ed952 : FAILURE in 45m 37s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/168d76fbdecd4fa7a0d6840b5d128c5b : FAILURE in 35m 57s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/bd072806d2784d8fa19d0214896b451c : FAILURE in 20m 59s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/6909d674a21d4c35bea93c521833445f : FAILURE in 5m 33s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/da74da2e05da48a4a8e835dfa7cb15c3 : FAILURE in 7m 32s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4e95189a797749178a240b19c8ecfb71 : SUCCESS in 2h 46m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1322e2dabdf56f669c6fd6d67d5a146cd80d1c54","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-10-14 01:17:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d4d6ed3292257c835cefd7e784268573cfcc77f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 04:33:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/46753e2b51dd487698ff0ec62436e8b7 : SUCCESS in 4m 17s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f61e0148b1c54398a88d063d3a98d8ed : SUCCESS in 3m 52s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/1aee49e04dc44939b10330ef33d6abda : FAILURE in 7m 30s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/3498141de34d493e85c3b72bf72bd768 : SUCCESS in 57m 49s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/6d8b517d006240ff8df6f7a1c78760fb : FAILURE in 1h 02m 58s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/00f6d4d2d3b44dd39993da11f084e11c : SUCCESS in 1h 08m 36s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/7b05fe47a3e94f848ba5ab930bd252fd : SUCCESS in 45m 16s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/54cd46c69c79443591bdaf4db0eb33c9 : FAILURE in 31m 29s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4a827513bfcf4dd58fb6a1fabbd36fb2 : FAILURE in 31m 01s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b1c29e6acff345d1ba649ee74306e31d : FAILURE in 22m 15s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/da5b043e79334c719dbdebbaf40df059 : FAILURE in 22m 53s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/54867d24278b45ab95ee46515745741d : FAILURE in 7m 40s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/17ba005837db4bd5bce31c0ba9535652 : FAILURE in 8m 59s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6465e0597eeb4c6d93abdc244e5e201d : SUCCESS in 3h 10m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6e451850be9922469da85cb0f1565b62ec565f9c","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-10-15 05:52:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4378a0061458a3643c9d8032f5d6630ff5e1f43f","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-10-15 07:35:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c6a1b529da23c9564158589c02601a6dde027154","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 09:13:26.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e0dd5f0acdab4aeebd19181800836db9 : SUCCESS in 4m 04s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/05cf66dd389d4bb4a3f57e9f8a582cff : SUCCESS in 4m 02s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/0865178e9373411a86aac41aff80da40 : SUCCESS in 33m 35s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/8c45b9b32ada4ec59ca624c24330555e : SUCCESS in 53m 54s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/9a9c50c7cf494ff08271b4314a78d3f7 : SUCCESS in 58m 06s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/76d4811ecf574fc7a6dc4aa06a1181bb : FAILURE in 1h 11m 41s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/a19d2e707527440e8d17a4e46d65bd0e : SUCCESS in 50m 01s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/fc295edc897445aa827cc6f4d8ec8f1f : FAILURE in 29m 57s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4420ccee301142d897ef6ff1eb92ff7a : FAILURE in 33m 58s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9fd568e9b1a24aaf969e8e699b623064 : FAILURE in 21m 30s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/fc124922a119433db00f59a14feabbbd : FAILURE in 20m 33s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/e47e9f4abd7147d1b37b7adc4fd0c190 : SUCCESS in 18m 21s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/602b670e89794a7b905121680bd5732b : FAILURE in 9m 15s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/803f4c3f31fa4474b5928f90bae7e32f : SUCCESS in 3h 19m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"344cd99cf8286c70c2a81eb24d8a2587eb125975","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-10-16 09:33:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5edc28129d2a4bb11447572a8299a0f1b1668359","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 12:49:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3ab811bdf20e4c75807ed3193d8f5376 : SUCCESS in 4m 06s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/44a812a5e8b54e82a916362093b5d39a : SUCCESS in 5m 23s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/03872bda84ce467aa3dffe54156885d1 : SUCCESS in 35m 41s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/5b1b8544112c488c94a7803f27ce7ca3 : SUCCESS in 1h 05m 53s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/61956c28ce4843acb93ca7860fc96ac8 : SUCCESS in 56m 56s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/bc69774bb8e440b8bb4306745ad21c3e : SUCCESS in 1h 08m 01s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/67719a302a2c4dfb8b0ba1899e2aa61b : SUCCESS in 50m 01s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/fa1754d88bef4c208a804d59ac196e4d : FAILURE in 36m 01s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fc78f57255cf491d834a60786209a2dc : FAILURE in 31m 25s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/41fa06968d7947c4842c6158e45f796e : FAILURE in 21m 15s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/83bacd85b35f442f9e894aadc56b9ffa : FAILURE in 34m 58s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/a9520cdf444b401f981c4be7afd590ca : SUCCESS in 23m 51s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6afd83c0ae53482bb0f49943fbb23e33 : FAILURE in 10m 00s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/1188968ad48642c3b17ee2ad2a346bdc : SUCCESS in 3h 05m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"771803886410f9620422083264f6edbeff47343c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 12:49:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e5830c8075af155a0d0b02113faf78728eae59c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 16:00:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fff1661265ff46fc9f751787f573332d : SUCCESS in 4m 40s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cbb8e2379f524b83bf598a6e69d52516 : SUCCESS in 4m 42s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/262d572a4dd94af68302febafd9dd8d3 : SUCCESS in 33m 12s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/9ba8b6ba72df4dfa85c23dab1df6c09f : SUCCESS in 1h 02m 13s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/ce56076298364d3589e45df57edc127b : SUCCESS in 58m 56s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/e1bb52530d0d4fd78060e2b113149bfd : SUCCESS in 1h 35m 27s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/534068deca21459fb2138b80314ac48b : SUCCESS in 1h 02m 26s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/cf0f51dee5864331a4be4d733c3b66b3 : SUCCESS in 21m 54s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/78767bee485847ad9b4cf7f290db4ac1 : SUCCESS in 3h 04m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"23fc1b603a231c1cff97b70255d75078754eb3a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 16:00:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3dc0585e05e79bafb4beaa66d16d251e22fb05f5","revisions":{"3dc0585e05e79bafb4beaa66d16d251e22fb05f5":{"kind":"REWORK","_number":1,"created":"2019-10-09 23:10:13.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/17/687717/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/17/687717/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/17/687717/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/17/687717/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/17/687717/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/17/687717/1"}}},"commit":{"parents":[{"commit":"6e02338857782e2383e14c2e3e8365d57bdb0a24","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/6e02338857782e2383e14c2e3e8365d57bdb0a24"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-10-09 23:10:04.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-10-09 23:10:04.000000000","tz":120},"subject":"Switch to Train","message":"Switch to Train\n\nChange-Id: I930582694aabc9f6c800cca9cf528259e1d29198\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/3dc0585e05e79bafb4beaa66d16d251e22fb05f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/3dc0585e05e79bafb4beaa66d16d251e22fb05f5"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
