)]}'
{"id":"openstack%2Fpuppet-neutron~691793","triplet_id":"openstack%2Fpuppet-neutron~master~I9c9b681d356b11ff4bafeb37c7c19459bb99cf2d","project":"openstack/puppet-neutron","branch":"master","topic":"bug/1850236","hashtags":[],"change_id":"I9c9b681d356b11ff4bafeb37c7c19459bb99cf2d","subject":"Fix rabbitmq dependency","status":"MERGED","created":"2019-10-29 01:32:18.000000000","updated":"2020-04-03 08:36:00.000000000","submitted":"2019-10-29 01:59:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691793-1572314394133-01639688","meta_rev_id":"af95acc5c9db0fcaac9f9242c59bc6c9cfb88680","_number":691793,"virtual_id_number":691793,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-29 01:59:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2019-10-29 01:51:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2019-10-29 01:51:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-29 01:59:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"099b4f87fe26a8051afc350597ea5f1d68a14e03","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-10-29 01:32:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1adf5e392fa9b6704c2d917b6ea69389d1d89865","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 01:41:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/b9ce80f3785146c48bcf5810e273b68f : SUCCESS in 1m 51s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d653c0125f9249f3a3122796bdecfb00 : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":1},{"id":"a0ee7b5c677eb65d66e20ff1bf7aeeee008da0d2","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-10-29 01:51:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c051f8429bcff2397195db67cf6c500455b7ee40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 01:51:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a2b79538fa0d53e8544a73903b2a38fb5cb54b37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 01:59:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/a00f38285d604eee9441cea4f05d7c8f : SUCCESS in 1m 57s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c313b5d88e44412c940514d41a2a8042 : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":1},{"id":"4d44f7785b94c8008768a4d2a0859feb18f655b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 01:59:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d3f3c5ebc2ba91ec87e30ef43186358e4f7f3258","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-03 08:36:00.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 5e41b95ec01a5ca4dc643bd12335621afa2a8c45","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"29f5049be21b57b32de8290935774bbe03a51dbc","revisions":{"29f5049be21b57b32de8290935774bbe03a51dbc":{"kind":"REWORK","_number":1,"created":"2019-10-29 01:32:18.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/93/691793/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/93/691793/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/93/691793/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/93/691793/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/93/691793/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/93/691793/1"}}},"commit":{"parents":[{"commit":"3478011d64fcf57fa8d034f10621ce48cb5aff88","subject":"OVN: Add ovn_emit_need_to_frag configuration option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/3478011d64fcf57fa8d034f10621ce48cb5aff88"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-10-29 01:31:05.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-10-29 01:31:05.000000000","tz":480},"subject":"Fix rabbitmq dependency","message":"Fix rabbitmq dependency\n\npuppetlabs/rabbitmq [1] is deprecated, and max version is 5.6.0 which can\nnot be fulfilled by this dependency.\n\nCorrect it to puppet-rabbitmq [2] instead which is being actively maintained\nand has the right version.\n\n[1]: https://forge.puppet.com/puppetlabs/rabbitmq\n[2]: https://forge.puppet.com/puppet/rabbitmq\n\nChange-Id: I9c9b681d356b11ff4bafeb37c7c19459bb99cf2d\nCloses-Bug: #1850236\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/29f5049be21b57b32de8290935774bbe03a51dbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/29f5049be21b57b32de8290935774bbe03a51dbc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
