)]}'
{"id":"openstack%2Fpuppet-neutron~692808","triplet_id":"openstack%2Fpuppet-neutron~stable%2Fstein~I8a1bff166ee78dca402457fd1498e5906380c9ed","project":"openstack/puppet-neutron","branch":"stable/stein","topic":"bug/1844688","hashtags":[],"change_id":"I8a1bff166ee78dca402457fd1498e5906380c9ed","subject":"Add agents.l3 \"radvd_user\" configuration option","status":"MERGED","created":"2019-11-04 11:49:05.000000000","updated":"2019-11-05 11:45:32.000000000","submitted":"2019-11-05 11:45:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692808-1572954332664-df6c426c","meta_rev_id":"b11a125782c9317e061015f321b0bd9c514e253e","_number":692808,"virtual_id_number":692808,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-05 11:45:32.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-11-05 08:46:34.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-11-04 16:50:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-11-05 08:46:34.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-04 16:50:19.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-11-05 08:46:34.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-11-05 11:45:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7c325c3522ac595274cebbe4930c03841c0af34d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-11-04 11:49:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ae227d046edd3d99651e4a4eebf2aaf59e03e38b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 14:23:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9925d8a8a9a7459eb5085cf863e33517 : SUCCESS in 4m 25s\n- puppet-openstack-syntax-4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1d9e7eeaec914e57a0bd7385ecca7599 : SUCCESS in 5m 37s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/98c9fc00d5f14c01856292421e4afd4c : SUCCESS in 5m 54s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7c7add297d20461981901e16d2c4b454 : SUCCESS in 4m 22s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/42573a6d23f0403898b3791604e09ec9 : SUCCESS in 30m 18s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/fd130b5ac182480aa7656b02df247ca0 : SUCCESS in 30m 22s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/406568a846d14f4ca622bbd5153286ee : SUCCESS in 1h 14m 10s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/ff6fdf5d7ca34d3ca52edb3db2dffc6a : SUCCESS in 1h 17m 11s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/e781b415aaaf40398bb2b170242e4995 : SUCCESS in 56m 36s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/83f57f8c5e3f44b1b3906a44e7f7ec01 : SUCCESS in 45m 11s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/faac163363344976ac966587b199f8d5 : FAILURE in 22m 44s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/47d44bff30594864abbd8960b0c890f9 : FAILURE in 24m 02s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/76524f240730447d93953c10a40de166 : FAILURE in 23m 28s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/2901318f34714ba58e0d0a89fb7cd53a : FAILURE in 24m 59s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/f6dd7e01f8d245bc8a1ca1d4d63763ca : SUCCESS in 17m 33s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/74acf4e494fa48fb93399eb78f6eaf55 : FAILURE in 17m 33s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6ec1d8c170764ad39a20a032cb96eaaa : SUCCESS in 2h 31m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"10eafdf681b164a4c308be95a5e49b45f4d82d54","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-11-04 16:50:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8a0f7497930282111c13b17ab1ed27b3e0846b7e","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-11-05 08:46:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c88fe032b0c4586a149ff942bbefd004a708a6a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 08:46:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"299239bc562c6933700a3ac9dfec4d6aba410778","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 11:45:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"028952c6630facbc7e4d83731c365dfc4b6f44e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 11:45:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/df8919020c824f2282b8c599930b052f : SUCCESS in 5m 04s\n- puppet-openstack-syntax-4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/55a2f467726f46759330e847b0b3bdea : SUCCESS in 5m 10s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f5da8b76be2042ffb9f3e6c78be3c099 : SUCCESS in 4m 30s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/63acf1f149d14d548eef9b8e03eccc8a : SUCCESS in 3m 59s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/b6c02d1745b0481d8cf6715221eafeaf : SUCCESS in 28m 12s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/73e779728a0b4de48099ed7c38ce1e06 : SUCCESS in 29m 34s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/e499930be57848949c651a03927b4f80 : SUCCESS in 49m 43s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/9c616f3a7b75477994bce3e41c22a01a : SUCCESS in 58m 17s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/624a4c0ce0f64f91a6b6e8ef6283e4eb : SUCCESS in 1h 01m 18s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/26c281b01f36482a820d13279cd4384e : SUCCESS in 18m 29s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8d350230906b45ff94842541bf5da832 : FAILURE in 17m 39s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/907fc3f06a8d4152927b185330624c67 : SUCCESS in 2h 54m 55s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"470cd27ef6e719b612217ffb7a3649e8ac29d067","revisions":{"470cd27ef6e719b612217ffb7a3649e8ac29d067":{"kind":"REWORK","_number":1,"created":"2019-11-04 11:49:05.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/08/692808/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/08/692808/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/08/692808/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/08/692808/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/08/692808/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/08/692808/1"}}},"commit":{"parents":[{"commit":"26765458ef4f3999e09ebf5d9ac84d4abf841d75","subject":"Merge \"Ensure tunnel types are absent if not configured\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/26765458ef4f3999e09ebf5d9ac84d4abf841d75"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-10-29 10:57:34.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-11-04 11:48:39.000000000","tz":0},"subject":"Add agents.l3 \"radvd_user\" configuration option","message":"Add agents.l3 \"radvd_user\" configuration option\n\nBased on new Neutron configuration option [1], it will allow to define\nthe user to run the radvd daemon in the L3 agent.\n\nBy default, the user will be \"root\", to solve the problem found in the\nrelated bug.\n\n[1] Ie9a6fbf04d453a3c1c0bddf9ecaa3d4d6467e8ff\n\nConflicts:\n      manifests/agents/l3.pp\n\nChange-Id: I8a1bff166ee78dca402457fd1498e5906380c9ed\nCloses-Bug: #1844688\n(cherry picked from commit ab1049ae014d927f29b08079a89914446bb9acb3)\n(cherry picked from commit 1111e3dd2f8ae4f5c4bf921286d73dc833db868a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/470cd27ef6e719b612217ffb7a3649e8ac29d067"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/470cd27ef6e719b612217ffb7a3649e8ac29d067"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
