)]}'
{"id":"openstack%2Fpuppet-neutron~693603","triplet_id":"openstack%2Fpuppet-neutron~master~I4d48f811aff6bfb54f7089130d513ee52de2b8a7","project":"openstack/puppet-neutron","branch":"master","topic":"missing-defalut-value-radvd_user","hashtags":[],"change_id":"I4d48f811aff6bfb54f7089130d513ee52de2b8a7","subject":"Add default value for radvd_user option","status":"MERGED","created":"2019-11-09 01:50:34.000000000","updated":"2019-11-09 09:05:32.000000000","submitted":"2019-11-09 09:05:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693603-1573290332520-df6c426c","meta_rev_id":"ed4b34e473ad72ee11c06ec7483f9ff571b4557b","_number":693603,"virtual_id_number":693603,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-09 09:05:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-11-09 02:36:57.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2019-11-09 05:51:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2019-11-09 05:51:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-09 02:36:57.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-09 09:05:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7063fbf4344977ee8ea4e08a58459dcfa5e69f61","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-11-09 01:50:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8737c167972c4b9f3e2986567028c529994b5d34","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-09 02:36:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/03/693603/1/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/ad23951/ : FAILURE in 35m 37s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/03/693603/1/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/72fb6ad/ : SUCCESS in 37m 04s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/03/693603/1/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/c99e5ed/ : SUCCESS in 44m 50s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/03/693603/1/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/0d776f4/ : SUCCESS in 34m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8f1f5c326a57c2dae265b21639e31a7a71739a1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-09 04:58:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/41741084bb074092b645a32edb574511 : SUCCESS in 4m 21s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d4d1d60e601d4c4dbc912265f570bd3c : SUCCESS in 4m 22s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b2d926837d444188bcb68452010a3566 : SUCCESS in 4m 03s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/21b8daab648a446883d3610d807625dd : SUCCESS in 30m 14s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/180987bb2ad346f0814865cf96a27291 : SUCCESS in 16m 49s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f414146b35ef4598b17bc3cf532b7eab : FAILURE in 18m 25s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/73d16234ddfd4fdabb4e87a5c44e93fe : SUCCESS in 53m 31s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/2e85df9bd5a8468abf0a4c94dba4cfbd : SUCCESS in 58m 33s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/ec67cc2234b8410398b7e8257cf3999a : SUCCESS in 1h 01m 18s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/123ec27d44db4d2b9211f5902a4101bc : SUCCESS in 51m 24s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/1f45eb9242b44961bc5fb17ea166a648 : SUCCESS in 44m 11s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/81ef8ace684e4644b99f19a24f8477cc : SUCCESS in 44m 38s\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6a97119f9dca40a398d0389ed670fff2 : SUCCESS in 37m 58s\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/de1bec709cfc4deabf3b5fffd5f80cf6 : SUCCESS in 36m 36s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/90cfd766e54c48c9b1837e361d0aeb86 : SUCCESS in 21m 05s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/303a1cd3232b47c696e4f8e74ff641ba : FAILURE in 22m 10s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a53c5653aa9e4cd3bab60135fcfced0d : SUCCESS in 3h 06m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"04d154c545014fae523056b278804beeaf485987","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-11-09 05:51:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3ec5c030a4717f2d1dda1ceeb2467b43f7d706e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-09 05:51:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c91a1a49ba8ef13a0e65096107ad5e2c4472f067","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-09 09:05:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fa527537041af6d51c7a15c21fcade7fdf328e6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-09 09:05:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/813343b3d5f447ccb1656dad95342120 : SUCCESS in 4m 33s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/500bb4bbb10b48cba81e8793a6e02515 : SUCCESS in 6m 20s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e00b5e48416c4b86aca516bedc1610cf : SUCCESS in 5m 14s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/1a3b7ef8c61d4a39b04a826c0b01fa1e : SUCCESS in 28m 16s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/20b5c89c214d4149a7b773f171348618 : SUCCESS in 40m 12s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/bdb2860639a843449653bcb80a4f69c6 : SUCCESS in 53m 21s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/0c8ef343c28d4a64b196c05789bad59f : SUCCESS in 1h 00m 47s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/3267707251504a5da08922cdbf129050 : SUCCESS in 1h 31m 21s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/3280168decae47d2b52021cca6aace1c : SUCCESS in 49m 45s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/0da89c91f90a477081af1ae7f46b71e0 : SUCCESS in 17m 26s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e6c350cc46e0456789640f80edb0416d : TIMED_OUT in 3h 13m 22s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0d9bc9d96cad7ff76cd5af063c2b188fb9b37047","revisions":{"0d9bc9d96cad7ff76cd5af063c2b188fb9b37047":{"kind":"REWORK","_number":1,"created":"2019-11-09 01:50:34.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/03/693603/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/03/693603/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/03/693603/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/03/693603/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/03/693603/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/03/693603/1"}}},"commit":{"parents":[{"commit":"25d3fa3f15ad081382f04cd6e590139400686a8e","subject":"Merge \"Update the constraints url\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/25d3fa3f15ad081382f04cd6e590139400686a8e"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-11-09 01:49:50.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-11-09 01:49:50.000000000","tz":480},"subject":"Add default value for radvd_user option","message":"Add default value for radvd_user option\n\nChange-Id: I4d48f811aff6bfb54f7089130d513ee52de2b8a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/0d9bc9d96cad7ff76cd5af063c2b188fb9b37047"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/0d9bc9d96cad7ff76cd5af063c2b188fb9b37047"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
