)]}'
{"id":"openstack%2Fpuppet-neutron~693604","triplet_id":"openstack%2Fpuppet-neutron~master~Iea4a47e5a58021a2244057996ed3f5765cb8978f","project":"openstack/puppet-neutron","branch":"master","topic":"add-unit-test-radvd_user","hashtags":[],"change_id":"Iea4a47e5a58021a2244057996ed3f5765cb8978f","subject":"Add unit test for radvd_user option","status":"MERGED","created":"2019-11-09 01:57:02.000000000","updated":"2019-11-09 08:51:54.000000000","submitted":"2019-11-09 08:51:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693604-1573289514868-dd5696ed","meta_rev_id":"b5f216e8d56f10f93a249a02ec1fdebc64447828","_number":693604,"virtual_id_number":693604,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-09 08:51:54.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2019-11-09 05:51:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2019-11-09 05:51:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-09 08:51:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fb51b326db21afb6662596bfb94b71563c2b4d54","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-11-09 01:57:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff85093317c502babf54ff0cd247ec620fde61e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-09 04:55:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/978bfb4ade484e8f82d17bf520336140 : SUCCESS in 4m 34s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2f1df56774a54aad97c58b03a582dc8c : SUCCESS in 5m 49s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3ca78e1261c94ed2875e29bda46b401a : SUCCESS in 5m 51s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/79cdd5dcaa3e4c71b5957d5a7210639e : SUCCESS in 24m 43s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/14cc6168e8fb4e618a9971bff02f0098 : SUCCESS in 15m 26s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bb7bef6068884f4e849476bde58dff2a : FAILURE in 16m 10s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/03dec6f837754ca58e96a3adc233b886 : SUCCESS in 2h 57m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f8b575a481f04f9509508f4ab4755d57993ed5b5","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-11-09 05:51:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dcace437cac4e4a1d6fbfd098a81c7a08f5db284","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-09 05:51:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0c4e432df5f739ec6dad2f828d9c7ea239aa4b01","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-09 08:51:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d098aaa79d942fe6f0fed6a7c01a707cf50e15c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-09 08:51:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6982b182a803446ead19cce76575dd28 : SUCCESS in 4m 18s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/48236384c8fa41e29c92409b7863d69c : SUCCESS in 3m 41s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/82802d4095c746d3a18a55ff3ee341e9 : SUCCESS in 5m 59s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/fa728829f01f425abf27cad0941316ee : SUCCESS in 24m 04s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2491e3c4beeb40e3a8023120f29936bc : SUCCESS in 42m 10s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f2a35565a6ef49f6aac33e0c1bde8e75 : SUCCESS in 2h 59m 57s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f590e019eb28dbc159f07c021a5712b55c14ef69","revisions":{"f590e019eb28dbc159f07c021a5712b55c14ef69":{"kind":"REWORK","_number":1,"created":"2019-11-09 01:57:02.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/04/693604/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/04/693604/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/693604/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/693604/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/04/693604/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/04/693604/1"}}},"commit":{"parents":[{"commit":"25d3fa3f15ad081382f04cd6e590139400686a8e","subject":"Merge \"Update the constraints url\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/25d3fa3f15ad081382f04cd6e590139400686a8e"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-11-09 01:56:38.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-11-09 01:56:38.000000000","tz":480},"subject":"Add unit test for radvd_user option","message":"Add unit test for radvd_user option\n\nChange-Id: Iea4a47e5a58021a2244057996ed3f5765cb8978f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/f590e019eb28dbc159f07c021a5712b55c14ef69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/f590e019eb28dbc159f07c021a5712b55c14ef69"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
