)]}'
{"id":"openstack%2Fpuppet-neutron~695577","triplet_id":"openstack%2Fpuppet-neutron~master~Id2bbddfa664fe3ba43992ab8721ef8117475df94","project":"openstack/puppet-neutron","branch":"master","topic":"add_ovn_metadata_remote_probe_interval","hashtags":[],"change_id":"Id2bbddfa664fe3ba43992ab8721ef8117475df94","subject":"Add the ability to configure ovn-remote-probe-interval","status":"MERGED","created":"2019-11-21 21:03:42.000000000","updated":"2019-11-25 11:32:11.000000000","submitted":"2019-11-25 11:31:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695577-1574681466132-1353a74c","meta_rev_id":"bbf6eb45bacd3cf0b7e2c1bd579debdedf297667","_number":695577,"virtual_id_number":695577,"owner":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-11-22 18:05:26.000000000","_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-25 11:31:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-11-22 18:59:13.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":2,"date":"2019-11-25 08:31:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-11-22 21:58:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2019-11-25 06:49:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":1,"date":"2019-11-25 08:31:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-22 18:59:13.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-22 21:58:00.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-11-25 06:49:56.000000000","updated_by":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"reviewer":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"state":"REVIEWER"},{"updated":"2019-11-25 08:31:15.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2019-11-25 11:31:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e9c7142f7d5d89591324bf571621c1055f137306","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-11-21 21:03:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c9d0f615328469412526f5623f568f486219a9ea","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-11-21 21:05:42.000000000","message":"Topic set to add_ovn_metadata_remote_probe_interval","accounts_in_message":[],"_revision_number":1},{"id":"9bbf0ba5e06c184639730c99de8135823d8b7c16","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-21 22:03:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/77/695577/1/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/9abf2ec/ : SUCCESS in 36m 58s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/77/695577/1/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/8d9a865/ : SUCCESS in 55m 25s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/77/695577/1/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/22d9418/ : SUCCESS in 57m 38s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/77/695577/1/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/20fd145/ : SUCCESS in 33m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c6ad8fdd658970ed1acbbdc65fb91ef5f332b80b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 00:22:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2dcaf2939db641c280b0c176b0a4b39b : SUCCESS in 4m 55s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/34bff4c7eba24a04b3ad174b8265d414 : SUCCESS in 4m 21s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7bf9e3b516de4b27ad53a50543462102 : SUCCESS in 5m 03s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/9243e16b401a477b9c89210992373759 : SUCCESS in 25m 31s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c6bb7030f342458cb751df6d7a430bb4 : SUCCESS in 17m 40s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5409b376a67e4786965b1a3a0251fc56 : FAILURE in 17m 03s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/09ea728983644b2d958efe888b892313 : FAILURE in 1h 04m 20s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/f2cd8af08757461d853ca4292622ab0d : SUCCESS in 56m 55s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/ef81d97888584c868a8dafc062534c64 : SUCCESS in 1h 06m 51s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/bcc527afbb65440a8a7e4bcb37c945f5 : SUCCESS in 53m 08s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/a2b0a1a56b3a41f28e621cd7cef9cdd4 : SUCCESS in 43m 08s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/90028586ed1f41ebbea2b27e667b55df : SUCCESS in 49m 53s\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/309adb8648cf403b91bf9e56765341ca : SUCCESS in 47m 04s\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/dbab17dab4f44c6797deb4e2277330b5 : SUCCESS in 37m 43s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/f58487c63d0c422aad1d0ea82e56c8f9 : SUCCESS in 18m 34s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/429a97ec0cb24a6ab881d3a05c2aba27 : FAILURE in 17m 48s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b88a3e65ea2f468da5e7eefc86eeb544 : SUCCESS in 3m 12s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a075738356bb4fa793bf691ce09ab842 : TIMED_OUT in 3h 14m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4b485edc06d607596d5b6de7d8cf0056735e00ad","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-11-22 00:43:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2c7eac964182e8b750eeadca2e24c70ce1b85a44","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-22 01:42:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/77/695577/1/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/176768e/ : SUCCESS in 40m 28s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/77/695577/1/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/f71bbef/ : SUCCESS in 56m 39s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/77/695577/1/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/e867a55/ : SUCCESS in 54m 23s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/77/695577/1/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/95aee4e/ : SUCCESS in 40m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3c608f045f4cb62a22cc3fb7d72df92d38f0a379","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 03:58:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0e115c65e6704832886158ac508629b7 : SUCCESS in 4m 08s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5e5c8918429343beb902652c96b40983 : SUCCESS in 4m 27s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f5a8c75927a44ebaa233929e4002f8b5 : SUCCESS in 3m 54s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/fb47dfb86e544dc1bd81b9560ab7ea6d : SUCCESS in 30m 18s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/733de22450e2477d90cbf9019e388bfc : SUCCESS in 17m 26s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/039df92eba5a4c55b5b090728ff3abde : FAILURE in 22m 29s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/629bbd82a6894b8a94793628b9ebddca : SUCCESS in 52m 29s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/60c1736bd64c46d89bb23f8a7dcc7df8 : SUCCESS in 1h 03m 25s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/f1dd919887d24e29953247e3819094f2 : SUCCESS in 1h 08m 40s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/16d8b8f4f7c84004ac84c4987a78c8fa : SUCCESS in 50m 51s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/d53734b3f4ec4c25b7d6acaf7b464f35 : SUCCESS in 40m 01s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a213442b3ccb4a89a3164ece17d0b524 : SUCCESS in 51m 54s\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/da535416fab047faa65cd3d512460fa6 : SUCCESS in 38m 07s\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/d17584ddf153467eaf9a31767c306e6f : FAILURE in 1h 15m 39s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/5145457d8f794c139ee1d7b3e27eed5b : SUCCESS in 15m 42s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0b403f556e2545e7b0bc4a63aaed4df2 : FAILURE in 16m 45s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d16f01612a2744c8a50732f2cdee64ad : SUCCESS in 3m 22s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b8346c459a744fe59ad5470f5d5edca0 : TIMED_OUT in 3h 14m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"24934c6474c8f79c044c12890c9fd963723330ff","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-11-22 16:17:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"636f8c7bbcb89bbcf80aa9e0afbb8625b556c42e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-22 17:06:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/77/695577/1/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/5538a68/ : SUCCESS in 36m 34s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/77/695577/1/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/c3c921a/ : SUCCESS in 46m 19s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/77/695577/1/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/3416ee9/ : SUCCESS in 47m 17s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/77/695577/1/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/30a9df0/ : SUCCESS in 31m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"426c95ea3c487c4318bebe39a5338168a44e624b","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-11-22 18:05:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ee8a0b6e75fc09415f0a11d009e8477974a189d5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-22 18:59:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/77/695577/1/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/872e861/ : SUCCESS in 34m 49s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/77/695577/1/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/96b481b/ : SUCCESS in 41m 34s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/77/695577/1/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/3d8f071/ : SUCCESS in 52m 16s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/77/695577/1/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/c9a44ae/ : SUCCESS in 31m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"920e6bf21e1cda15a2a9ff27b017a4ff45fc7c8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 19:35:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ecbfcd6427b04b41a4a1b7263822775d : SUCCESS in 4m 26s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2ba860193eb94652a7f85fbc1622eb4f : SUCCESS in 4m 39s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5e02d761a563456b9a9ebab4e4b244c0 : SUCCESS in 5m 04s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/7471479818354374a2daceb1ad241e93 : SUCCESS in 29m 55s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6f1e1ffa27ff46c89f7aa46c209bb592 : SUCCESS in 14m 20s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e5d7af81f0954778b5729114e334bd3f : FAILURE in 16m 30s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/69f566ea624741e68dd22e092d61433b : SUCCESS in 1h 04m 17s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/c4697de18bcc4e32b51d21a96526ba97 : SUCCESS in 1h 08m 13s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/c2cfdea7a93341b2af81a645ed46cf8c : SUCCESS in 1h 11m 59s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/639d78c7f60346ccbd13d854913e8bfc : SUCCESS in 47m 53s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/8af6d6935fef444b8a7bf07b08d5da04 : SUCCESS in 42m 11s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ed957176ae6646adab7c9cf752bb4574 : SUCCESS in 45m 11s\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fc654819f5fa4626ade10fc3de083ba3 : SUCCESS in 37m 15s\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/67633d88a58e4180b2a6caf9b2d33dd0 : SUCCESS in 48m 54s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/2dc19e533d934a2a835cabf937a6b70c : SUCCESS in 27m 16s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/07a55f2ac9a04a85883c530388632d73 : FAILURE in 29m 56s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f7306d0a63043768e67725b228b980c : SUCCESS in 3m 11s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/429c9fe485174e34ba67b9a2fb5e5592 : SUCCESS in 3h 10m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"36a123757bc8688497a88b7ae0fa58be7994bc5c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-11-22 21:58:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"40a54142c3b30436d109d09f1e27204fa557959a","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2019-11-25 06:49:56.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3ca2ebec7011abbdf6ab046f197ebf4a30a84107","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-11-25 08:31:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3b45a32b1a093afb9287a1778efc6fecf482fda9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 08:31:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"55544ba5e3bd277c4cc812b118f4019a32ad2e8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 11:31:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/406ea390730d45029228786f8f3f4a76 : SUCCESS in 5m 13s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2fe0ac0c3eca435793405680617705a9 : SUCCESS in 5m 33s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e302a65f66c44c46a94586bacd592c01 : SUCCESS in 4m 18s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/db51c02d27364d92bbac26b3493fd1e9 : SUCCESS in 27m 34s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7cc2242b01b34a378eaa6ceec16f6e95 : SUCCESS in 19m 06s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/793689724a1d45919f5c1ffb664cff9e : SUCCESS in 51m 06s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/5ae1f60a07ed422bbcd3de3f24236602 : SUCCESS in 56m 12s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/2139b48a3e214a3c9a85c9e64db3617d : SUCCESS in 1h 02m 29s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/453ad70f1e224f898dd8c04ddbc6530f : SUCCESS in 42m 57s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/294ee270fe9d44ddbd6ff969f3cd5b39 : SUCCESS in 17m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/844fe1b90189432d99a4f3b181d571ae : SUCCESS in 3m 19s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3739b509bf2745dfb0770d476a1e6116 : SUCCESS in 2h 58m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0a26db9375117cce62cd147482ac3d1b47f40328","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 11:31:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"bbf6eb45bacd3cf0b7e2c1bd579debdedf297667","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 11:32:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cadddc85e67c44ebb305d431b0f2baa9 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c6b3800f7dd4e977f3234b5482d79d0959712855","revisions":{"c6b3800f7dd4e977f3234b5482d79d0959712855":{"kind":"REWORK","_number":1,"created":"2019-11-21 21:03:42.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/77/695577/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/77/695577/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/77/695577/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/77/695577/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/77/695577/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/77/695577/1"}}},"commit":{"parents":[{"commit":"2b1d8b6be81e79f7c8f18b5e4cd07da483f9b23c","subject":"Add executor_thread_pool_size","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/2b1d8b6be81e79f7c8f18b5e4cd07da483f9b23c"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2019-11-21 20:47:03.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2019-11-21 20:48:35.000000000","tz":-300},"subject":"Add the ability to configure ovn-remote-probe-interval","message":"Add the ability to configure ovn-remote-probe-interval\n\nThis parameter sets inactive probe interval of the JSON\nsession from ovn-metadata to the OVN SB database.\nBy default this it is 5s which not be sufficient in\nloaded systems or during high control-plane activity spikes,\nleading to unnecessary reconnections to OVSDB server.\nNow it is extended by default to 1 min and it\nis configurable.\n\nChange-Id: Id2bbddfa664fe3ba43992ab8721ef8117475df94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c6b3800f7dd4e977f3234b5482d79d0959712855"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c6b3800f7dd4e977f3234b5482d79d0959712855"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
