)]}'
{"id":"openstack%2Fpuppet-neutron~699417","triplet_id":"openstack%2Fpuppet-neutron~stable%2Frocky~Iee2f869c4054f1120d224994aba8c827f981e6c0","project":"openstack/puppet-neutron","branch":"stable/rocky","topic":"rocky/cisco","hashtags":[],"change_id":"Iee2f869c4054f1120d224994aba8c827f981e6c0","subject":"CiscoNexus: Update and Remove deprecated vars","status":"MERGED","created":"2019-12-17 13:55:43.000000000","updated":"2019-12-18 03:19:22.000000000","submitted":"2019-12-18 03:19:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"699417-1576639162062-fead5b33","meta_rev_id":"8b7d0ee4ac232f558a8f86db597780307a4e5801","_number":699417,"virtual_id_number":699417,"owner":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-18 03:19:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2019-12-18 01:31:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-12-18 01:21:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2019-12-18 01:31:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-18 01:21:34.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-12-18 01:31:55.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2019-12-18 03:19:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"badd66f068e27856a4578a54de6a3d62fbe63a1b","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-12-17 13:55:43.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"be06b6490715a835f11078ff4f47230b3628b2d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 15:39:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/9f7c5d95c589463683d650114c467cb4 : SUCCESS in 5m 37s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/b4267d45a2aa43a694c6de65291fe5c0 : SUCCESS in 4m 17s\n- puppet-openstack-syntax-5 https://zuul.opendev.org/t/openstack/build/3ae2b803a0a1470490789fe62c923d7d : SUCCESS in 4m 15s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/a333708c0f56498a8d30f3e1910135d4 : SUCCESS in 27m 34s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/275dcd8beef343588064405e78fe20ce : SUCCESS in 55m 43s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/b32d34e26f1146209f46851e6971f23a : SUCCESS in 59m 56s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/6247c2bab13b486da71e90d7f3724033 : SUCCESS in 1h 05m 20s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic https://zuul.opendev.org/t/openstack/build/8f26eca27cd649e28eb4850305948900 : SUCCESS in 42m 54s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/efa3a9abfb7d4755a579cbe90b851ebb : TIMED_OUT in 1h 31m 16s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5df03d819f6f4e5b9f8a7b763deb2f18 : TIMED_OUT in 1h 30m 58s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8df30936daf344c58fd9ec771b5dc567 : TIMED_OUT in 1h 31m 26s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/fdf13093c52a471f816d18a90927f33f : TIMED_OUT in 1h 31m 14s (non-voting)\n- puppet-openstack-integration-4-scenario001-tempest-debian-stable-luminous https://zuul.opendev.org/t/openstack/build/d6498589228e4dce814045afa9baa12b : TIMED_OUT in 1h 31m 31s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-debian-stable https://zuul.opendev.org/t/openstack/build/8cb63bb79d3f4fa5b70076e6cd9bb314 : FAILURE in 53m 52s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-debian-stable https://zuul.opendev.org/t/openstack/build/34dd469c436a4656b27bb81a2577f6ad : TIMED_OUT in 1h 31m 18s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-debian-stable-luminous https://zuul.opendev.org/t/openstack/build/bd839f8690f541c4af8636de8d5e6b47 : TIMED_OUT in 1h 31m 29s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/bfcd3a667e18425a95b77768efbff67a : SUCCESS in 17m 58s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ea3d7928fa794c108c208ef21dac9e96 : SUCCESS in 21m 43s\n- tripleo-puppet-ci-centos-7-undercloud-oooq https://zuul.opendev.org/t/openstack/build/132078792a8b42608eee6af4b416b54e : SUCCESS in 1h 06m 37s\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/dc20ca5722e9493dada585c7f7e6cda7 : SUCCESS in 1h 31m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ccc6309c76b22cd93dcedcadf33d66028c742aa3","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-12-17 16:35:45.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)\n\nI know we don\u0027t really support this thing but I think we shouldn\u0027t backport the hard removals","accounts_in_message":[],"_revision_number":1},{"id":"31db917b6e4e4f10ec028cb36bfbd45c5b36486f","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-12-17 20:58:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9fb206e6e150425dbf6991fb28fd58d6228d92ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 22:56:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/b5365b3142e349d0b4b6ee6ede4238fb : SUCCESS in 6m 11s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/4c50de1a435d4c3a87eec84fbe33ede5 : SUCCESS in 4m 23s\n- puppet-openstack-syntax-5 https://zuul.opendev.org/t/openstack/build/9c7399baef75491bbfc7c7c61c482dcb : SUCCESS in 4m 27s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/5167bdf5c8da4d0ab20701fbde7550d0 : SUCCESS in 31m 59s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/20c925c103db484ba2591478f16580cd : SUCCESS in 51m 04s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/7eb63a6c9a9f4caebee2c77851ac705e : SUCCESS in 55m 51s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/ee2bbb0039e24a9b912de409d735eba1 : SUCCESS in 1h 01m 23s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic https://zuul.opendev.org/t/openstack/build/0d508dd35ef14b2498ecfe08ef8df4ee : SUCCESS in 42m 53s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/648b191f2d7a4600810a96009c6f0ce4 : TIMED_OUT in 1h 30m 53s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e2e95b388833492bb3fec2986565fe30 : TIMED_OUT in 1h 31m 32s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4051ec400210480c81d9e26a1c7e5609 : TIMED_OUT in 1h 31m 01s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/8173119b33c04c3083f2510be25f50ee : TIMED_OUT in 1h 31m 21s (non-voting)\n- puppet-openstack-integration-4-scenario001-tempest-debian-stable-luminous https://zuul.opendev.org/t/openstack/build/4838d53ed99a40828b0d81480bb28f4c : TIMED_OUT in 1h 31m 02s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-debian-stable https://zuul.opendev.org/t/openstack/build/2cc2152a361e48fb9d5ab8695e430c7f : FAILURE in 50m 45s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-debian-stable https://zuul.opendev.org/t/openstack/build/ca1d54f1be214cba82221b2f9cd6d4ae : TIMED_OUT in 1h 31m 00s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-debian-stable-luminous https://zuul.opendev.org/t/openstack/build/1662b680103f4e8ba0e6e6d9329405c7 : FAILURE in 7m 53s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/6a1534d6d83d4f3a95a2dc0f304cb32f : SUCCESS in 18m 02s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/16753407600b492ab04bd071eeab3233 : SUCCESS in 14m 38s\n- tripleo-puppet-ci-centos-7-undercloud-oooq https://zuul.opendev.org/t/openstack/build/7d45ca1e2b384a01bb6165f80f2f2665 : SUCCESS in 1h 08m 26s\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/4ad35bf4703f4708b2266852754dad04 : SUCCESS in 1h 48m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d2a73aa6aa4303e89efa8fce63775d40ab89a7cc","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-12-18 01:21:34.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2a82800f02e3308756c2b2c582b2ccbe43845d59","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-12-18 01:31:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3d31363c44f3a5a0482545a42432bc23521c2152","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 01:32:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ff931851a245de91dcea24875fdb713bd1cfe8de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 03:19:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/b5a5d2587eeb4996a1730fdab34967e4 : SUCCESS in 4m 31s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/21875fd15b73402a873fabad127cd30d : SUCCESS in 4m 32s\n- puppet-openstack-syntax-5 https://zuul.opendev.org/t/openstack/build/f4a3b592aa2b4d3596e862d9b40f9318 : SUCCESS in 4m 06s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/20d831e96ef0437794eed79b689adb8e : SUCCESS in 27m 31s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/f6386d46886247f6a7eca8d5c2eda1f8 : SUCCESS in 48m 54s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/15bbc1802d524122be18f0f227e70dcc : SUCCESS in 58m 45s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/73a02dab8925435da6d55b9a1767698c : SUCCESS in 59m 27s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic https://zuul.opendev.org/t/openstack/build/ad5b393ceded42c9b94e20ffb2706e97 : SUCCESS in 40m 29s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/c180fb6e9e304b379b1eeb88d774bed5 : SUCCESS in 18m 43s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ac3392031eac4ccc89dad8d8e71710cf : SUCCESS in 17m 10s\n- tripleo-puppet-ci-centos-7-undercloud-oooq https://zuul.opendev.org/t/openstack/build/d09df5b7ebbf434b97bae2cb243c63da : SUCCESS in 1h 17m 50s\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/52b20c0c08474c8fa3f6845ea55bdbcf : SUCCESS in 1h 42m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"849c3e61af440132350adedcfab01b3844bdfa1f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 03:19:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f1ac3e32ea5efc5ab054caeed3578277581a0cba","revisions":{"28152f4a305e69fec964a3a27587fa496060455f":{"kind":"REWORK","_number":1,"created":"2019-12-17 13:55:43.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/17/699417/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/17/699417/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/17/699417/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/17/699417/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/17/699417/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/17/699417/1"}}},"commit":{"parents":[{"commit":"0919aa8cd72d933e9925a9e64c08a5bcb0ae2087","subject":"Add agents.l3 \"radvd_user\" configuration option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/0919aa8cd72d933e9925a9e64c08a5bcb0ae2087"}]}],"author":{"name":"Carol Bouchard","email":"caboucha@cisco.com","date":"2018-10-01 14:03:49.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-12-17 13:55:43.000000000","tz":0},"subject":"CiscoNexus: Update and Remove deprecated vars","message":"CiscoNexus: Update and Remove deprecated vars\n\nUpdate and Remove deprecated variables from Cisco Nexus Config.\n* Deprecated variables include vlan_name_prefix, svi_round_robin,\n  provider_vlan_name_prefix, persistent_switch_config,\n  never_cache_ssh_connection, ssh_port, switch_replay_count,\n  nexus_driver, and host_key_checks.\n* Change config replay to on by defaulting switch_heartbeat_time\u003d30.\n* Replaced method of configurating host_ports_mapping config.\n\nChange-Id: Iee2f869c4054f1120d224994aba8c827f981e6c0\nCloses-bug:  #1793379\n(cherry picked from commit f45169f38db4333d60841136fe3fab951221eeb6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/28152f4a305e69fec964a3a27587fa496060455f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/28152f4a305e69fec964a3a27587fa496060455f"}]},"branch":"refs/heads/stable/rocky"},"f1ac3e32ea5efc5ab054caeed3578277581a0cba":{"kind":"REWORK","_number":2,"created":"2019-12-17 20:58:21.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/17/699417/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/17/699417/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/17/699417/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/17/699417/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/17/699417/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/17/699417/2"}}},"commit":{"parents":[{"commit":"0919aa8cd72d933e9925a9e64c08a5bcb0ae2087","subject":"Add agents.l3 \"radvd_user\" configuration option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/0919aa8cd72d933e9925a9e64c08a5bcb0ae2087"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-12-17 20:57:57.000000000","tz":-300},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-12-17 20:57:57.000000000","tz":-300},"subject":"CiscoNexus: Update and Remove deprecated vars","message":"CiscoNexus: Update and Remove deprecated vars\n\nBackward compatible backport (modified by hand).\n\nUpdate and Remove deprecated variables from Cisco Nexus Config.\n* Deprecated variables include vlan_name_prefix, svi_round_robin,\n  provider_vlan_name_prefix, persistent_switch_config,\n  never_cache_ssh_connection, ssh_port, switch_replay_count,\n  nexus_driver, and host_key_checks.\n* Change config replay to on by defaulting switch_heartbeat_time\u003d30.\n* Replaced method of configurating host_ports_mapping config.\n\nChange-Id: Iee2f869c4054f1120d224994aba8c827f981e6c0\nCloses-bug:  #1793379\n(cherry picked from commit f45169f38db4333d60841136fe3fab951221eeb6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/f1ac3e32ea5efc5ab054caeed3578277581a0cba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/f1ac3e32ea5efc5ab054caeed3578277581a0cba"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
