)]}'
{"id":"openstack%2Fpuppet-neutron~704579","triplet_id":"openstack%2Fpuppet-neutron~master~I682e22d56d65ec160a09780f6780467b1c8718ba","project":"openstack/puppet-neutron","branch":"master","hashtags":[],"change_id":"I682e22d56d65ec160a09780f6780467b1c8718ba","subject":"Remove neutron::agents::metadata::metadata_ip","status":"MERGED","created":"2020-01-28 14:17:57.000000000","updated":"2020-01-30 19:21:46.000000000","submitted":"2020-01-30 19:20:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"704579-1580412014622-641ca554","meta_rev_id":"7682d4c9043c91c4b29489da115fdbd6c3dd4a4a","_number":704579,"virtual_id_number":704579,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-30 19:20:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-01-29 14:59:15.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-01-28 14:29:23.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-01-30 15:57:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2020-01-30 15:57:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-28 14:29:23.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-01-29 14:59:15.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-01-30 15:57:50.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-01-30 19:20:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3c722964290b9984e2b27abcd9b7a580123a303d","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-01-28 14:17:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"79a47e8b40a9f1a6f5386f198d93b1672b81d5df","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-01-28 14:29:23.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f9efdd8ba22aa795626f4fb0bc8ee62ee7750283","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 17:05:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9662886676b54fb9948bc03d11526867 : SUCCESS in 5m 18s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5b17b6879e274a5aa0bf84f9a4be5c80 : SUCCESS in 4m 42s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b7e8f97b49f84cabb7301228320af495 : SUCCESS in 7m 28s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/c552da525b9042aba8f230e90d3715a5 : SUCCESS in 30m 16s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b11e885d4bef49cbb47b59ce48c7a022 : SUCCESS in 21m 52s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fba6ca8acf6c4357a1f418256ab831b1 : FAILURE in 24m 23s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/497bea65e2e44b3dadb9ad0e123b6406 : SUCCESS in 1h 04m 13s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/56e854b754c442fe925d9a218af2999e : SUCCESS in 1h 09m 23s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/bfdd5a05483e4af09812ca8f09f02e3c : SUCCESS in 57m 54s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/d2a509d00ce941d3979f51e9cb625e4f : SUCCESS in 47m 13s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/00cd47ac5b554a7284beb84f949bc6fd : SUCCESS in 47m 44s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/236df879e8e849dbad1f67821114fca9 : SUCCESS in 50m 27s\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e1987400dc2244fe8c5ba88f863e59d1 : SUCCESS in 49m 07s\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/d8841b5d881b477a9ad81c19225a56db : SUCCESS in 42m 26s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/eec7a852b204438e9ec2eea8d66754af : SUCCESS in 54m 31s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b6406a115ddd45a8aac43dac3f8d53f2 : SUCCESS in 1h 14m 35s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/dbf4599f12194253aad581a100e73e92 : SUCCESS in 58m 20s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b43e1e6e6bce4705bbf907eebe8049eb : SUCCESS in 48m 54s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/d240889420454ccf93320b75056eac74 : SUCCESS in 21m 23s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6aeed09c237d4b3f8de6c683f99f4950 : FAILURE in 20m 03s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d021f45e583445649545ba2b47e0ce0e : SUCCESS in 5m 41s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/1dce82af19f94dceb80f3a8a495d4047 : SUCCESS in 2h 44m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1aeffb55d5a976a84e410dab37e9410cab60910b","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-01-29 14:59:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2bd440f85632e5e3c1004a1c34ab9df83606445d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-01-30 15:57:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8550768a835cb23c524e17baa6bb1dd95f5fa796","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 15:58:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"946f9cc603f12e9720b7124ee96a958b7a26fbd4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 19:20:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ca35458c243714c73e39a56e259e1abb819cfa66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 19:20:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bde6ca0861ba42dcb6cbf1d16a326ada : SUCCESS in 5m 59s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cecef2be74524b7b8071d1c549135c30 : SUCCESS in 6m 30s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c292a9fae23849d88f3c38d6944566c8 : SUCCESS in 11m 35s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/e1296b7b90694b35b1ee94fad22f636f : SUCCESS in 39m 04s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/474a378b03944992b870a9d913533638 : SUCCESS in 20m 37s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/b707892e73dc43f8baa9049726396353 : SUCCESS in 58m 12s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/c46f5f9509f544d2abaa3ac237147f44 : SUCCESS in 1h 13m 50s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/a74fa29aa8a2412baf37f88a8e871223 : SUCCESS in 1h 34m 22s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/70542243a96a44f7893ca64c485b4294 : SUCCESS in 51m 45s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/1435d760577148d288fe923a1890ecfa : SUCCESS in 22m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/667427be46104c21a07c09310fbf4c4f : SUCCESS in 5m 54s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2ca30969a784423892e098f9253e9c50 : TIMED_OUT in 3h 17m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7682d4c9043c91c4b29489da115fdbd6c3dd4a4a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 19:21:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d6e8c0c304b840c6b5550eb2fc3edf39 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e3c8f1423ba2f40b6f1e44cc082fddaebf17b439","revisions":{"e3c8f1423ba2f40b6f1e44cc082fddaebf17b439":{"kind":"REWORK","_number":1,"created":"2020-01-28 14:17:57.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/79/704579/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/79/704579/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/79/704579/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/79/704579/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/79/704579/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/79/704579/1"}}},"commit":{"parents":[{"commit":"9f5c39cdcf79a435ded68e57ea6444e86359fc31","subject":"Fix README","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/9f5c39cdcf79a435ded68e57ea6444e86359fc31"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-01-28 14:17:35.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-01-28 14:17:35.000000000","tz":60},"subject":"Remove neutron::agents::metadata::metadata_ip","message":"Remove neutron::agents::metadata::metadata_ip\n\nDeprecated in favor of metadata_host.\n\nChange-Id: I682e22d56d65ec160a09780f6780467b1c8718ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/e3c8f1423ba2f40b6f1e44cc082fddaebf17b439"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/e3c8f1423ba2f40b6f1e44cc082fddaebf17b439"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
