)]}'
{"id":"openstack%2Fpuppet-neutron~705909","triplet_id":"openstack%2Fpuppet-neutron~master~Id73e8ae40f4db67247bde36156e3b84c456b9f29","project":"openstack/puppet-neutron","branch":"master","hashtags":[],"change_id":"Id73e8ae40f4db67247bde36156e3b84c456b9f29","subject":"Add OVS ovsdb_timeout parameter","status":"MERGED","created":"2020-02-05 10:02:09.000000000","updated":"2020-02-08 01:39:19.000000000","submitted":"2020-02-08 01:37:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"705909-1581125871942-4edda447","meta_rev_id":"98559ae5974f8fd60fc825895d6d42fa755a5e51","_number":705909,"virtual_id_number":705909,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2020-02-06 15:30:06.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-08 01:37:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-02-06 18:59:36.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2020-02-07 22:17:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2020-02-07 22:17:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-06 18:59:36.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-02-07 22:17:10.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-02-08 01:37:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7cf1053e3976b29f22553bf5dd73a5e614e0ec0d","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-02-05 10:02:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"42610db5505a65e054496222b706f17733095677","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 13:23:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1c2b5b9894a047c1adead6e9b1ee342e : SUCCESS in 4m 47s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6c40d8f58dde4904b38e26d74e73f7d0 : SUCCESS in 5m 40s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/03a0b07e19374c3ab38c750e4f434dbd : SUCCESS in 5m 40s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/e73ce31cb44e4eea8166844a03e4a90f : SUCCESS in 30m 15s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b7a2b0bd542c478596e5a31126147d0b : SUCCESS in 16m 44s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/541dab849ffb4c56a80694df4e79166b : FAILURE in 18m 28s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/a40c5bd09f8c43ef80e45ae0ff8a5375 : SUCCESS in 53m 33s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/3547459f698b401da7d4ed7537a7db38 : SUCCESS in 1h 01m 06s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/064c6158a41244e2a238b9c9ddc0dd90 : SUCCESS in 1h 00m 27s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/55a2f765111e44589293262ffaa9a44a : SUCCESS in 49m 33s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/cb9dee78c4f845c8aa6ead5216e28a71 : FAILURE in 34m 24s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c771a1cff331448e82c86803cc0f4b8c : FAILURE in 37m 05s\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/91989af0f940479e8d3ee749fe2fdf8c : FAILURE in 28m 30s\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/1b41c539143d4adfa7098ffab5bb274c : FAILURE in 30m 07s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/0d615178b515469293ab517c1d7b2fa8 : SUCCESS in 54m 00s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ce2cf2bdfdf6483291f23a5f6d266f23 : SUCCESS in 1h 09m 20s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/fdc7be1f08e84a61b3e2a6d8656b819a : SUCCESS in 1h 28m 59s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/93daa304228e45bf84a2170166a5710e : SUCCESS in 43m 23s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/5f060d46ab7843e5bc1b9b386fa973f5 : SUCCESS in 20m 28s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a587aaa25c994d56baae690b5d04a89a : FAILURE in 16m 26s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/163d7618806648958fdab0d714ff6018 : SUCCESS in 4m 21s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/212069e3fc81489d9839fbc6a8337ade : TIMED_OUT in 3h 15m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c9c7c146711709eebc23e523493ead9fec46d5f1","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-02-06 15:30:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8e182476714ef16b35bf499d4e70cd3db656ce78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 17:06:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a6938ae608ef493884da29bd16aee41c : SUCCESS in 5m 11s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/89c9aabcd3d349fa92729d75a41e95f4 : SUCCESS in 5m 40s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dcb0ee54264c47b6ab7448180b787583 : SUCCESS in 6m 24s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/79b43273d64949f3b9bbf4b92d531565 : SUCCESS in 33m 05s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d39406d46047478d8164150c39a8e460 : SUCCESS in 21m 56s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/951e385bb8be40b1a61b9607a5a60baa : FAILURE in 20m 18s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/b8fcefac4c7f41788dd8d0014ac82476 : SUCCESS in 1h 04m 19s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/01b4c5695a884baf8ee984f4181e2706 : SUCCESS in 1h 11m 46s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/491110c0a59640e7a95d737ee2094c96 : SUCCESS in 1h 10m 07s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/5c94a2ef81984252a1e387bd23c8473a : SUCCESS in 1h 01m 46s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/8091e30adab54e1998f1146ff8fbed8e : FAILURE in 48m 56s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f8d548b47d264324a991cb30e2fbc00a : FAILURE in 49m 31s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8ae6459d98484a77bc08e535e994222b : FAILURE in 35m 51s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/b2f65a19c4aa4338b05ffc5438bf6c40 : FAILURE in 35m 00s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/30d0d924ec764df4be993807f5edbd8c : SUCCESS in 1h 00m 42s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/28f097c408f349a5814e7bad5a81df74 : SUCCESS in 1h 16m 22s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/640320a8e81a415a87bf2a97a3dccd93 : SUCCESS in 1h 07m 08s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5f0365a69ed84c83b2b00e1921354949 : SUCCESS in 56m 52s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/c2cfa5610e184554b802a3f215acca75 : SUCCESS in 26m 30s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b6e17f37fb284ef6b642dc435762a2d3 : FAILURE in 25m 01s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1d33575741245d3a83b102d0ff5a5fc : SUCCESS in 5m 23s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/bcabb25750774d7b80953547a694ca06 : FAILURE in 1h 29m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8cf3a64e3cea9210981ca89af67dbb67e824e50d","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-02-06 18:59:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fa70c2d23a10a60046b2bb5203a43687fa2e7410","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-02-07 22:17:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"024443a5349bab1b98edbe97ddb1a877a69c735b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 22:17:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"209ec0fa2616403e344ba19624d3633db1982b8a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-08 01:37:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aa3734f0f631486db03d859ea6fe628b : SUCCESS in 6m 19s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/188f7af2d9a54a60a6d8e177145e59e7 : SUCCESS in 6m 46s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bdbbc871a5334ab4a61aae7b62ed4c11 : SUCCESS in 6m 01s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/ab4042e2630348d0baa5dcc8eb4111fe : SUCCESS in 25m 25s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2f3bd75485bf40b9863f5e67ac546dec : SUCCESS in 22m 07s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/480cb094315c4139ba1140db80b0e465 : SUCCESS in 1h 07m 29s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/5adf01556e6146c1879426faafdd44b0 : SUCCESS in 1h 11m 52s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/0152ea706430476e82015ba47174397b : SUCCESS in 1h 12m 12s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/4f892b86a1384508b032515cbc78abf9 : SUCCESS in 1h 01m 17s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/77ae67cde5f641b0a0429b67727377dc : SUCCESS in 24m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/591a369abf1a44d892848a428bf1a3c6 : SUCCESS in 6m 06s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/38cd6e33baa54fd683712330936383c0 : SUCCESS in 3h 11m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e3d264df657d4cc1ec49ef5c92035ea5599add74","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-08 01:37:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"98559ae5974f8fd60fc825895d6d42fa755a5e51","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-08 01:39:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9188ebbbccbc4ec9957e2c135068347e : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fbafd406fcb507cb00e9e0d21dcff2eb3f6b6681","revisions":{"fbafd406fcb507cb00e9e0d21dcff2eb3f6b6681":{"kind":"REWORK","_number":1,"created":"2020-02-05 10:02:09.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/09/705909/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/09/705909/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/09/705909/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/09/705909/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/09/705909/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/09/705909/1"}}},"commit":{"parents":[{"commit":"e3c8f1423ba2f40b6f1e44cc082fddaebf17b439","subject":"Remove neutron::agents::metadata::metadata_ip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/e3c8f1423ba2f40b6f1e44cc082fddaebf17b439"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-02-05 10:00:58.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-02-05 10:01:39.000000000","tz":60},"subject":"Add OVS ovsdb_timeout parameter","message":"Add OVS ovsdb_timeout parameter\n\nThis adds the ovsdb_timeout to the ovs config\nsection which is the seconds that it should try\novsdb commands before failing. [1]\n\n[1] https://github.com/openstack/neutron/blob/master/neutron/conf/agent/ovs_conf.py#L24\n\nChange-Id: Id73e8ae40f4db67247bde36156e3b84c456b9f29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/fbafd406fcb507cb00e9e0d21dcff2eb3f6b6681"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/fbafd406fcb507cb00e9e0d21dcff2eb3f6b6681"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
