)]}'
{"id":"openstack%2Fpuppet-neutron~706506","triplet_id":"openstack%2Fpuppet-neutron~master~I6b9cc2aff14ce38a56c2757d9ad6c56898e27247","project":"openstack/puppet-neutron","branch":"master","topic":"ovn-md-workers","hashtags":[],"change_id":"I6b9cc2aff14ce38a56c2757d9ad6c56898e27247","subject":"[OVN] Change the default metadata_workers to 2","status":"MERGED","created":"2020-02-07 13:46:48.000000000","updated":"2020-02-10 18:37:27.000000000","submitted":"2020-02-10 18:32:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706506-1581359535856-798fc2db","meta_rev_id":"2d6ac9c2720dd41e5a13fea2d46d70464e851cc5","_number":706506,"virtual_id_number":706506,"owner":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-10 18:32:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"recommended":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":1,"date":"2020-02-10 15:27:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":2,"date":"2020-02-07 16:03:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2020-02-07 22:17:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2020-02-10 09:42:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2020-02-10 14:56:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-07 16:03:32.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-02-10 09:42:40.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2020-02-10 14:56:41.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-02-10 15:27:23.000000000","updated_by":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"reviewer":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"state":"REVIEWER"},{"updated":"2020-02-10 18:32:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"04b12f87ae243864ee6e9fef3abe3cf9ba1c7c44","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-02-07 13:46:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"394f232c8ad063b31b9e68be16b54a31da691039","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-02-07 14:01:40.000000000","message":"Patch Set 1:\n\n(1 comment)\n\ncould you add a release note? also see comment","accounts_in_message":[],"_revision_number":1},{"id":"54d448da09c88f10174acf9e6188a0a4c2e6bb91","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-02-07 14:10:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d0819f5817aacdea8b1f42732b0751f55b15d7d9","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-02-07 14:11:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9afb76ee0023121f4bbc16d0138032f7b43263cd","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-02-07 14:15:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b5f07d8377dbdfa8a64e32f2576ef769f91625c9","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-02-07 15:19:54.000000000","message":"Patch Set 2: Code-Review+2\n\nPatch lgtm but need a release note that default value is changed :)","accounts_in_message":[],"_revision_number":2},{"id":"2c25795651dabc2ffd01e4068390d9db4629f9c1","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-02-07 15:48:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b285c8ea15c12f14e934983f04562f477c597e5c","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-02-07 15:48:43.000000000","message":"Patch Set 2:\n\n\u003e Patch lgtm but need a release note that default value is changed :)\n\nThanks Tobias! Added one in the next patch-set","accounts_in_message":[],"_revision_number":2},{"id":"022ba46756498781622d738eace82598b080abbf","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-02-07 16:03:32.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"8f57860289611a3739d6b745425aa2f0efc98151","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 19:01:44.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f6fdd4b8bf0646ec9d36372e4186e074 : SUCCESS in 6m 59s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a77174e44fa14efeb8923aab2bd2ec98 : SUCCESS in 6m 20s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7d7a85b259f14606b5ead4b8947297e6 : SUCCESS in 5m 58s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/f478aed8ecad4ae1911d34cb95688180 : SUCCESS in 31m 20s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a48b3b0d08cc45f9ac14b5e18d6b85a6 : SUCCESS in 20m 30s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/217dab50983148e093ed9f5299332696 : FAILURE in 20m 59s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/aa573f3d58604e9ebba5a00f88ebeba4 : SUCCESS in 54m 57s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/f4f337b9328547ccbee73233dc1dc2f2 : SUCCESS in 1h 04m 42s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/214b59809d24459890fde29c15fd062c : SUCCESS in 1h 10m 51s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/654fa0e51b674557b4876bcff92afd3c : FAILURE in 48m 43s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/694b99d028aa4eb49105b1a26c81a4ac : FAILURE in 34m 55s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/770735e4999a41cbae88a888597993b2 : FAILURE in 39m 36s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b87bbfcac9c34463830ce27228d7f24f : FAILURE in 32m 00s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/58268567d74241628e9a240f96a52ebb : FAILURE in 32m 36s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f3c936bd7d6f4a0198305ce912f23ca5 : SUCCESS in 56m 54s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8a39085e68604832874789fd01bf5d89 : SUCCESS in 1h 08m 26s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/894499288ff94f0786de10b06e2c14ea : SUCCESS in 1h 04m 12s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a92ce5bfeb594c698d6f00e1695db1f2 : SUCCESS in 49m 04s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/abb384e9350b4a11b60966b406bb7e1c : SUCCESS in 22m 27s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c4ac2a9ddbc344149fe73f7976b44d50 : FAILURE in 19m 18s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/017e5a03c4bb40149aa0a19fce12ce64 : SUCCESS in 6m 06s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/91e4c29fe9674177ab47fdf9c1d6d30a : SUCCESS in 3h 07m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4d9d614c5d3345d6865e47344529ad899598f910","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-02-07 22:17:01.000000000","message":"Patch Set 3: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"5e391bf25c4db817bb625660f025c41f094d9a54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-08 01:38:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/912d0c8a2c5242ad8ac6a4f55cc4fcf1 : SUCCESS in 6m 54s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e7f34783a67344f482764c6002589eb1 : SUCCESS in 7m 10s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2bd15b042ce449218d782aaffbdcedce : SUCCESS in 8m 12s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/395081164ed6413c95cfcc54ea1c4f9a : SUCCESS in 36m 05s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e4b5aa23333f4455bb7b6db7ca192ff2 : SUCCESS in 22m 41s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b678ffeaacdf45f0afe1697a0b158070 : FAILURE in 26m 20s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/b471873f5f1d4c30bb88bb8611322fe7 : SUCCESS in 1h 06m 27s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/2901347408dc4810b44ddaeabe71202b : SUCCESS in 1h 11m 06s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/f22fea68c4074cf2aa555d000fbe8a87 : SUCCESS in 1h 17m 13s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/14c8dfd58e8e402482d8af0e559298cc : SUCCESS in 1h 02m 37s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/2b30a4e439e44e8684d786e69b7a78c2 : FAILURE in 44m 47s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/224dbc7fd4e242729cb95a2a417ac3d3 : FAILURE in 47m 46s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5a955024c0964eb4aa80bac5b5d16566 : FAILURE in 48m 58s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/adc9c1dd63414239b80acbe5a7aeca72 : FAILURE in 28m 32s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7f0c46647e94430d9f77ba86f0f90a75 : SUCCESS in 1h 06m 07s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7b34216219104c61a4f8fe84176d61fc : SUCCESS in 1h 14m 14s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/837a1c448b40439c92f9f4c1616b2872 : SUCCESS in 1h 01m 58s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e9150e56a7ea4a3197c77aa3cd09503c : SUCCESS in 45m 43s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/a6369a87f75c49a29c6bac8936b4232a : SUCCESS in 28m 12s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8ab2fb4faf0c4b03af26210e00fe7491 : FAILURE in 19m 46s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69ea5d378d52492a84fe803b900dc8ba : SUCCESS in 4m 55s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/14656c52c64d4934aebc8588ed1960d8 : SUCCESS in 3h 19m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e2813eef49a4e3e2a4431676a4652e04905d66be","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2020-02-10 09:42:40.000000000","message":"Patch Set 3: Code-Review+1\n\nThanks! As discussed we might want to have a separate option in THT [0] to be able to tweak this number if needed.\n\n[0] https://opendev.org/openstack/tripleo-heat-templates/src/branch/master/deployment/neutron/neutron-metadata-container-puppet.yaml#L48","accounts_in_message":[],"_revision_number":3},{"id":"951aefe74b5c72c9c78cc0a0131300b08d50d27e","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-02-10 14:56:41.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d125ad37abecd743101ea7ff8a84aae20933f3ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 14:56:53.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"47a8995dcdaeb49df1105e17ff4e5b1edd7dbae7","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-02-10 15:27:23.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"8d9e8c8284efdbf29a17a61aea3dde5aaa8eb57a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 18:32:15.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d89607f83f944b38a1999dfca7168f04 : SUCCESS in 6m 49s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/695f23f5860c4d6bbcba970c14869879 : SUCCESS in 6m 50s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/206d0dbd34c54a139ea569282cd53dcf : SUCCESS in 6m 52s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/e8084ce98b3045eb8451fc7b10de0d51 : SUCCESS in 34m 38s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/254de63dcd244a8e97ed7eca2dcec4c1 : SUCCESS in 19m 31s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/9abea086935a4c979accf0bfa392ed5e : SUCCESS in 1h 07m 33s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/eb815cf7c0ec4b41aaa46e866e4e7361 : SUCCESS in 1h 06m 09s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/622eba2c32204e8b95c944bf2774cf67 : SUCCESS in 1h 10m 21s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/458d4bff9a404c0fa2cfc329a40cc852 : SUCCESS in 56m 33s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/5aa7ee8f4d7a4c40b46e7928cdd1b82c : SUCCESS in 23m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c8b5a255e8ea45f793f69478f0299da4 : SUCCESS in 6m 58s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a23bd0e919c44e26b89edd44c66659b5 : TIMED_OUT in 3h 15m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e4fec57134e17da45f3f7e9430fa8dde7ff32373","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 18:32:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"472a82893471096713e8c6d6d982529ae1921d3b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 18:37:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01e670a4ee984941b097f032ff2b8bf7 : SUCCESS in 1m 50s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"847f434140ee8435ee842801748a0deccdff8155","revisions":{"bb10e1e833fb751c9e8031fc4cabb507253cfaae":{"kind":"REWORK","_number":1,"created":"2020-02-07 13:46:48.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/06/706506/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/06/706506/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/06/706506/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/06/706506/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/06/706506/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/06/706506/1"}}},"commit":{"parents":[{"commit":"e3c8f1423ba2f40b6f1e44cc082fddaebf17b439","subject":"Remove neutron::agents::metadata::metadata_ip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/e3c8f1423ba2f40b6f1e44cc082fddaebf17b439"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-02-07 13:34:45.000000000","tz":0},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-02-07 13:46:37.000000000","tz":0},"subject":"[OVN] Change the default metadata_workers to 2","message":"[OVN] Change the default metadata_workers to 2\n\nPrior to this patch, the metadata_workers parameter defaults to\n$os_workers (calculated here [0]) but, even tho it\u0027s capped at 12 workers\nthis number still redundant.\n\nThere are a few differences of how the OVN metadata agent works compared\nwith the OVS metadata agents:\n\n1) The OVN metadata workers are deployed onto the compute nodes and not\n   on the controllers/gateways as the OVS ones are. This means it\u0027s\n   better distributed.\n\n2) Even tho we could have multiple workers, the OVN SB IDL in the OVN\n   metadata agents acquires a lock while listening to OVSDB events.\n   Therefore, even tho we could have dozens of agents only one of them\n   is listening to the events.\n\n3) Instantiating IDLs means maintaining in-memory replicas of the\n   OVSDB (or, the tables of that DB that is being watched). This puts a\n   burden on OVSDBs to maintain all these replicas updated, reducing\n   this number workers also reduces the burden on the OVSDBs.\n\nThis patch is reducing the default number of OVN metadata agents to two\n(per compute). Two because that keeps it HA, in case the active agente\ndies the lock moves to the other one and the service will keep working.\n\n[0]\nhttps://github.com/openstack/puppet-openstacklib/blob/74e0f5a3170b09632ad4ede7e20d3ebeb4514586/lib/facter/os_workers.rb#L37\n\nChange-Id: I6b9cc2aff14ce38a56c2757d9ad6c56898e27247\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/bb10e1e833fb751c9e8031fc4cabb507253cfaae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/bb10e1e833fb751c9e8031fc4cabb507253cfaae"}]},"branch":"refs/heads/master"},"082c5aee4bc22d85311186efb57d4630f5c8b98e":{"kind":"REWORK","_number":2,"created":"2020-02-07 14:15:11.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/06/706506/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/06/706506/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/06/706506/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/06/706506/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/06/706506/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/06/706506/2"}}},"commit":{"parents":[{"commit":"e3c8f1423ba2f40b6f1e44cc082fddaebf17b439","subject":"Remove neutron::agents::metadata::metadata_ip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/e3c8f1423ba2f40b6f1e44cc082fddaebf17b439"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-02-07 13:34:45.000000000","tz":0},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-02-07 14:15:00.000000000","tz":0},"subject":"[OVN] Change the default metadata_workers to 2","message":"[OVN] Change the default metadata_workers to 2\n\nPrior to this patch, the metadata_workers parameter defaults to\n$os_workers (calculated here [0]) but, even tho it\u0027s capped at 12 workers\nthis number still redundant.\n\nThere are a few differences of how the OVN metadata agent works compared\nwith the OVS metadata agents:\n\n1) The OVN metadata workers are deployed onto the compute nodes and not\n   on the controllers/gateways as the OVS ones are. This means it\u0027s\n   better distributed.\n\n2) Even tho we could have multiple workers, the OVN SB IDL in the OVN\n   metadata agents acquires a lock while listening to OVSDB events.\n   Therefore, even tho we could have dozens of agents only one of them\n   is listening to the events.\n\n3) Instantiating IDLs means maintaining in-memory replicas of the\n   OVSDB (or, the tables of that DB that is being watched). This puts a\n   burden on OVSDBs to maintain all these replicas updated, reducing\n   this number workers also reduces the burden on the OVSDBs.\n\nThis patch is reducing the default number of OVN metadata agents to two\n(per compute). Two because that keeps it HA, in case the active agente\ndies the lock moves to the other one and the service will keep working.\n\n[0]\nhttps://github.com/openstack/puppet-openstacklib/blob/74e0f5a3170b09632ad4ede7e20d3ebeb4514586/lib/facter/os_workers.rb#L37\n\nChange-Id: I6b9cc2aff14ce38a56c2757d9ad6c56898e27247\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/082c5aee4bc22d85311186efb57d4630f5c8b98e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/082c5aee4bc22d85311186efb57d4630f5c8b98e"}]},"branch":"refs/heads/master"},"847f434140ee8435ee842801748a0deccdff8155":{"kind":"REWORK","_number":3,"created":"2020-02-07 15:48:28.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/06/706506/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/06/706506/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/06/706506/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/06/706506/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/06/706506/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/06/706506/3"}}},"commit":{"parents":[{"commit":"e3c8f1423ba2f40b6f1e44cc082fddaebf17b439","subject":"Remove neutron::agents::metadata::metadata_ip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/e3c8f1423ba2f40b6f1e44cc082fddaebf17b439"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-02-07 13:34:45.000000000","tz":0},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-02-07 15:48:17.000000000","tz":0},"subject":"[OVN] Change the default metadata_workers to 2","message":"[OVN] Change the default metadata_workers to 2\n\nPrior to this patch, the metadata_workers parameter defaults to\n$os_workers (calculated here [0]) but, even tho it\u0027s capped at 12 workers\nthis number still redundant.\n\nThere are a few differences of how the OVN metadata agent works compared\nwith the OVS metadata agents:\n\n1) The OVN metadata workers are deployed onto the compute nodes and not\n   on the controllers/gateways as the OVS ones are. This means it\u0027s\n   better distributed.\n\n2) Even tho we could have multiple workers, the OVN SB IDL in the OVN\n   metadata agents acquires a lock while listening to OVSDB events.\n   Therefore, even tho we could have dozens of agents only one of them\n   is listening to the events.\n\n3) Instantiating IDLs means maintaining in-memory replicas of the\n   OVSDB (or, the tables of that DB that is being watched). This puts a\n   burden on OVSDBs to maintain all these replicas updated, reducing\n   this number workers also reduces the burden on the OVSDBs.\n\nThis patch is reducing the default number of OVN metadata agents to two\n(per compute). Two because that keeps it HA, in case the active agente\ndies the lock moves to the other one and the service will keep working.\n\n[0]\nhttps://github.com/openstack/puppet-openstacklib/blob/74e0f5a3170b09632ad4ede7e20d3ebeb4514586/lib/facter/os_workers.rb#L37\n\nChange-Id: I6b9cc2aff14ce38a56c2757d9ad6c56898e27247\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/847f434140ee8435ee842801748a0deccdff8155"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/847f434140ee8435ee842801748a0deccdff8155"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
