)]}'
{"id":"openstack%2Fpuppet-neutron~706677","triplet_id":"openstack%2Fpuppet-neutron~stable%2Ftrain~Id73e8ae40f4db67247bde36156e3b84c456b9f29","project":"openstack/puppet-neutron","branch":"stable/train","hashtags":[],"change_id":"Id73e8ae40f4db67247bde36156e3b84c456b9f29","subject":"Add OVS ovsdb_timeout parameter","status":"MERGED","created":"2020-02-08 22:23:03.000000000","updated":"2020-02-10 20:51:01.000000000","submitted":"2020-02-10 20:47:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706677-1581367679636-25e6796c","meta_rev_id":"56a02ce7cf7c0a3b7f66ac36ce9e79a9f17a81eb","_number":706677,"virtual_id_number":706677,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-10 20:47:59.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-02-10 17:07:46.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-02-10 15:49:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-02-10 17:07:46.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-10 15:49:42.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-02-10 17:07:46.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-02-10 20:47:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7b220823a15565b74bab0edc75230df2b8628827","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-02-08 22:23:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aed9d2846e83773024e7c607b8dc5440c71cbe37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-09 01:45:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f6b2bf047eea4776acffbb5444b00d5e : SUCCESS in 4m 56s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/04d43e830095499397be89f9e25c5769 : SUCCESS in 7m 17s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1232a9a46f424c9baa5f53e8dea21257 : SUCCESS in 5m 09s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/195e4376d43f40629dc0be62dc3f4b27 : SUCCESS in 28m 16s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/f3a9ae7742ff4f53ae19f0b88d0b3c94 : SUCCESS in 49m 47s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/9914fe3cfc544725862378f7ea52fa0e : SUCCESS in 59m 15s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/b8b6e08b27a248e29958f5d6126f748a : SUCCESS in 54m 36s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/c12845aa082b4e71852011e4bab81d3f : SUCCESS in 45m 00s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/c52d92720b2a43c59508bdc5ed3b51ef : FAILURE in 37m 04s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d760865cf99b48a9b133637366abbf37 : FAILURE in 31m 57s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0cdad3e6dc524ef6a167af747a070d55 : FAILURE in 35m 59s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/e91d57ddd06b48adb68905923b9c8471 : FAILURE in 23m 24s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/2404ea3d1ed149f7a1f011a6e5423661 : SUCCESS in 19m 31s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cbaa65f20cec4d08a617b48fd85d4f67 : FAILURE in 17m 29s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63505c11a88d48318b1f58bf06e2276e : SUCCESS in 3m 21s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c7ad18644f05463f81987d58352f6eb4 : TIMED_OUT in 3h 16m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3688318ead03c818901e99770ad04aac606538ec","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-02-10 15:49:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"aba61d73f01ca8fbe09834c41d1d14f02824d7dd","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-02-10 17:07:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d97642f05af0517acbe183142c0da8dbb943db09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 17:08:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a6a2e6735e4acdb0527bccb7cde9cc1469b411d8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 20:47:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e0927d18544e32eb4fe7fe208728ca552cc6f63c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 20:47:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/14750cbb9efc4e568de9b73d1b145eba : SUCCESS in 7m 26s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a9897e9303084275a31472c67cc38961 : SUCCESS in 6m 45s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/15c9525972a643c697a65a86a1df69b5 : SUCCESS in 6m 03s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/b2fcf643b56c48f28df508187b454a91 : SUCCESS in 33m 46s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/26e9f30e29184ce9ad4a6d105580927f : SUCCESS in 1h 00m 09s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/9c93e9e1c98c41fc9a6e70e78e6ba0db : SUCCESS in 1h 04m 04s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/90425d9b58704d8dbb72210569804734 : SUCCESS in 1h 06m 46s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/95ad1e4158c64194995d8607bf4aef6e : SUCCESS in 51m 20s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/e1cacf7c74c8468e8676a75917e84b17 : SUCCESS in 22m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e7de5290ebb4cffbbffea6e92b8008a : SUCCESS in 6m 38s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/1884db6c814d41bfb78893edffcbf15a : TIMED_OUT in 3h 24m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"56a02ce7cf7c0a3b7f66ac36ce9e79a9f17a81eb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 20:51:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/888f5cc5e54d49598c714b8615bcf0cd : SUCCESS in 2m 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"235c69cbc8051d05fd03de2cb9214c8d3fa445f6","revisions":{"235c69cbc8051d05fd03de2cb9214c8d3fa445f6":{"kind":"REWORK","_number":1,"created":"2020-02-08 22:23:03.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/77/706677/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/77/706677/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/77/706677/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/77/706677/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/77/706677/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/77/706677/1"}}},"commit":{"parents":[{"commit":"1111e3dd2f8ae4f5c4bf921286d73dc833db868a","subject":"Add agents.l3 \"radvd_user\" configuration option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/1111e3dd2f8ae4f5c4bf921286d73dc833db868a"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-02-05 10:00:58.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-02-08 22:22:42.000000000","tz":60},"subject":"Add OVS ovsdb_timeout parameter","message":"Add OVS ovsdb_timeout parameter\n\nThis adds the ovsdb_timeout to the ovs config\nsection which is the seconds that it should try\novsdb commands before failing. [1]\n\n[1] https://github.com/openstack/neutron/blob/master/neutron/conf/agent/ovs_conf.py#L24\n\nChange-Id: Id73e8ae40f4db67247bde36156e3b84c456b9f29\n(cherry picked from commit fbafd406fcb507cb00e9e0d21dcff2eb3f6b6681)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/235c69cbc8051d05fd03de2cb9214c8d3fa445f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/235c69cbc8051d05fd03de2cb9214c8d3fa445f6"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
