)]}'
{"id":"openstack%2Fpuppet-neutron~706985","triplet_id":"openstack%2Fpuppet-neutron~stable%2Fstein~Id73e8ae40f4db67247bde36156e3b84c456b9f29","project":"openstack/puppet-neutron","branch":"stable/stein","hashtags":[],"change_id":"Id73e8ae40f4db67247bde36156e3b84c456b9f29","subject":"Add OVS ovsdb_timeout parameter","status":"MERGED","created":"2020-02-10 21:48:35.000000000","updated":"2020-02-12 16:09:12.000000000","submitted":"2020-02-12 16:06:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706985-1581523604482-18a4250f","meta_rev_id":"f4646e1e6cc7894a80f6e3a0ae29a40fb682e3fc","_number":706985,"virtual_id_number":706985,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"date":"2020-02-12 09:12:49.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-12 16:06:44.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2020-02-11 08:35:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2020-02-11 00:04:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2020-02-11 08:35:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-11 00:04:12.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-02-11 08:35:41.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2020-02-12 16:06:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7a077109ad2044a3c8692768063f4ae480409139","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-02-10 21:48:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c2e2764e58fa37a6312881ae8f3bde5cdbfad9fe","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-02-11 00:04:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a26fb7ee88bf6ff8af098a901fee99a42485a286","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 01:14:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a5c254db202546c88caea6ea2a40f69c : SUCCESS in 5m 30s\n- puppet-openstack-syntax-4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f3817afa41f94d6fad5890a129320a67 : SUCCESS in 9m 33s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/619192a2e0494d218dec9e15c10f5dc8 : SUCCESS in 6m 21s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/262534a1d1cf4f7b97ee8d141514f2f0 : SUCCESS in 6m 22s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/94a8dbfe26de4cb4b91dd6c70b88587d : SUCCESS in 38m 15s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/d323bc645718416c8d44f0435d8bd7f9 : SUCCESS in 35m 14s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/72940d9967294e77ab48d847d51371a3 : SUCCESS in 56m 53s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/9073bef3150149629ee805b2310ae1b0 : SUCCESS in 1h 08m 58s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/0006d243f0394c1fa091daae42f2af40 : SUCCESS in 1h 09m 11s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/d9adb4c8e3ea46ff96ca13a6aee1f631 : SUCCESS in 45m 55s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/7213c4d9be86404480efeb7ba9c5f6d6 : FAILURE in 30m 22s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0e7b222a389e4e43aa370dbae6f13010 : FAILURE in 25m 45s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0c096ae25f3b4338881f5fd5a879d7b0 : FAILURE in 26m 22s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/e4e6994a09c340008df97170404e6988 : SUCCESS in 39m 52s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/787a50f877204318bb7072b9915e8f16 : SUCCESS in 19m 38s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a718a51ee0784443a4e56dda99d6113d : FAILURE in 18m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e768aa13afd411389ee0744e6854d6a : SUCCESS in 7m 41s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d9bc8ec2e9ba4b0e810366059532f224 : TIMED_OUT in 3h 19m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d6b6005f0a7315fdae25a69e516659043cb21bb5","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-02-11 08:35:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1e914b026eb4833719366f84f1e11a2ef002183e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 08:35:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4833b43fdafb4b13739ae29177dbbef5ad35aad8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 09:13:56.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/92dbe89dbffe4ef2978b43959230656f : SUCCESS in 4m 34s\n- puppet-openstack-syntax-4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1f18f3f9472c40b490a49a479471528d : SUCCESS in 4m 24s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bcc4ef3e415c46bca36e393cb9b8bb6e : SUCCESS in 4m 19s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a9f969d3a7364ad391fab827c3e31b2e : SUCCESS in 4m 25s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/889a3a6c5f4c47b082acf3c451f6f9fb : RETRY_LIMIT in 4m 59s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/0b72c4a2db1b44dc86fc3076f09f9965 : RETRY_LIMIT in 5m 06s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/50506f490f4247c9aaafe27f77cc7466 : RETRY_LIMIT in 8m 17s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/c33d70ae0edf4ae1aaffabf1ca974b26 : RETRY_LIMIT in 5m 57s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/fdf608a0ab7949828086a90da8c6cd72 : RETRY_LIMIT in 8m 12s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/f01c82f1930f468ba1f49d2b5756c45d : RETRY_LIMIT in 5m 14s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a23b786a14144d85b13693e4b9d4334d : FAILURE in 18m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75bda5e2a70749428d8b8ffe0f22e29e : SUCCESS in 3m 59s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0b9fe4e826d74af3a950e2d7a3308cc8 : FAILURE in 6m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9653d70b846773e66b6f6d821d746f24a8860f42","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-02-11 13:12:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5ecf973c79210dbcba0b38ea86f730c5afdc462e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 13:59:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7f4342989f0649cda64f8a7f35514632 : SUCCESS in 8m 40s\n- puppet-openstack-syntax-4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/408d01cc91754f3c810fe6777dfc99ea : SUCCESS in 6m 17s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f32e7cea498f4b02a6ef59dc37a1db2b : SUCCESS in 6m 00s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/886513db41674c98b675eef284d79dd4 : SUCCESS in 7m 38s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/099b4dd18beb4ba090d121743a846f81 : RETRY_LIMIT in 4m 19s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/fa57ec9a16594a328566021312c30046 : RETRY_LIMIT in 5m 30s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/482ea02c22d14b4886d598cf23ffa838 : RETRY_LIMIT in 5m 11s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/83d352f8d8bd40c7ab56d4637d1ba835 : RETRY_LIMIT in 6m 01s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/cfcbd4bec674475c9eb3c749e07ee0ee : RETRY_LIMIT in 4m 49s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/bf0de33f4f99498fbc39b6112009278f : RETRY_LIMIT in 6m 12s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/dc16decc90ec4a9d88f7eb43c5c98377 : FAILURE in 31m 48s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4bb892d0cb2645519853aeebb4620d90 : FAILURE in 32m 27s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/788c332c4753411aa88f1fc3349a96b4 : FAILURE in 20m 41s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/a2e94a6885844bd386079f3a4c800856 : SUCCESS in 43m 02s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/4f7a993740e84624a9ab740fc6e8b492 : RETRY_LIMIT in 6m 02s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/902277e167044ba1bdc7eba18d4384ae : FAILURE in 22m 52s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4781af41dd784362a5043182c62ab7d0 : SUCCESS in 4m 26s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7c09eee25ac84cf1b349c52802f144d7 : FAILURE in 7m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f2a319068744ee1bb479f30d94ee0a4bd48555f4","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-02-12 09:12:49.000000000","message":"Patch Set 1:\n\nrecheck infra fixed it now?","accounts_in_message":[],"_revision_number":1},{"id":"09544bfd0ddeb947b1f8ddf0e67357c3fbb863c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 12:43:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/08d342a9257b481bad0e2f5150e27342 : SUCCESS in 6m 32s\n- puppet-openstack-syntax-4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e8f3cc7b89c54c0fad6c4645e9170915 : SUCCESS in 6m 17s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4cffbe89fc8145419d80ebbb0fbd576a : SUCCESS in 5m 10s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/12923d48b59e443682b6f5bcb5b90a7a : SUCCESS in 5m 53s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/21c315d29a6a4cbda1cf948f682e4b59 : SUCCESS in 36m 45s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/33d63d1e06034bedb1a01f989e5f9806 : SUCCESS in 35m 02s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/f5745931614d403da7178ea31b90ee63 : SUCCESS in 1h 01m 13s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/f3011c1302e0448d93e80e7867368ad0 : SUCCESS in 1h 04m 04s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/c72e6cce2bad4a188fd75e27a5972171 : SUCCESS in 54m 52s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/4d1b2f1b48f2434a874f2e3a356631a5 : SUCCESS in 45m 10s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/fbf948443f5f43bbbffca6466b78a5fc : FAILURE in 23m 03s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7c26ae787f344be69f37f1aec258b6e1 : FAILURE in 23m 55s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b6530e571f6b4c58ae6fd65ebcb5889e : FAILURE in 23m 59s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/a63beaeababf447ba0edfbb9685c3080 : SUCCESS in 36m 02s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/0bd11b5c20ce43a2b481bc697a5ae9c4 : SUCCESS in 20m 29s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/591885d104cb4aec8070e5132ec66b16 : FAILURE in 18m 20s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6102147a064e4d19b4865c14c9cc7bea : SUCCESS in 4m 40s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0740bacd604e4bd68377cbf248462fd3 : TIMED_OUT in 3h 28m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"16b38d4fda971fdb535af24e3e7941810fb7b910","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 12:43:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e56a5ef3dc683778a95f8271be56b4090b5905b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 16:06:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"540cdecd2078fd95aec083bebe85e0ead46d4aa4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 16:06:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/31ce99bfd9c043c199d27b3d93722cf4 : SUCCESS in 6m 15s\n- puppet-openstack-syntax-4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6ff58939b09d4106b4bade4de2f68601 : SUCCESS in 7m 22s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/10ef64572a104717a8d64dd01d31baba : SUCCESS in 6m 15s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cb729de9595e4e5eac95c049087de0fc : SUCCESS in 7m 47s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/5707dd07717548b8b1dfb3cff6ddf0e6 : SUCCESS in 26m 44s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/5a475f8743c246c6928a268e45701c4e : SUCCESS in 32m 18s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/097b3d3e83c043508c89c41986ceedee : SUCCESS in 53m 49s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/ffdc20591bbd42a9baa88037e09dc054 : SUCCESS in 1h 13m 54s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/6af1fac78a1e499699119f771f2f1443 : SUCCESS in 1h 02m 45s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/31fb0089910744e5a78880b876c74f01 : SUCCESS in 21m 25s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b7f2eee4b24a4949a251cbca7bcad1fe : FAILURE in 17m 18s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f93b1e7ee7184754947ffddf8feeba3a : SUCCESS in 6m 03s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/af76250041264d3e8104eba103495ab3 : TIMED_OUT in 3h 17m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f4646e1e6cc7894a80f6e3a0ae29a40fb682e3fc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 16:09:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc2b8ef4afc34033821f4ecef7354fa3 : SUCCESS in 2m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6c3b600d6c0568527c10923254a252e50b9dca7b","revisions":{"6c3b600d6c0568527c10923254a252e50b9dca7b":{"kind":"REWORK","_number":1,"created":"2020-02-10 21:48:35.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/85/706985/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/85/706985/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/85/706985/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/85/706985/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/85/706985/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/85/706985/1"}}},"commit":{"parents":[{"commit":"470cd27ef6e719b612217ffb7a3649e8ac29d067","subject":"Add agents.l3 \"radvd_user\" configuration option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/470cd27ef6e719b612217ffb7a3649e8ac29d067"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-02-05 10:00:58.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-02-10 21:48:13.000000000","tz":60},"subject":"Add OVS ovsdb_timeout parameter","message":"Add OVS ovsdb_timeout parameter\n\nThis adds the ovsdb_timeout to the ovs config\nsection which is the seconds that it should try\novsdb commands before failing. [1]\n\n[1] https://github.com/openstack/neutron/blob/master/neutron/conf/agent/ovs_conf.py#L24\n\nChange-Id: Id73e8ae40f4db67247bde36156e3b84c456b9f29\n(cherry picked from commit fbafd406fcb507cb00e9e0d21dcff2eb3f6b6681)\n(cherry picked from commit 235c69cbc8051d05fd03de2cb9214c8d3fa445f6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/6c3b600d6c0568527c10923254a252e50b9dca7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/6c3b600d6c0568527c10923254a252e50b9dca7b"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
