)]}'
{"id":"openstack%2Fpuppet-neutron~707367","triplet_id":"openstack%2Fpuppet-neutron~master~I1d96899de29cdd994f2a45b701923b0177a7edd8","project":"openstack/puppet-neutron","branch":"master","topic":"igmp","hashtags":[],"change_id":"I1d96899de29cdd994f2a45b701923b0177a7edd8","subject":"Add ovs/igmp_snooping_enable support","status":"MERGED","created":"2020-02-12 11:44:52.000000000","updated":"2020-03-10 00:27:01.000000000","submitted":"2020-03-10 00:27:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":20,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707367-1583800021158-72e87165","meta_rev_id":"401bec288cdbe23774c4ae1a6fe8a7f70c97a3a5","_number":707367,"virtual_id_number":707367,"owner":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2020-03-09 16:13:53.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-10 00:27:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-03-09 23:16:55.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-03-09 17:34:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-03-09 23:16:55.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-17 09:25:48.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2020-03-03 11:25:49.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-03-09 17:34:56.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-03-09 23:16:55.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-03-10 00:27:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"68c47ab5ade8db32f4f05020ba54a844648a554d","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-02-12 11:44:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"43e6f1386b3095224bf047e8aeb0fce0aefa0159","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 11:45:09.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"c4514e23b59fc450bebfe80920982b3ed0139654","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-02-12 13:29:11.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7558bfa935bca4f661f6a3b83b8394f8f2988b47","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-02-12 13:29:14.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"48499b35d766540552c916069be6914d47887a4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 16:55:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/938b8a0205d1463a804ae6aca55ea810 : SUCCESS in 5m 20s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5e20cdd4d07e4d178b903da0bb7e67a3 : SUCCESS in 6m 42s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2abbf92c9847421dac7b461dc6c29ff1 : SUCCESS in 7m 37s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/42bb5261ec25405ea8afe49c6f59233a : FAILURE in 32m 49s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a9ee203b6ea241608c1ddcd5556a5931 : FAILURE in 19m 32s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/39c0c2e4e0f74bf7aacc99cfde1a105a : FAILURE in 21m 38s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/9c17e295cf3c4957b79f43edd115de37 : SUCCESS in 1h 14m 39s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/f4b2ba4c4e5b45c3920a72aa13ab1ac5 : SUCCESS in 1h 08m 19s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/e498dd2654144b7b8515ffced6035032 : SUCCESS in 1h 12m 48s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/94b46406dd1d48ae9e3566143b8effa4 : SUCCESS in 1h 02m 34s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/43cf1f4b5d1840feb41318a2aa6aeccd : SUCCESS in 47m 50s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c106b40700be4349810e411a766c7d78 : SUCCESS in 1h 06m 21s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9bf835b2b53f432b81369b534cfb904a : SUCCESS in 56m 40s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/3ce5686d296b4c8b8ae4bbc9f26d37ff : SUCCESS in 49m 12s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ce766f26fc26466287d6ce2aa469bc89 : SUCCESS in 1h 13m 54s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e66088a8fdb04dc3adbc61a11b7afc57 : SUCCESS in 1h 12m 46s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f3eb7dfb146a4e7f858b322a4db016bd : SUCCESS in 1h 08m 06s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d879a35afc38413396f42b129cf1ef4f : SUCCESS in 58m 13s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/1113ee2b644144e29b97b8af20b8e098 : SUCCESS in 22m 20s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/39937be87b6049598de4926143b08456 : FAILURE in 17m 51s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8e03ba50f32e4548a3c63f359579baf8 : TIMED_OUT in 3h 19m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7f49f0ef205bb66179d9604084cd8f8278f01554","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-02-14 15:16:44.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3ac454d87e177d11888c75e64d321f0d39866d50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 16:46:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1dd100adeaa54a5993f147600478c999 : SUCCESS in 7m 17s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/60320efd5ae04afe8098877e27950bba : SUCCESS in 7m 45s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c0fce0e9a27941628d36706b4c5a9317 : SUCCESS in 7m 10s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/cb6d7526099b4aee9ec436b22242f044 : SUCCESS in 29m 35s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/85a3346d1f7c469e9fe492edb33c7753 : SUCCESS in 23m 48s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/029eeeb6257143f184177017be293459 : FAILURE in 24m 39s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/95e997d2a3734ccda872fa4b9a161d0b : SUCCESS in 59m 35s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/2132d06e3ef244d5b14ae6ae0dd0ec75 : SUCCESS in 1h 04m 11s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/4e58f26371684c938579882a5d200616 : SUCCESS in 1h 23m 22s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/305bb64e0fb24907ae09bec9d53a53cc : SUCCESS in 54m 25s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/0bc5d40a308b4d5cb8a2a83e58c32d96 : SUCCESS in 55m 03s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2d4260f112c74bcfbc3d1897b565dc06 : SUCCESS in 55m 29s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/993488e0f4804951b36b76c19f87f52d : SUCCESS in 41m 44s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/d2068740a0dc406e8daf39fdd5d797dc : SUCCESS in 52m 32s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/92455c2e7fe641779e428429a0d01cec : SUCCESS in 1h 17m 24s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/0da6755cc0694b208d6b17eb28eb08ac : SUCCESS in 1h 02m 53s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/2e17848c588c46a8b78cffa835fd4df9 : SUCCESS in 1h 04m 18s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/0f595893b0324448af9d26e65fb8d770 : SUCCESS in 54m 29s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/e5e5d048c3184c9081f0c12a1da242c5 : SUCCESS in 25m 30s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/46b8785f6d644e44a41dd0228e4ac8e8 : FAILURE in 19m 23s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b40f746beaf647509b2d86264b6f429e : FAILURE in 34m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"05d79cc768c9b2b605bbb6818a0cf64b8669c48f","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2020-02-17 09:25:48.000000000","message":"Patch Set 3: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"47c226bb9a24f769bd1d16d91bff580be4bdb924","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-02-17 20:31:55.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fd23cb430df1ecc9aa3f0ee6cf8b78af5b3f6803","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-02-20 10:52:23.000000000","message":"Patch Set 3: Workflow-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"b0071af95abccfae6e4bbc2dae4b7ba37d921cba","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-02-20 10:53:42.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f92c05948398600bf5c25d13a632f76478e534cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 12:25:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ead20b0598ff4f3fbd8b4a943aefeea7 : SUCCESS in 7m 18s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e80b45b3d88448d8949745b42b1082a5 : SUCCESS in 6m 26s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4406389629c54977997131a47ce74902 : SUCCESS in 6m 51s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/fe290f21f2644b64bd6c01f59c0f27a9 : SUCCESS in 24m 50s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7e806fd57e584d04a30f5ff89c9ef9c5 : SUCCESS in 18m 08s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7334ae7b60574e7097697b6eb1d3c6d2 : FAILURE in 21m 28s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/fd9e3889d26d4fc2b8c5c9cda6fa1e96 : SUCCESS in 1h 15m 09s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/b618e4d5dcf041b7b9241faaea69e03f : SUCCESS in 1h 17m 48s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/68a044cd166b410ebc6f15ec124cec64 : SUCCESS in 1h 09m 52s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/ec39487e151c41b1ac9bfe6f1775a28a : SUCCESS in 54m 02s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/b8c10019c5ea4f8298911e1b03de73e3 : SUCCESS in 1h 01m 25s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/74fa5a0290aa485f94f78eda7fd17c9d : SUCCESS in 1h 02m 32s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4bf3bd1376e645c6b22c69bfcbf1aad9 : SUCCESS in 50m 05s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/2bc6e622493c49bf8f93986697a4fedd : SUCCESS in 45m 42s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f898c771201544669242641ba74fd006 : SUCCESS in 1h 08m 03s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ca5630a79b7a496498263d01437ee1cd : SUCCESS in 1h 19m 06s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9955f58f62954a0b98f4ea6e3b524694 : SUCCESS in 1h 07m 18s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/73cab7b782b74681903ad1f7c08716ed : SUCCESS in 48m 39s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/2b2c2b4365b341b99f558941e3234078 : SUCCESS in 22m 20s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8aae4e0f956d48ef9164f59926e55714 : FAILURE in 17m 16s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c31f96cfde6b4ee9b48b0b4bf48f4b97 : FAILURE in 25m 18s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3dfc6d8b427c0ac992537a7b39c91d81e50db6f3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-02-21 12:33:15.000000000","message":"Patch Set 4: Code-Review-1\n\n(6 comments)\n\nAlso, IMO it would be nice to have a release note about the change because it introduced some new parameters.","accounts_in_message":[],"_revision_number":4},{"id":"6d2f972afed0eb992ad318f028b3cbdacede7dee","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-03-03 09:52:29.000000000","message":"Patch Set 4: Workflow-1\n\n(6 comments)\n\nThanks for the review","accounts_in_message":[],"_revision_number":4},{"id":"c70d81d33003d5286b83c11ca806f1c3c9718cf6","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-03-03 10:08:49.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f8577d3cfe7f3ea26fcfd922b14b2a72be53d091","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 10:33:34.000000000","message":"Patch Set 5: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"f344fc72bdb96bda1d1741a4ee428cfcf546b307","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-03-03 11:25:49.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"3e7ed0f8adec4c90a729336fb451498ef3429b3a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-03-03 11:49:55.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"125476ba1020490a174300d755cac726d1202671","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-03-03 11:51:42.000000000","message":"Patch Set 6:\n\nThanks for fixing it Takashi!","accounts_in_message":[],"_revision_number":6},{"id":"eab5dc22ed03fd9f89be881f3afdcf87a8984eb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 13:25:37.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/992c35a1646f450aac42e1242e29e23e : SUCCESS in 5m 37s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e790df5b571044888b38b125fa2b559d : SUCCESS in 6m 08s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9022e37cd8d1431db963f17feefbc425 : SUCCESS in 5m 46s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/3eb2477f13654fd89f911e2925c073f4 : FAILURE in 15m 22s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7abd257ff4ef47cc941075bc71bfb770 : FAILURE in 14m 57s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2a433a1df5684e73b712343242d6fc4d : FAILURE in 13m 38s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/eb6e621bd7124a16b48b7ad61c60a15c : SUCCESS in 1h 02m 43s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/ae28393cf69a4cba92e51cf60c7a4af2 : SUCCESS in 1h 07m 15s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/4fb9a5ef9d864f67b61a0063abd73077 : SUCCESS in 1h 02m 21s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/5a5e019cbbac409f81bdb8fa98760237 : SUCCESS in 53m 09s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/a5ca933b1b04424aa54d08fb8abc58c0 : SUCCESS in 55m 33s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6ccc0544dc184dc1aed82bdce3514e8a : SUCCESS in 1h 06m 48s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7624b99b75974b158e40c52ff9e09185 : SUCCESS in 52m 07s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/25dac99958304ce4a0cd6ab8c79752aa : SUCCESS in 51m 50s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9de51b784b9e4806b99b434082336c1b : SUCCESS in 1h 11m 14s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/de529b407497444ab13cc7e8a666b0fd : FAILURE in 57m 32s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7abd8f1396f84c37b62f018cb4b1a030 : SUCCESS in 59m 13s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/773267cf8d3e40009c876b7e60716055 : SUCCESS in 59m 10s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/58fdaefe969d4df9b03e9dfe792a504e : SUCCESS in 21m 33s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a152e14952184dccb63965966cdb5e1d : FAILURE in 19m 23s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"fed52436c991f97368b8355a194f63d2c1fa6d06","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-03-03 13:34:33.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"33046e45db255e35f061b5918b9413fa207a15c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 14:51:44.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3bd1fed77587406685de78a13a41f157 : SUCCESS in 7m 27s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/29425eaa7090467498abb6cba446e71a : SUCCESS in 7m 36s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0469aa1c1a4342ee926145573cf595c1 : SUCCESS in 6m 46s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/c5fd8bf727bd4ea0931c8488f4058cd7 : SUCCESS in 27m 10s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e27eeb1038b147d88a03ba8945784124 : SUCCESS in 17m 41s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/232e8320679542cf9af074d47049718f : FAILURE in 24m 53s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/902aa07a74c94e8dafaab902c71c72cc : SUCCESS in 56m 36s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/64947d4cd0d24d6bac11942c7db72b0c : SUCCESS in 1h 00m 19s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/3fa05ae6eb3d4b01b2fdae521e8be242 : SUCCESS in 59m 33s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/759c70175c434dca81a6f161ce126e21 : SUCCESS in 57m 38s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/4c0c390d19334701870e36e86c79e9b3 : SUCCESS in 49m 05s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4ae299ff640f4060aa065ff782a9aa9e : SUCCESS in 55m 45s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e97862d0649b4e9c8d208a5c5fc027d2 : SUCCESS in 41m 38s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/ed60886b099e4263af5da29553ca50de : SUCCESS in 40m 14s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/2f64aafa04a84caa9473266ce7ac2645 : SUCCESS in 56m 01s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/15ce1c0e420840cb8ed12951bcbb3849 : SUCCESS in 1h 01m 43s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b5ac72c1c56d4891a078328c554e24ef : SUCCESS in 1h 12m 16s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c1b766f53dbb41559b1d25efd1f2b4b6 : SUCCESS in 59m 04s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/9ab60d33a3a1465ebbee78a4d7c1fb24 : SUCCESS in 27m 02s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b360945f272948e2a637316c2b6e1074 : FAILURE in 24m 28s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"30e91dfa5f916313a1affc7dc7dc017c7e5b535b","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-03-09 16:13:11.000000000","message":"Patch Set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"8224688e931291bac0d39835a54cef23f5678516","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-03-09 16:13:53.000000000","message":"Patch Set 8:\n\nCommit message updated, IGMP support for OVN in neutron is now merged","accounts_in_message":[],"_revision_number":8},{"id":"7e9c529c44f2e6cbd19ff90cdcf2f8b515319d33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 17:32:05.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e3e83b4e41d14ac9a78f44c0cffaf3cc : SUCCESS in 7m 11s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ec6db503b0d6417db0f5c4a623ff9a93 : SUCCESS in 6m 17s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/48b7dc65509742339a941db841b10e94 : SUCCESS in 6m 14s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/3ee622705c57470fb97380ee6addcb22 : SUCCESS in 35m 35s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/0400a2c7c16c49459bc9dc5d588a03dc : FAILURE in 9m 57s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/94c68eae257049b08671e80a611e56ad : FAILURE in 8m 55s (non-voting)\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9bbf40c1ddf343129f71834d4dda94d2 : SUCCESS in 22m 31s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ebda4a3b03c74994a6dcf5f86430d2a8 : FAILURE in 24m 38s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/59c704fcbb734616a694d5b88437f724 : SUCCESS in 1h 03m 25s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/663e03a6296f4276ab40a8ea8bf0de11 : SUCCESS in 1h 08m 29s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/d0d9ddce86514971a105b96095a69286 : SUCCESS in 1h 05m 11s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/d71381a662cb43dc8da55c67712752fc : SUCCESS in 46m 58s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/8ed4b3678a6c4f72aabe7fba48f6a635 : SUCCESS in 55m 15s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a150e3b9782946a0969a533f6da84abd : SUCCESS in 59m 07s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/91c55c96dbdf4c33ab0de3df1ac7e03b : SUCCESS in 44m 09s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/58e05827cdb948d79dc62f49e7567bad : SUCCESS in 43m 21s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c111c4606d4541eb8fcd419caa71733c : SUCCESS in 1h 05m 12s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/37eb241e0c014c30a40783cc9d9be053 : SUCCESS in 1h 09m 30s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d49f945ea1f847a6bcd22c23d2ab7fe7 : SUCCESS in 1h 04m 07s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f3af380ac0e0493898c3c8dade4032d0 : SUCCESS in 44m 46s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/f6bcca38f6b142c68fed33e01936172a : SUCCESS in 22m 25s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/be2777aa6abe4313b968db6c9d16eafe : FAILURE in 21m 56s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"fcdcf78e08f93e5721d6134a4e9b6ac37c8b1a8e","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-03-09 17:34:56.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"063c7e0797ebf3f8ac1d45531922695e601c1daa","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-03-09 23:16:55.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"947abc6c857da2eafb7f1056c491a2ac382a2b3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 23:17:07.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"67718f9800f890d4694ce99653ad5365b3939eb1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 00:27:00.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ff535826e8e949f088a687b1fac84074 : SUCCESS in 5m 09s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ce9b6db397b146019de63aed55e4de03 : SUCCESS in 4m 29s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2359b55adef040ceab5d4374c8b1df46 : SUCCESS in 5m 04s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/01e7a9d96fdd4237a1dc6e8fd820dcc0 : SUCCESS in 33m 29s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/195a0094f3a541289e13978f65b1f4b1 : SUCCESS in 18m 48s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/e2dff44b48d34fbe85f45342edfdd908 : SUCCESS in 1h 04m 57s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/9fce596618f84b67a18e54e56dfb59aa : SUCCESS in 1h 01m 57s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/6fdec53e5af34bb796322e8ee28b3399 : SUCCESS in 1h 08m 17s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/9a572408a93445fcaddf1ef1a2afb291 : SUCCESS in 49m 33s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/beb8eab674fb44ed908a79fb019099fd : SUCCESS in 21m 22s","accounts_in_message":[],"_revision_number":8},{"id":"04de021d71613434b956ea83cb7a074060c22f39","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 00:27:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"382352a893c2a693eb10329365357282d68d75e4","revisions":{"33e6d0de009f60d55e6d6e8c95d0e6c1f3960740":{"kind":"REWORK","_number":1,"created":"2020-02-12 11:44:52.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/67/707367/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/67/707367/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/67/707367/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/67/707367/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/67/707367/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/67/707367/1"}}},"commit":{"parents":[{"commit":"46cd24bf1a3eeaf789d890d274d250746ea18f74","subject":"Merge \"[OVN] Change the default metadata_workers to 2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/46cd24bf1a3eeaf789d890d274d250746ea18f74"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-02-12 11:32:13.000000000","tz":0},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-02-12 11:40:16.000000000","tz":0},"subject":"Add ovs/igmp_snooping_enable support","message":"Add ovs/igmp_snooping_enable support\n\nThis patch is adding support to the \"ovs/igmp_snooping_enable\"\nconfiguration option in the neutron.conf and ml2_conf.ini configuration\nfiles.\n\nBoth, ML2/OVS and ML2/OVN reuse the same configuration option to enable\nIGMP but, for ML2/OVN this option could can be set in the neutron.conf\n(since it will be consumed on the controllers) and and for ML2/OVS it\u0027s\nconfigured for the agents (ml2_conf.ini)\n\nDepends-On: I32f61ba3dd06d7eacf76a74c5c44e1286f90e584\nChange-Id: I1d96899de29cdd994f2a45b701923b0177a7edd8\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/33e6d0de009f60d55e6d6e8c95d0e6c1f3960740"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/33e6d0de009f60d55e6d6e8c95d0e6c1f3960740"}]},"branch":"refs/heads/master"},"520ce0196e5240293fb45f4f477e93d50cef3bdf":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-02-12 13:29:11.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/67/707367/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/67/707367/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/67/707367/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/67/707367/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/67/707367/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/67/707367/2"}}},"commit":{"parents":[{"commit":"46cd24bf1a3eeaf789d890d274d250746ea18f74","subject":"Merge \"[OVN] Change the default metadata_workers to 2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/46cd24bf1a3eeaf789d890d274d250746ea18f74"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-02-12 11:32:13.000000000","tz":0},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-02-12 13:29:07.000000000","tz":0},"subject":"Add ovs/igmp_snooping_enable support","message":"Add ovs/igmp_snooping_enable support\n\nThis patch is adding support to the \"ovs/igmp_snooping_enable\"\nconfiguration option in the neutron.conf and ml2_conf.ini configuration\nfiles.\n\nBoth, ML2/OVS and ML2/OVN reuse the same configuration option to enable\nIGMP but, for ML2/OVN this option could can be set in the neutron.conf\n(since it will be consumed on the controllers) and and for ML2/OVS it\u0027s\nconfigured for the agents (ml2_conf.ini)\n\nChange-Id: I1d96899de29cdd994f2a45b701923b0177a7edd8\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/520ce0196e5240293fb45f4f477e93d50cef3bdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/520ce0196e5240293fb45f4f477e93d50cef3bdf"}]},"branch":"refs/heads/master"},"0fd20f6f09faee7d0e1f1be97316d0349d76bb93":{"kind":"REWORK","_number":3,"created":"2020-02-14 15:16:44.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/67/707367/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/67/707367/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/67/707367/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/67/707367/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/67/707367/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/67/707367/3"}}},"commit":{"parents":[{"commit":"46cd24bf1a3eeaf789d890d274d250746ea18f74","subject":"Merge \"[OVN] Change the default metadata_workers to 2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/46cd24bf1a3eeaf789d890d274d250746ea18f74"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-02-12 11:32:13.000000000","tz":0},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-02-14 15:16:25.000000000","tz":0},"subject":"Add ovs/igmp_snooping_enable support","message":"Add ovs/igmp_snooping_enable support\n\nThis patch is adding support to the \"ovs/igmp_snooping_enable\"\nconfiguration option in the neutron.conf and ml2_conf.ini configuration\nfiles.\n\nBoth, ML2/OVS and ML2/OVN reuse the same configuration option to enable\nIGMP but, for ML2/OVN this option could can be set in the neutron.conf\n(since it will be consumed on the controllers) and and for ML2/OVS it\u0027s\nconfigured for the agents (ml2_conf.ini)\n\nDepends-On: I32f61ba3dd06d7eacf76a74c5c44e1286f90e584\nChange-Id: I1d96899de29cdd994f2a45b701923b0177a7edd8\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/0fd20f6f09faee7d0e1f1be97316d0349d76bb93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/0fd20f6f09faee7d0e1f1be97316d0349d76bb93"}]},"branch":"refs/heads/master"},"c5119192474c159daa829535e8f477cafc6b2a73":{"kind":"REWORK","_number":4,"created":"2020-02-20 10:53:42.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/67/707367/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/67/707367/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/67/707367/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/67/707367/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/67/707367/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/67/707367/4"}}},"commit":{"parents":[{"commit":"b646b7cf1acfb708814cb8adba276b68361588ac","subject":"[ussuri][goal] Cleanup for python 2.7 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b646b7cf1acfb708814cb8adba276b68361588ac"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-02-12 11:32:13.000000000","tz":0},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-02-20 10:53:33.000000000","tz":0},"subject":"Add ovs/igmp_snooping_enable support","message":"Add ovs/igmp_snooping_enable support\n\nThis patch is adding support to the \"ovs/igmp_snooping_enable\"\nconfiguration option in the neutron.conf and ml2_conf.ini configuration\nfiles.\n\nBoth, ML2/OVS and ML2/OVN reuse the same configuration option to enable\nIGMP but, for ML2/OVN this option could can be set in the neutron.conf\n(since it will be consumed on the controllers) and and for ML2/OVS it\u0027s\nconfigured for the agents (ml2_conf.ini)\n\nDepends-On: I32f61ba3dd06d7eacf76a74c5c44e1286f90e584\nChange-Id: I1d96899de29cdd994f2a45b701923b0177a7edd8\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c5119192474c159daa829535e8f477cafc6b2a73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c5119192474c159daa829535e8f477cafc6b2a73"}]},"branch":"refs/heads/master"},"c41cd972ea9731cac7a8e2d95f569f59b81d6c0f":{"kind":"REWORK","_number":5,"created":"2020-03-03 10:08:49.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/67/707367/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/67/707367/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/67/707367/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/67/707367/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/67/707367/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/67/707367/5"}}},"commit":{"parents":[{"commit":"b646b7cf1acfb708814cb8adba276b68361588ac","subject":"[ussuri][goal] Cleanup for python 2.7 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b646b7cf1acfb708814cb8adba276b68361588ac"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-02-12 11:32:13.000000000","tz":0},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-03-03 10:08:39.000000000","tz":0},"subject":"Add ovs/igmp_snooping_enable support","message":"Add ovs/igmp_snooping_enable support\n\nThis patch is adding support to the \"ovs/igmp_snooping_enable\"\nconfiguration option in the neutron.conf and ml2_conf.ini configuration\nfiles.\n\nBoth, ML2/OVS and ML2/OVN reuse the same configuration option to enable\nIGMP but, for ML2/OVN this option could can be set in the neutron.conf\n(since it will be consumed on the controllers) and and for ML2/OVS it\u0027s\nconfigured for the agents (ml2_conf.ini)\n\nDepends-On: I32f61ba3dd06d7eacf76a74c5c44e1286f90e584\nChange-Id: I1d96899de29cdd994f2a45b701923b0177a7edd8\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c41cd972ea9731cac7a8e2d95f569f59b81d6c0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c41cd972ea9731cac7a8e2d95f569f59b81d6c0f"}]},"branch":"refs/heads/master"},"ab77a01547012da4e2345ee6b97abe350fa4bbf4":{"kind":"REWORK","_number":6,"created":"2020-03-03 11:49:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/67/707367/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/67/707367/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/67/707367/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/67/707367/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/67/707367/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/67/707367/6"}}},"commit":{"parents":[{"commit":"b646b7cf1acfb708814cb8adba276b68361588ac","subject":"[ussuri][goal] Cleanup for python 2.7 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b646b7cf1acfb708814cb8adba276b68361588ac"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-02-12 11:32:13.000000000","tz":0},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-03 11:49:37.000000000","tz":0},"subject":"Add ovs/igmp_snooping_enable support","message":"Add ovs/igmp_snooping_enable support\n\nThis patch is adding support to the \"ovs/igmp_snooping_enable\"\nconfiguration option in the neutron.conf and ml2_conf.ini configuration\nfiles.\n\nBoth, ML2/OVS and ML2/OVN reuse the same configuration option to enable\nIGMP but, for ML2/OVN this option could can be set in the neutron.conf\n(since it will be consumed on the controllers) and and for ML2/OVS it\u0027s\nconfigured for the agents (ml2_conf.ini)\n\nDepends-On: I32f61ba3dd06d7eacf76a74c5c44e1286f90e584\nChange-Id: I1d96899de29cdd994f2a45b701923b0177a7edd8\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/ab77a01547012da4e2345ee6b97abe350fa4bbf4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/ab77a01547012da4e2345ee6b97abe350fa4bbf4"}]},"branch":"refs/heads/master"},"b1ee0179bfff5f0fe1941efe7a866136cc891c33":{"kind":"REWORK","_number":7,"created":"2020-03-03 13:34:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/67/707367/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/67/707367/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/67/707367/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/67/707367/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/67/707367/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/67/707367/7"}}},"commit":{"parents":[{"commit":"b646b7cf1acfb708814cb8adba276b68361588ac","subject":"[ussuri][goal] Cleanup for python 2.7 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b646b7cf1acfb708814cb8adba276b68361588ac"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-02-12 11:32:13.000000000","tz":0},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-03 13:34:28.000000000","tz":0},"subject":"Add ovs/igmp_snooping_enable support","message":"Add ovs/igmp_snooping_enable support\n\nThis patch is adding support to the \"ovs/igmp_snooping_enable\"\nconfiguration option in the neutron.conf and ml2_conf.ini configuration\nfiles.\n\nBoth, ML2/OVS and ML2/OVN reuse the same configuration option to enable\nIGMP but, for ML2/OVN this option could can be set in the neutron.conf\n(since it will be consumed on the controllers) and and for ML2/OVS it\u0027s\nconfigured for the agents (ml2_conf.ini)\n\nDepends-On: I32f61ba3dd06d7eacf76a74c5c44e1286f90e584\nChange-Id: I1d96899de29cdd994f2a45b701923b0177a7edd8\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b1ee0179bfff5f0fe1941efe7a866136cc891c33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b1ee0179bfff5f0fe1941efe7a866136cc891c33"}]},"branch":"refs/heads/master"},"382352a893c2a693eb10329365357282d68d75e4":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2020-03-09 16:13:11.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/67/707367/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/67/707367/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/67/707367/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/67/707367/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/67/707367/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/67/707367/8"}}},"commit":{"parents":[{"commit":"b646b7cf1acfb708814cb8adba276b68361588ac","subject":"[ussuri][goal] Cleanup for python 2.7 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b646b7cf1acfb708814cb8adba276b68361588ac"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-02-12 11:32:13.000000000","tz":0},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-03-09 16:13:08.000000000","tz":0},"subject":"Add ovs/igmp_snooping_enable support","message":"Add ovs/igmp_snooping_enable support\n\nThis patch is adding support to the \"ovs/igmp_snooping_enable\"\nconfiguration option in the neutron.conf and ml2_conf.ini configuration\nfiles.\n\nBoth, ML2/OVS and ML2/OVN reuse the same configuration option to enable\nIGMP but, for ML2/OVN this option could can be set in the neutron.conf\n(since it will be consumed on the controllers) and and for ML2/OVS it\u0027s\nconfigured for the agents (ml2_conf.ini)\n\nChange-Id: I1d96899de29cdd994f2a45b701923b0177a7edd8\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/382352a893c2a693eb10329365357282d68d75e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/382352a893c2a693eb10329365357282d68d75e4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
