)]}'
{"id":"openstack%2Fpuppet-neutron~707983","triplet_id":"openstack%2Fpuppet-neutron~master~Id66ad15ca22fabe965f06bd15bf4e2aa28102f16","project":"openstack/puppet-neutron","branch":"master","topic":"drop-py27-support","hashtags":[],"change_id":"Id66ad15ca22fabe965f06bd15bf4e2aa28102f16","subject":"[ussuri][goal] Cleanup for python 2.7 drop","status":"MERGED","created":"2020-02-16 01:33:02.000000000","updated":"2020-02-18 06:08:03.000000000","submitted":"2020-02-18 06:05:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707983-1582005959508-b4850ae3","meta_rev_id":"933ee082618668fab0031a083ac4eb10d09c7176","_number":707983,"virtual_id_number":707983,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-18 06:05:59.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2020-02-18 02:42:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2020-02-17 20:26:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2020-02-18 02:42:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-17 20:26:14.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-02-18 02:42:59.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2020-02-18 06:05:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1aa138af37adbee5f5d2d790cd34faeae00a7649","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-16 01:33:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"93c9753ce84e770dc55536e35116b82903c9eecd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-16 04:43:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8ba1253814a14f8a8f3f520a1cb724f9 : SUCCESS in 5m 14s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b01acebad7ef4bb780cfd9c34f9bdd18 : SUCCESS in 5m 46s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/898463d091024033943dd2797db14384 : SUCCESS in 5m 54s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/0dabc6f011604d6b8a0f57394458dc63 : SUCCESS in 24m 19s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/91489656ab38404ebd4125a35ad872bd : SUCCESS in 21m 15s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/daf5f74f8df0418f8ae1d7f3f8a1018f : FAILURE in 17m 39s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/d26ac98cbe5d466aa61787966cdc30c8 : SUCCESS in 1h 06m 57s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/a3781861bf744745b701db70c47bfbd5 : SUCCESS in 1h 20m 08s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/28e235aa5270427da419409121d17243 : SUCCESS in 1h 13m 06s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/1d563309d225493199acd3394a2c6ee4 : SUCCESS in 47m 37s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/08018be3de294a27987fa48932fef7a6 : SUCCESS in 56m 14s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e97f9b4a6a3c441f8a6ec8a063b6a0b3 : SUCCESS in 1h 08m 24s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/10bb1439a4f042e482fcd1c5641d2418 : SUCCESS in 39m 06s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/9fe5d14b599c4e3b85ced40b6fd641a7 : SUCCESS in 51m 59s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ca6f7b48f4e34cb9a8be8de4269b2afc : SUCCESS in 1h 01m 46s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/dc3d94ce7e3c4bb9ae78831d3c88186c : SUCCESS in 57m 19s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/846f590b99574de28893e5b65fe4000e : SUCCESS in 59m 02s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/dc0b10e0eab34260b5f1a14db9574475 : SUCCESS in 57m 47s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/f30ef5e4a6d344489f422931a5b948ec : SUCCESS in 23m 59s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/31b928d277044f97bcba364d3d7be1a8 : FAILURE in 18m 11s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd4e2c2044e145bcaba6f5d203522c16 : SUCCESS in 4m 58s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/dc4cb7fe8ea9414383d0e131b0496d8c : SUCCESS in 3h 08m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"58b517037522a0ace798352efbc6aba7818c5f5e","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-02-17 20:26:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e553d4a35605584d1b181c667945cac3d74eb35b","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-02-18 02:42:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e7b87ef1fa415b2c48c68be1067fd78ab912e778","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 02:43:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3ceb2ea23acdc262901fa6fd0c31f0d2ebafcf6f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 06:05:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"44e7226b082a685a3364c905ba564e31b09f9657","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 06:05:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/58641bf425394d82aef7cc28f0fed03d : SUCCESS in 7m 33s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b2e6d07ab6704fe187ca52cfbd4ce31a : SUCCESS in 6m 04s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cafefab4e71641848046b226f0357c8c : SUCCESS in 6m 59s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/77d8c5684785446fba6a9550d9d5fdd7 : SUCCESS in 25m 52s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5c21bb5b849d4a4db932e5ead8640f4e : SUCCESS in 18m 10s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/255b81e947f548c492d5081be2dcad87 : SUCCESS in 1h 01m 06s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/53f922829ac647c783ac60e38b57489d : SUCCESS in 1h 07m 07s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/58799ba1380645f9a9dda9cd33260add : SUCCESS in 1h 08m 44s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/315aa9ad38314e6690b7f34d403966a0 : SUCCESS in 59m 39s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/4ffe3f7123ed4e03bf24f11d4d408bfa : SUCCESS in 24m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6fc8eb1947274487b9c4c2a65bd179a9 : SUCCESS in 7m 50s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/14908757bf1a4e83a62684c24d318515 : TIMED_OUT in 3h 17m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c43749e7bcc1078d7036585af2e4560e11f9d7c1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 06:08:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/300049f112064c7e9888ba17adfca043 : SUCCESS in 1m 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b646b7cf1acfb708814cb8adba276b68361588ac","revisions":{"b646b7cf1acfb708814cb8adba276b68361588ac":{"kind":"REWORK","_number":1,"created":"2020-02-16 01:33:02.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/83/707983/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/83/707983/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/83/707983/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/83/707983/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/83/707983/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/83/707983/1"}}},"commit":{"parents":[{"commit":"3c7692488345622fc144294258874dbc3fd6e69b","subject":"Prepare Ussuri M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/3c7692488345622fc144294258874dbc3fd6e69b"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-02-16 01:32:20.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-02-16 01:33:01.000000000","tz":-360},"subject":"[ussuri][goal] Cleanup for python 2.7 drop","message":"[ussuri][goal] Cleanup for python 2.7 drop\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\npuppet-* repos have to cleanup the requirement and tox\nfor py2.7 drop.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nChange-Id: Id66ad15ca22fabe965f06bd15bf4e2aa28102f16\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b646b7cf1acfb708814cb8adba276b68361588ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b646b7cf1acfb708814cb8adba276b68361588ac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
