)]}'
{"id":"openstack%2Fpuppet-neutron~711944","triplet_id":"openstack%2Fpuppet-neutron~stable%2Fstein~I4290bb1dbb96ea0d57e6ad8c6a686c2bf103c907","project":"openstack/puppet-neutron","branch":"stable/stein","topic":"bug/1866206-stable/stein","hashtags":[],"change_id":"I4290bb1dbb96ea0d57e6ad8c6a686c2bf103c907","subject":"Force cast bridge mappings as an array","status":"MERGED","created":"2020-03-09 16:09:46.000000000","updated":"2020-03-10 06:59:38.000000000","submitted":"2020-03-10 06:59:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711944-1583823578902-72e87165","meta_rev_id":"0282c8b86a7f7e12c3f784090cc71947184e694d","_number":711944,"virtual_id_number":711944,"owner":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-10 06:59:38.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-03-09 19:08:24.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-03-10 03:38:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-09 19:08:24.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-03-10 03:38:37.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2020-03-10 06:59:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a009d995e590a8da7da9265775ba4b4430a62339","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-03-09 16:09:46.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"93e2a10a7c49b6525883db0540341363e9ad15aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 19:06:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d03627dd97fb49a4aaa2dd8f3645557d : SUCCESS in 8m 36s\n- puppet-openstack-syntax-4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f2f44ff7a1264817be247bb8c1347089 : SUCCESS in 6m 00s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7c5354dcda0f4df0af811dcbe998eaf9 : SUCCESS in 6m 05s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/10adbd9384ad4915869f6e80361f2745 : SUCCESS in 6m 11s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/b6c6826cf9d346ee9d903d5dadcc121b : SUCCESS in 32m 39s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/730ae46292d44a9fadd455ac4beef792 : SUCCESS in 33m 07s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/537bfab77c104e23b1d09f094d475086 : SUCCESS in 56m 31s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/e8b7d2daa50747a1a39a1c1ec5e6c56b : SUCCESS in 1h 05m 43s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/6f81532235e8466c9554f55f72b0b4d8 : SUCCESS in 1h 22m 15s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/056b4a1ed663416e9d9063886a0e4384 : SUCCESS in 53m 55s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/4e9562717045459e80cfb0cdfc6c2bd5 : FAILURE in 27m 59s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1e630190c09f47ddbc96259e82af8dbd : FAILURE in 25m 32s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/71b92b65a1c64c6bb70fd31434ddf054 : FAILURE in 23m 52s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/07e552eec54e4090a6be220f85f455fa : FAILURE in 42m 11s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/1bceaa45396f4851927a5121877f36b2 : SUCCESS in 23m 14s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b44a8a32d53d40bbb705c1c7edde6e77 : FAILURE in 20m 39s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/fe937b3738744014a2972f8125ea3f5b : SUCCESS in 2h 52m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8c2044e3919e8f63575a4ea5fea09db487ab9efe","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-03-09 19:08:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"14be1814580cdf35deb53983d921e436c28144e3","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-03-10 03:38:37.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ee7df514a00554ea46ee9e55be82ff976d0184b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 03:38:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d326cd683fc17fe9f517d39e482ef3fb640ee3fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 06:59:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"90bd9403cebd4db787441a79e2b738925f720467","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 06:59:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f04dbdb7a6404bc2ab67ca0ba210669a : SUCCESS in 4m 49s\n- puppet-openstack-syntax-4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/52d5d4d43746442b9b266b0f697bc6be : SUCCESS in 5m 10s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7ac3c16d359548ae97310c346ea1d17d : SUCCESS in 4m 04s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6b6686945ecd47c3a4b3c2c938baf3c2 : SUCCESS in 4m 09s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/8d3ad38aede84c8085c611e8a7d7fded : SUCCESS in 34m 58s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/43923ecb87a4453c9bf1a1bb886c5a79 : SUCCESS in 32m 07s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/32d925c137c647efa20684285dc99d49 : SUCCESS in 55m 00s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/64cdd69264934776a78aac71cc6f002b : SUCCESS in 1h 03m 20s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/876045d495734a798f9450b6bbfd76ac : SUCCESS in 1h 01m 12s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/e103e558762149079185de37ce916e73 : SUCCESS in 18m 28s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/543f3840f325436c9f588bced9b8c7d8 : FAILURE in 15m 23s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6a7525b46e9b4c10bdabf24fba84f42b : SUCCESS in 3h 13m 03s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3991b18c8419e67ab13eed83da5a9d96672cf452","revisions":{"3991b18c8419e67ab13eed83da5a9d96672cf452":{"kind":"REWORK","_number":1,"created":"2020-03-09 16:09:46.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/44/711944/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/44/711944/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/44/711944/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/44/711944/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/44/711944/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/44/711944/1"}}},"commit":{"parents":[{"commit":"d7bdf94e46e4dac9fc2d558a015a1d94508606c2","subject":"Add of_connect_timeout and of_request_timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/d7bdf94e46e4dac9fc2d558a015a1d94508606c2"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-03-05 16:09:39.000000000","tz":60},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-03-09 16:09:46.000000000","tz":0},"subject":"Force cast bridge mappings as an array","message":"Force cast bridge mappings as an array\n\nIn some cases, the parameter will not passed as an array but as a\nstring, causing conversion error when generating the matching list\n\nForce array parameter conversion to cover all cases\n\nChange-Id: I4290bb1dbb96ea0d57e6ad8c6a686c2bf103c907\nCloses-Bug: #1866206\n(cherry picked from commit 73d64f2cc1742b855dc72dbc90f1d81cb6543413)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/3991b18c8419e67ab13eed83da5a9d96672cf452"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/3991b18c8419e67ab13eed83da5a9d96672cf452"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
