)]}'
{"id":"openstack%2Fpuppet-neutron~711946","triplet_id":"openstack%2Fpuppet-neutron~stable%2Fqueens~I4290bb1dbb96ea0d57e6ad8c6a686c2bf103c907","project":"openstack/puppet-neutron","branch":"stable/queens","topic":"bug/1866206-stable/queens","hashtags":[],"change_id":"I4290bb1dbb96ea0d57e6ad8c6a686c2bf103c907","subject":"Force cast bridge mappings as an array","status":"MERGED","created":"2020-03-09 16:11:51.000000000","updated":"2020-03-10 17:36:56.000000000","submitted":"2020-03-10 17:36:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711946-1583861816026-3bf77f9f","meta_rev_id":"74f00b8a1241b7c194d006f7defd82daceb3dfc2","_number":711946,"virtual_id_number":711946,"owner":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-10 17:36:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-03-09 18:00:51.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-03-10 15:49:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-03-10 15:49:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-09 18:00:51.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-03-10 15:49:02.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-03-10 17:36:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e1f40050dadc3e0d96e42a93dd0c60e8d55ff2e6","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-03-09 16:11:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"78294b606fd523903967e9135b9a97fcf8264a0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 17:54:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/dbf3c8cac1524eafa674747e656f3c6f : SUCCESS in 6m 02s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/c5d2794688c8436ebf7849946e7edb22 : SUCCESS in 6m 07s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/42c59251f7bd4c68b6b8b1f4ecb21aad : SUCCESS in 31m 16s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/c5e016ab8a114006b3086412c636d479 : SUCCESS in 27m 31s\n- puppet-openstack-integration-4-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/11d49879f5d749a2ad2b9a974a71104a : SUCCESS in 1h 03m 49s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/39d9bb6ba44b42dea5c7361e1ef32e8a : SUCCESS in 1h 00m 12s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/3ea5d6e274b642b88064e2c58fadf606 : SUCCESS in 1h 03m 04s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/3cbb0ddf14414e8da90f17b16d896335 : SUCCESS in 45m 37s\n- puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial-luminous https://zuul.opendev.org/t/openstack/build/eaf91bd88ff64abb9c5d4aeee28cf5d1 : TIMED_OUT in 1h 31m 16s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/60d06eef85eb49af9fa02c0cda409f0a : TIMED_OUT in 1h 32m 33s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/e35dfaa8f1f3486cabe932832d56f005 : TIMED_OUT in 1h 31m 48s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial-luminous https://zuul.opendev.org/t/openstack/build/d9196c2166cc423eb76bd61b48968225 : TIMED_OUT in 1h 31m 46s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/34bd8797c55d4faca9c23829e4546640 : SUCCESS in 22m 39s\n- puppet-openstack-beaker-xenial https://zuul.opendev.org/t/openstack/build/19edeb59f3d043f1b726c72df5c433d9 : SUCCESS in 17m 50s","accounts_in_message":[],"_revision_number":1},{"id":"e0e0f66d44815bf8c35d19470cd172494b00da13","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-03-09 18:00:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b0502cf9529228fe5a788ebeccd9bbf50e3eff82","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-03-10 15:49:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"88834debfe9521ce7947e79116f074a9715ee81f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 15:49:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"15d1598bc3512b2a050833c7c6dd5ad2ca61f182","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 17:36:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/5cfd0f04a9f64efb9e70227f7e4a6a74 : SUCCESS in 6m 00s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/5a78bee87de14a769002fb34434ecc1b : SUCCESS in 6m 36s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/69222121b4bf4f8ab80d3a84088de436 : SUCCESS in 29m 54s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/2cd2886fe0f640c3afc84e698ed78838 : SUCCESS in 26m 05s\n- puppet-openstack-integration-4-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/e814b44e81854b33b3bd017ea5a3aeb3 : SUCCESS in 58m 17s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/1a67328f8e1a49a48de9ecd7f9a1676b : SUCCESS in 1h 15m 24s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/abd4730f2e964fa3b7af66c209d1cdca : SUCCESS in 1h 15m 28s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/b0a2ce7b23364dc7867453617fb31706 : SUCCESS in 50m 13s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/602af8511fbb4cab963b3af5617610f0 : SUCCESS in 21m 22s\n- puppet-openstack-beaker-xenial https://zuul.opendev.org/t/openstack/build/9fd5b34fe6374c6ea73ba9dae72a3519 : SUCCESS in 16m 47s","accounts_in_message":[],"_revision_number":1},{"id":"40a6a92d5ddb2a13d366e606d7b6d7f34dc9e641","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 17:36:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0f0e824d8da8dd77e9821ef38934fff017005bb3","revisions":{"0f0e824d8da8dd77e9821ef38934fff017005bb3":{"kind":"REWORK","_number":1,"created":"2020-03-09 16:11:51.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/46/711946/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/46/711946/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/46/711946/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/46/711946/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/46/711946/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/46/711946/1"}}},"commit":{"parents":[{"commit":"b0196fd0e4029bee0fc1431638f4db442b3d9b83","subject":"CiscoNexus: Update and Remove deprecated vars","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b0196fd0e4029bee0fc1431638f4db442b3d9b83"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-03-05 16:09:39.000000000","tz":60},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-03-09 16:11:30.000000000","tz":60},"subject":"Force cast bridge mappings as an array","message":"Force cast bridge mappings as an array\n\nIn some cases, the parameter will not passed as an array but as a\nstring, causing conversion error when generating the matching list\n\nForce array parameter conversion to cover all cases\n\nConflicts:\n\tmanifests/agents/ml2/linuxbridge.pp\n\nChange-Id: I4290bb1dbb96ea0d57e6ad8c6a686c2bf103c907\nCloses-Bug: #1866206\n(cherry picked from commit 73d64f2cc1742b855dc72dbc90f1d81cb6543413)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/0f0e824d8da8dd77e9821ef38934fff017005bb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/0f0e824d8da8dd77e9821ef38934fff017005bb3"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
