)]}'
{"id":"openstack%2Fpuppet-neutron~716952","triplet_id":"openstack%2Fpuppet-neutron~master~If3b2c8a8f8c4bbb4a1391207c2ed2ae4660a9bab","project":"openstack/puppet-neutron","branch":"master","topic":"bug/1870257","hashtags":[],"change_id":"If3b2c8a8f8c4bbb4a1391207c2ed2ae4660a9bab","subject":"Limit scope of puppet-neutron-tripleo-standalone tempest tests","status":"ABANDONED","created":"2020-04-02 11:17:36.000000000","updated":"2020-04-02 14:52:17.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ac4b4ac22c8215b3cc4da3a6b93e6ca9aafb22d0","_number":716952,"virtual_id_number":716952,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-04-02 14:18:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2020-04-02 11:44:58.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"date":"2020-04-02 14:26:44.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2020-04-02 14:27:13.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"date":"2020-04-02 14:26:53.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"date":"2020-04-02 14:26:32.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-02 11:28:53.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2020-04-02 11:29:43.000000000","updated_by":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"reviewer":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"state":"REVIEWER"},{"updated":"2020-04-02 11:44:58.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2020-04-02 14:18:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-02 14:26:32.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-04-02 14:26:44.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-04-02 14:26:53.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-04-02 14:27:13.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"}],"messages":[{"id":"5042003a15567d60a4b3f6cdebd9bb2d9bbb5249","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-02 11:17:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"adef9efe97b588209c5dfc8852de2960175e7c65","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-02 11:19:48.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9ff7f75b3baf2d3f666cecb0f0ffacde8b03fc79","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-02 11:33:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"968c8e22561bd1d21516501a334142d475ca4598","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-02 11:36:06.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"f6d4a23942053ef3d2c23806a8972886bac880c6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-02 11:38:54.000000000","message":"Patch Set 4:\n\nNote that unit tests should fail because of another fix we need but currently is blocked by this.\n\n https://review.opendev.org/#/c/716197/\n\nAfter we get +2+2+A on this patch, I\u0027ll reach to infra team to ask them to force-merge this.\nOf cause we can merge these 2 patches into 1 patch to avoid gate failure, I don\u0027t like to have these 2 unrelated things in the same commit.","accounts_in_message":[],"_revision_number":4},{"id":"a5e6d4974f3b2c2a1a15893d5777a550e0f563ea","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-04-02 11:40:42.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ac8ed1fa9ba9e25d8f7e2194634872b67a5c5df0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-02 11:42:56.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"91ad277dc73690a80ad4be167c29c90079c47266","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-02 11:43:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bac84b0f6717bdc298f86e3c9584d0978ad5dbeb","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-04-02 11:44:58.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"be915108cec51da7725e2c2c934fdec42f488353","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 14:18:37.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9c298e136d3d47cf8a1d8d8cbfb18170 : SUCCESS in 6m 53s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2ed133d5083348309e702168e02ad378 : SUCCESS in 6m 19s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4386de1c2d124b6e9111e85962342234 : SUCCESS in 5m 05s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/572e15a2eb9e45d592510a8217d584ef : FAILURE in 20m 12s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/16359ef0414a483abe7395ecc01d7bba : FAILURE in 20m 33s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b8c11b54307c42af8d3787d4798ef6c6 : FAILURE in 22m 29s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/132d97f59f614a67bb30ec0bed6154df : FAILURE in 20m 18s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/4c74eb6152e6439baac7ff9022baa2a0 : SUCCESS in 1h 23m 34s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/efc6a6d4cfa44edfa812ac067eb241b1 : SUCCESS in 56m 24s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9a29daf4fb6c45eeafd3652d4599b2c7 : SUCCESS in 45m 05s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/170a7681ac474242a485b5895b0a78fb : SUCCESS in 40m 21s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8d7d949af0c84ec5b338a70e1ac6ed0f : SUCCESS in 52m 32s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8e4bf762c3d4496985db3f254933c27a : SUCCESS in 1h 01m 33s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c9f77db7182b47959c3f41d9d5fd6fac : SUCCESS in 1h 04m 40s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4695a63304374524ae91da58896a9fb5 : SUCCESS in 43m 24s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/1079810e74664593a6af7caa59cc7a20 : SUCCESS in 21m 27s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f55146b11fdc41a883f09e7a3f4af074 : FAILURE in 9m 34s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f898b4800ddf480389cee21e343c9898 : SUCCESS in 2h 33m 47s","accounts_in_message":[],"_revision_number":5},{"id":"c8bbc05abada42e93973005c0655955c52b8774a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-02 14:52:17.000000000","message":"Abandoned\n\nMerged into another gate fixing patch, after discussion with Alex.\n\nhttps://review.opendev.org/#/c/716197/","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"6cb1d1af735a0fbd2d811ec40fecbd7d1c987898","revisions":{"ace2ce649d6d73ce4e36993c39e06e3afde5c13f":{"kind":"REWORK","_number":1,"created":"2020-04-02 11:17:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/716952/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/52/716952/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/52/716952/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/52/716952/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/52/716952/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/52/716952/1"}}},"commit":{"parents":[{"commit":"616913ff9dd785cf500bd285ea75fdbc539b11e9","subject":"Merge \"Add CentOS 8 and Puppet 6 in metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/616913ff9dd785cf500bd285ea75fdbc539b11e9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-02 11:15:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-02 11:17:01.000000000","tz":540},"subject":"Limit scope of puppet-neutron-tripleo-standalone tempest tests","message":"Limit scope of puppet-neutron-tripleo-standalone tempest tests\n\nThe current scope used in tempest job it so much that it can cause\ntimeout in gate job often.\nThis patch limits the test scope for netowork api and basic network\nops scenario test to avoid timeout.\n\nRelated-Bug: #1870257\nChange-Id: If3b2c8a8f8c4bbb4a1391207c2ed2ae4660a9bab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/ace2ce649d6d73ce4e36993c39e06e3afde5c13f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/ace2ce649d6d73ce4e36993c39e06e3afde5c13f"}]},"branch":"refs/heads/master"},"a46fa59f9dfb4f1b20e48c04b572bd5e5fe33d4c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-04-02 11:19:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/716952/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/52/716952/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/52/716952/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/52/716952/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/52/716952/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/52/716952/2"}}},"commit":{"parents":[{"commit":"616913ff9dd785cf500bd285ea75fdbc539b11e9","subject":"Merge \"Add CentOS 8 and Puppet 6 in metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/616913ff9dd785cf500bd285ea75fdbc539b11e9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-02 11:15:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-02 11:19:32.000000000","tz":540},"subject":"Limit scope of puppet-neutron-tripleo-standalone tempest tests","message":"Limit scope of puppet-neutron-tripleo-standalone tempest tests\n\nThe current scope used in tempest job it so wide that it can cause\ntimeout in gate job continuously.\nThis patch limits the test scope for netowork api and basic network\nops scenario test to avoid timeout.\n\nRelated-Bug: #1870257\nChange-Id: If3b2c8a8f8c4bbb4a1391207c2ed2ae4660a9bab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/a46fa59f9dfb4f1b20e48c04b572bd5e5fe33d4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/a46fa59f9dfb4f1b20e48c04b572bd5e5fe33d4c"}]},"branch":"refs/heads/master"},"9753b76f3855321d30512439a87124fd052b91bc":{"kind":"REWORK","_number":3,"created":"2020-04-02 11:33:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/716952/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/52/716952/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/52/716952/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/52/716952/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/52/716952/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/52/716952/3"}}},"commit":{"parents":[{"commit":"616913ff9dd785cf500bd285ea75fdbc539b11e9","subject":"Merge \"Add CentOS 8 and Puppet 6 in metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/616913ff9dd785cf500bd285ea75fdbc539b11e9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-02 11:15:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-02 11:33:10.000000000","tz":540},"subject":"Limit scope of puppet-neutron-tripleo-standalone tempest tests","message":"Limit scope of puppet-neutron-tripleo-standalone tempest tests\n\nThe current scope used in tempest job it so wide that it can cause\ntimeout in gate job continuously.\nThis patch limits the test scope for netowork api and basic network\nops scenario test to avoid timeout.\n\nRelated-Bug: #1870257\nChange-Id: If3b2c8a8f8c4bbb4a1391207c2ed2ae4660a9bab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/9753b76f3855321d30512439a87124fd052b91bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/9753b76f3855321d30512439a87124fd052b91bc"}]},"branch":"refs/heads/master"},"e4f1213d434ea1a48380554ee5041f208523944a":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-04-02 11:36:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/716952/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/52/716952/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/52/716952/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/52/716952/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/52/716952/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/52/716952/4"}}},"commit":{"parents":[{"commit":"616913ff9dd785cf500bd285ea75fdbc539b11e9","subject":"Merge \"Add CentOS 8 and Puppet 6 in metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/616913ff9dd785cf500bd285ea75fdbc539b11e9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-02 11:15:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-02 11:35:52.000000000","tz":540},"subject":"Limit scope of puppet-neutron-tripleo-standalone tempest tests","message":"Limit scope of puppet-neutron-tripleo-standalone tempest tests\n\nThe current scope used in tempest job it so wide that it can cause\ntimeout in gate job continuously.\nThis patch limits the test scope for netowork api and basic network\nops scenario test to avoid timeout.\n\nCloses-Bug: #1870257\nChange-Id: If3b2c8a8f8c4bbb4a1391207c2ed2ae4660a9bab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/e4f1213d434ea1a48380554ee5041f208523944a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/e4f1213d434ea1a48380554ee5041f208523944a"}]},"branch":"refs/heads/master"},"6cb1d1af735a0fbd2d811ec40fecbd7d1c987898":{"kind":"REWORK","_number":5,"created":"2020-04-02 11:42:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/716952/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/52/716952/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/52/716952/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/52/716952/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/52/716952/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/52/716952/5"}}},"commit":{"parents":[{"commit":"616913ff9dd785cf500bd285ea75fdbc539b11e9","subject":"Merge \"Add CentOS 8 and Puppet 6 in metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/616913ff9dd785cf500bd285ea75fdbc539b11e9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-02 11:15:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-02 11:42:46.000000000","tz":540},"subject":"Limit scope of puppet-neutron-tripleo-standalone tempest tests","message":"Limit scope of puppet-neutron-tripleo-standalone tempest tests\n\nThe current scope used in tempest job it so wide that it can cause\ntimeout in gate job continuously.\nThis patch limits the test scope for netowork api and basic network\nops scenario test to avoid timeout.\n\nCloses-Bug: #1870257\nChange-Id: If3b2c8a8f8c4bbb4a1391207c2ed2ae4660a9bab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/6cb1d1af735a0fbd2d811ec40fecbd7d1c987898"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/6cb1d1af735a0fbd2d811ec40fecbd7d1c987898"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
