)]}'
{"id":"openstack%2Fpuppet-neutron~717152","triplet_id":"openstack%2Fpuppet-neutron~stable%2Ftrain~Ib47deec4a365a96e3e2b7972f65df089003bb68f","project":"openstack/puppet-neutron","branch":"stable/train","topic":"centos8-openswan-train","hashtags":[],"change_id":"Ib47deec4a365a96e3e2b7972f65df089003bb68f","subject":"Fix incorrect openswan package in CentOS8","status":"MERGED","created":"2020-04-03 00:01:20.000000000","updated":"2020-04-03 18:16:44.000000000","submitted":"2020-04-03 18:16:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"717152-1585937804397-6894be15","meta_rev_id":"4db32f4dd0be5dedb4e65e4fc52c622728d06716","_number":717152,"virtual_id_number":717152,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-03 18:16:44.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-04-03 14:14:49.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-04-03 08:51:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-04-03 14:14:49.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-03 08:51:58.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-04-03 14:14:49.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-04-03 18:16:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b15e2882c47fe0552b4c6cc3af1f8ef94a4b8358","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-03 00:01:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6c26cf8ef7c1eaa747773ef032e893a064cf2cf8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 02:56:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5dd05295acd94ddc99ddec5fb1096137 : SUCCESS in 4m 53s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/28bad63d5e9f41978fdd0312b85216ad : SUCCESS in 4m 21s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/24c6ad4c7312406c8669bb11188d1564 : SUCCESS in 5m 10s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/e001f277ac454b7ebb0d9d50dd233393 : SUCCESS in 31m 32s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/86a6d1d3c6fa441b8af9661a0c18e409 : SUCCESS in 22m 33s (non-voting)\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/42cd9ea123354987b5bc7c441832e12a : SUCCESS in 19m 52s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/9c6d4b90a15c480cb0eed96dafca0c47 : SUCCESS in 53m 18s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/1a687c675c204ae8abf37d75ac41cc1e : SUCCESS in 1h 07m 16s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/02977afbd0664ae2845eef9f798e092b : SUCCESS in 1h 01m 55s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/f348113778824a428f8f84584f08bf96 : SUCCESS in 46m 00s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/2b50ba34455c44248d91aa8b5e11fde9 : SUCCESS in 59m 25s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2e1e68c78b4341409eec991e6be5a334 : SUCCESS in 47m 38s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2549662e9cb44a70a76044b7bbb54689 : SUCCESS in 55m 16s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/99f4db6fbb464062a0d7ceb9537e0419 : SUCCESS in 51m 01s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5895179e7c9c482fa10316cf18408e5e : SUCCESS in 47m 13s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/690e0ccfa06b407abde62968f9bb7fed : SUCCESS in 1h 29m 15s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/913f347b89024894ae250f1d21ff7cd8 : SUCCESS in 1h 06m 21s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/155c9487c3c64fba8d47ee96cf8bb242 : SUCCESS in 57m 05s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/8d8817eebf7a4247b8093dcbe060020e : SUCCESS in 40m 06s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8f3e9de0621f4d9a8b9107aa1c08b087 : FAILURE in 8m 43s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/170c0d5f641548389a50f379d0373eca : SUCCESS in 2h 53m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"edac02fc9094bddde95bf972ccd0c19a259b4c74","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-04-03 08:51:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4ff66d2d00b48a7ae697bf5d8d7c287b5d1d8b51","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-04-03 14:14:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8fc357289efbd9e2791fd35ec81ae0488fc73220","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 14:15:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c9ab5e5eb03838a78044fdd9fe7673b76bd856ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 18:16:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"68e17480b2008077663f8fbe9858da17aeaba918","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 18:16:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c90760c4e0b6403292af253887735b33 : SUCCESS in 6m 31s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2dd130948c4d461e894bf34c0e72e9db : SUCCESS in 5m 50s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9688ce1ffb7a44d381c96374efbd2872 : SUCCESS in 5m 49s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/e0f2649ce2cd440b9538a26fe6bd8b27 : SUCCESS in 36m 34s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/c35db6fc449c4a34a037d335a7dd4e55 : SUCCESS in 58m 12s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/04548117859c4b0894f8367f4fcc3e14 : SUCCESS in 1h 02m 07s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/df27c798268d4ac7aab66026e5ec0e3b : SUCCESS in 1h 12m 28s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/3eec8df09ce94060a5a5a71437cae272 : SUCCESS in 51m 09s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/e692fb3d474d44d19cc4883245226789 : SUCCESS in 19m 58s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/57fcae3552c64e739e501e01b32f53d7 : SUCCESS in 2h 37m 51s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"445eb54d984ebcd5c5f54b32910c0cb7d31f2d50","revisions":{"445eb54d984ebcd5c5f54b32910c0cb7d31f2d50":{"kind":"REWORK","_number":1,"created":"2020-04-03 00:01:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/717152/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/52/717152/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/52/717152/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/52/717152/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/52/717152/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/52/717152/1"}}},"commit":{"parents":[{"commit":"175f063b182c627100404621961371b73b7e6f03","subject":"Limit scope of puppet-neutron-tripleo-standalone tempest tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/175f063b182c627100404621961371b73b7e6f03"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-02 23:49:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-03 00:01:06.000000000","tz":540},"subject":"Fix incorrect openswan package in CentOS8","message":"Fix incorrect openswan package in CentOS8\n\nThis is a partial backport of f7bc1dc4f207f45cc561e933a65f048813f1b8d1,\nand makes sure that we install libreswan instead of openswan in CentOS8\nand RHEL8.\n\nChange-Id: Ib47deec4a365a96e3e2b7972f65df089003bb68f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/445eb54d984ebcd5c5f54b32910c0cb7d31f2d50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/445eb54d984ebcd5c5f54b32910c0cb7d31f2d50"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
