)]}'
{"id":"openstack%2Fpuppet-neutron~717207","triplet_id":"openstack%2Fpuppet-neutron~master~I5847ced19cbe40d197ab807f008c5b4a21b8277d","project":"openstack/puppet-neutron","branch":"master","hashtags":[],"change_id":"I5847ced19cbe40d197ab807f008c5b4a21b8277d","subject":"Deprecate database params in neutron::server","status":"MERGED","created":"2020-04-03 07:43:35.000000000","updated":"2020-04-04 03:52:49.000000000","submitted":"2020-04-04 03:51:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"717207-1585972288121-e8a3316c","meta_rev_id":"fb39858c9f7a12a6ba7094e3dda09c3f98cb255f","_number":717207,"virtual_id_number":717207,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-04 03:51:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-04-03 14:13:32.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-04-04 01:37:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-04-04 01:37:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-03 14:13:32.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-04-04 01:37:41.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-04-04 03:51:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0eb135aac82c01d99a8dfc823aae9c7036ef7bb7","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-04-03 07:43:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"954c9bd746fef137dd2f30e64511d7ad1439c68a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 10:29:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a76cf691858d40de9dfeb68b5dae1a03 : SUCCESS in 7m 24s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2b8c69f19c6642f9a847631fd51c2ab5 : SUCCESS in 5m 57s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/361fd2759cbf45528afb1b8d546804f8 : SUCCESS in 7m 00s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/9e6f302fe5bd4420ab620fa30ddc03fa : SUCCESS in 25m 07s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/8ded3da7ca0242cda83f613310875f4c : SUCCESS in 24m 27s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/49213a114e6f4bf8b6e245bd3581c6cf : SUCCESS in 21m 47s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/522bd8f9b9114593a7738f8ec7902e6d : FAILURE in 22m 43s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/8c131f8fa2584f678a1198ef1b1432fa : SUCCESS in 54m 38s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/33322e882a5f481187ba4f41c9f753cf : SUCCESS in 1h 04m 31s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/375d790b68c043819e19d2c0b9544766 : SUCCESS in 54m 08s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/2b57113b64064d4fb0ee7aa30da10bc9 : SUCCESS in 39m 17s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/201e4b5018374ec8a3b64275788280fc : SUCCESS in 1h 11m 28s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/fa6441c373f147f19c389ec5cd692d5c : SUCCESS in 1h 05m 25s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/567485651ff04ddc97b2e0dbe0617c9a : SUCCESS in 1h 07m 11s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a4beb7c2c2b64864a70024b98be1aadf : SUCCESS in 47m 09s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/f1aec5c5d51d4b179e0e624479e40d45 : SUCCESS in 24m 40s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6c7f023829ba418680f1bf70cf220482 : FAILURE in 9m 59s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8dd00bd8a504c508337a36199f6eaaf : SUCCESS in 5m 34s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f3c8be9a8af24a69a5eb98f3e86dbbfd : SUCCESS in 2h 26m 44s","accounts_in_message":[],"_revision_number":1},{"id":"68db314962d4dc416052903f018deec30a5dbe19","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-04-03 14:13:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"de59d81874e6c62be643d7ca2dc3d0f198ed123d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-04 01:37:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"36940b6ec84b0b96d275d66316985f62ff4e3438","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 01:37:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7810b96ae3d9bb152f0729dc1c933d775bd85b8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 03:51:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aa095f79ad894e529bedba9779d3e2d6 : SUCCESS in 4m 23s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c5e168a05fb34e1a8ee9da2c302b4aa2 : SUCCESS in 4m 04s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/44bfb817b13c4c69a84accd34a58c8d0 : SUCCESS in 4m 15s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/7807470c38b748a0b4514d6bf464c5d3 : SUCCESS in 22m 58s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/cd6d7b892b244a6083023e29adc3b323 : SUCCESS in 24m 16s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6927f5dfe9ae4503b002538e752a6a8d : SUCCESS in 20m 15s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3c10c1e4f7cd4766a97f6ac4f69b27f1 : SUCCESS in 54m 19s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ea80c6be61c6443ab7df4d87ee6017b4 : SUCCESS in 59m 47s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/492a54f36f564eda9754b9af95f69c5c : SUCCESS in 1h 00m 22s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1ea957b208584f64b52fdd4053c41051 : SUCCESS in 42m 03s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/153bb537385548aca7527c312e4f666b : SUCCESS in 19m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63c0c8fb43544fad950eb297c7718068 : SUCCESS in 3m 22s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f2028640249c46749538c7e9107fb4c0 : SUCCESS in 2h 09m 23s","accounts_in_message":[],"_revision_number":1},{"id":"6ce02e8e58fcc14232cd59ee78d9f8997f27da04","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 03:51:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fb39858c9f7a12a6ba7094e3dda09c3f98cb255f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 03:52:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f2a9298f0d6d4fa9b25a9217f142130b : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f9330a50251a41a8ca29514442d187c385a2448b","revisions":{"f9330a50251a41a8ca29514442d187c385a2448b":{"kind":"REWORK","_number":1,"created":"2020-04-03 07:43:35.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/07/717207/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/07/717207/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/07/717207/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/07/717207/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/07/717207/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/07/717207/1"}}},"commit":{"parents":[{"commit":"3489c25a0fe9ed36a749f6ebed802581fae64f81","subject":"Deprecate min_pool_size option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/3489c25a0fe9ed36a749f6ebed802581fae64f81"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-04-03 07:43:10.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-04-03 07:43:10.000000000","tz":120},"subject":"Deprecate database params in neutron::server","message":"Deprecate database params in neutron::server\n\nLets get rid of the pick\u0027s that\u0027s been here\nfor years now.\n\nChange-Id: I5847ced19cbe40d197ab807f008c5b4a21b8277d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/f9330a50251a41a8ca29514442d187c385a2448b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/f9330a50251a41a8ca29514442d187c385a2448b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
