)]}'
{"id":"openstack%2Fpuppet-neutron~717239","triplet_id":"openstack%2Fpuppet-neutron~stable%2Ftrain~I9c9b681d356b11ff4bafeb37c7c19459bb99cf2d","project":"openstack/puppet-neutron","branch":"stable/train","topic":"bug/1850236-stable/train","hashtags":[],"change_id":"I9c9b681d356b11ff4bafeb37c7c19459bb99cf2d","subject":"Fix rabbitmq dependency","status":"MERGED","created":"2020-04-03 08:36:00.000000000","updated":"2020-04-03 14:27:40.000000000","submitted":"2020-04-03 14:27:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"717239-1585924060560-fe09d30a","meta_rev_id":"ed57e8dbcdd60ec2804f89c9dfe3735396c2a060","_number":717239,"virtual_id_number":717239,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-03 14:27:40.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-04-03 14:11:32.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2020-04-03 08:51:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-04-03 14:11:32.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2020-04-03 08:36:00.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-03 08:36:00.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2020-04-03 08:51:21.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-04-03 14:11:32.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-04-03 14:27:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"91da7382096f79e9c062edb75c6278a0316b9df8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-03 08:36:00.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"471b55bb6bf2de74a1da396f12a69159e45938a0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-03 08:36:17.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 0ea1a5daecc3f3d1579653b160249856878036b0","accounts_in_message":[],"_revision_number":1},{"id":"84b356c1909710a875d743b21a2dbb86ae740cdc","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-04-03 08:51:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b73623f2057f76585f2b0c72590caf11e3b45ebe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 08:53:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/2a1be11fb8414d5ab649fa870f3f150d : SUCCESS in 4m 34s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/16098f33517b43bd95e9dc5fa6b0e4c9 : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":1},{"id":"0d20e8ce238abf60b2e6325678ef6549ded02c8d","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-04-03 14:11:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b4083edf0eee2b80b91bfa26721776c0b99a2379","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 14:11:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9e595a15bc2518d3f9f055696177f1fcff11a51e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 14:27:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c373bc8cf1e00946e56b68298ee47c811e1cf31b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 14:27:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/44f3c170e6e94fbea31e7c93e34f1d92 : SUCCESS in 4m 55s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6041c9fc5d8c489a99acf4e51d404b28 : SUCCESS in 7m 29s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5e41b95ec01a5ca4dc643bd12335621afa2a8c45","revisions":{"5e41b95ec01a5ca4dc643bd12335621afa2a8c45":{"kind":"REWORK","_number":1,"created":"2020-04-03 08:36:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/39/717239/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/39/717239/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/39/717239/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/39/717239/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/39/717239/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/39/717239/1"}}},"commit":{"parents":[{"commit":"afc5750e191868099c1e5319dbf09e44e4f69e0f","subject":"Force cast bridge mappings as an array","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/afc5750e191868099c1e5319dbf09e44e4f69e0f"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-10-29 01:31:05.000000000","tz":480},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-03 08:36:00.000000000","tz":0},"subject":"Fix rabbitmq dependency","message":"Fix rabbitmq dependency\n\npuppetlabs/rabbitmq [1] is deprecated, and max version is 5.6.0 which can\nnot be fulfilled by this dependency.\n\nCorrect it to puppet-rabbitmq [2] instead which is being actively maintained\nand has the right version.\n\n[1]: https://forge.puppet.com/puppetlabs/rabbitmq\n[2]: https://forge.puppet.com/puppet/rabbitmq\n\nChange-Id: I9c9b681d356b11ff4bafeb37c7c19459bb99cf2d\nCloses-Bug: #1850236\n(cherry picked from commit 29f5049be21b57b32de8290935774bbe03a51dbc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/5e41b95ec01a5ca4dc643bd12335621afa2a8c45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/5e41b95ec01a5ca4dc643bd12335621afa2a8c45"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
