)]}'
{"id":"openstack%2Fpuppet-neutron~721539","triplet_id":"openstack%2Fpuppet-neutron~master~Id96ab9af7ac5776d5a0000889d26e494c68d9376","project":"openstack/puppet-neutron","branch":"master","topic":"domain_id","hashtags":[],"change_id":"Id96ab9af7ac5776d5a0000889d26e494c68d9376","subject":"Deprecate domain_id parameters","status":"MERGED","created":"2020-04-21 07:33:46.000000000","updated":"2020-04-23 20:55:09.000000000","submitted":"2020-04-23 20:53:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721539-1587675210281-a0c18567","meta_rev_id":"9214a87069908dc14ef5d5136acc2521e076f249","_number":721539,"virtual_id_number":721539,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-04-23 15:53:02.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-23 20:53:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-04-22 18:40:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2020-04-23 08:44:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2020-04-23 13:22:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-23 08:44:21.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-04-23 13:22:26.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-04-23 20:53:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8b9491fcb98c83e522d157f471e5011dcce5f8d1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-21 07:33:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4cc6e4aa4bfdeecf0f547b92a8649072514cdb65","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-21 07:35:21.000000000","message":"Topic changed from domain_id to domain_id-removal","accounts_in_message":[],"_revision_number":1},{"id":"3f634337baa99d9546215e8f16e4a4a142e8c46f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 10:22:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1d36e7232c9742ca9f11ce1d404a67dd : SUCCESS in 6m 44s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/41bb1a1a82724eef9b06bf91882ab66d : SUCCESS in 4m 59s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/992476c87954494e9191e866fea74340 : SUCCESS in 6m 42s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/930daf3c9f32453c88617a6bc9f00930 : FAILURE in 25m 03s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/1a9e232279a74cafafbade4848a019fa : FAILURE in 26m 51s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5f417960523242789e72e519134c2d65 : FAILURE in 20m 28s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/659f87702ec84737b8fff5d769a3b37c : FAILURE in 19m 55s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/91d20bf7f71b4dde925145284db5efe2 : SUCCESS in 44m 32s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fb020d2066d64a83ae93fc69cc67f2c5 : SUCCESS in 52m 41s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/138f47affaf947d89cf857ea34f04856 : SUCCESS in 40m 52s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/96edfb7a110445fcbc6de4d5e07fc715 : SUCCESS in 35m 33s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/be595866d747402086a64b6a9501ebbd : SUCCESS in 55m 20s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7726772edc264cb58909da860ffdd566 : SUCCESS in 1h 10m 08s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c369f874ae324c90a4e9d31c4f39e9b0 : SUCCESS in 1h 00m 48s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8e394e8b19bb4bad864c9523181a53dc : SUCCESS in 44m 20s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/f96d55d84a244405b6b413931737730c : SUCCESS in 31m 30s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/a38363b6579642e7b47b224cbd1c9c60 : FAILURE in 11m 02s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/42cdf600120843d8b298d3d6c83b089b : FAILURE in 7m 45s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8a1e8a1f35147bab29f6b311b942fdb : SUCCESS in 7m 00s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8fc70f2855f14306b234fdaad9532c10 : SUCCESS in 2h 45m 38s","accounts_in_message":[],"_revision_number":1},{"id":"a23d1e60f891dc407060b802589fd267ae32ab5e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-21 13:07:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8812639c45738d91ef5d43d25c3d2545d1548259","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 15:49:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/26aaaf17465d491cbdf4a64151fdc400 : SUCCESS in 5m 34s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ae5dab2fb5344702a0bb48bd0bf6f2b5 : SUCCESS in 5m 45s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/29058295689c486e9f963c6cbeb7be94 : SUCCESS in 5m 02s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/4d756ea99c8d4864b7988ea7cdbebb7c : SUCCESS in 18m 59s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/80403757496d405c9cbec3fef00f1f26 : SUCCESS in 23m 40s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/506116b88e68423091c601c0175037e2 : SUCCESS in 18m 01s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1d49b6bc2eb14284977a3f4cb1d57d7f : FAILURE in 22m 01s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/2ec16303475f4e6cb616bdb32c895017 : SUCCESS in 48m 27s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f55ac9b66b4a4502a1701397d5d35f59 : SUCCESS in 59m 41s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0dc1916d8513483598915dbf27aaa1c5 : SUCCESS in 44m 27s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/986efb098b524744b4e107931773483f : SUCCESS in 35m 17s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/76eac573995c493bbbb8b8f5e83c03a3 : SUCCESS in 1h 16m 28s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ed6b24c70c7944b0969f91bd307e823b : SUCCESS in 1h 07m 32s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/20fc998a7c2540afaf2319e478b594e8 : SUCCESS in 1h 06m 21s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/36941841a47f4d8394f232944f5309df : SUCCESS in 47m 26s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/0bd5aa3c9e3243d6a5da486deb93f3a7 : SUCCESS in 21m 12s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/b0a7552fea314f1eb562e1ad4fcd5bdd : FAILURE in 10m 06s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9ea0995da5124284abeabf960c81fb9f : FAILURE in 9m 03s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/832d87f47dc6499d8fbe63953fc530e0 : SUCCESS in 3m 49s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c88ce008876a45e9ada1f02338f60441 : SUCCESS in 2h 35m 42s","accounts_in_message":[],"_revision_number":2},{"id":"0d14b51fc1c529ab0adf750911c267f01bb8015f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-04-22 18:40:20.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8e4c9d3342b52d80e12bb07baa0bb0f34d3ab987","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-04-23 08:44:21.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"456b60ea89e05d022a8666d44f434f2bf17b9aae","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-04-23 13:22:26.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3ba3cd5c8b8831c1c7c8eb72c7f99be26c1c4381","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 13:22:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cd15f575080206a3210fd9c21da8dbe4ed2d274a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 15:46:00.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ef95817638dd463fbf3aeffa66b4499c : SUCCESS in 7m 35s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/19f750574e6b4a65bea0a0a112e4b8cb : SUCCESS in 8m 29s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ed38bafe9625462eba59a9aad86a3aff : SUCCESS in 5m 26s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/05aaf07f6bf64f15a8938ff2f3ec6572 : FAILURE in 18m 36s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/a3921f590e7a4b1083be71468c4e5e87 : FAILURE in 18m 31s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8272fb41066f43fc9d4dd57448bd67f0 : SUCCESS in 21m 53s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7cb15d7175604e8783e0c7b1ddee9ed6 : FAILURE in 15m 43s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f39acb4873d14f1ea23c683b72b6200b : FAILURE in 13m 00s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/64f2f143f29a43098990f69631ea48c8 : FAILURE in 14m 20s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d8aa3b0e72a6435fac4e8fffafae64c2 : FAILURE in 14m 12s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/39f859cdf83246c3a035471827965f69 : FAILURE in 19m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e37ceb9af4644a49efb67b620ae7df1 : SUCCESS in 6m 50s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0ed41e7a64334010ab0670092845449d : SUCCESS in 2h 16m 33s","accounts_in_message":[],"_revision_number":2},{"id":"e3cbdcac90a6d12a04c0ad6b65a475f40b2794f4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-23 15:53:02.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"44cf2a035c95aac391b6b272329ebe45d7bf1966","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 18:30:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e66727649c3a4e388215d3c3da8a9dd6 : SUCCESS in 4m 49s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/05f41220d2854938a2aaaa0bc842808c : SUCCESS in 5m 56s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dab6488c8c604f86b2c75df86d964563 : SUCCESS in 6m 26s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/4cc6a6b2f5f543f8a8e13b4e3461e2ec : SUCCESS in 21m 54s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/93f204c8fc7b4d9c93ef1fbd9744ff3f : SUCCESS in 23m 50s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c088c3f592a341128045263711ff38f9 : SUCCESS in 18m 52s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1d8ca2f7c44e44cca1ef37935c433727 : FAILURE in 21m 55s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/e9cfbab558c34940a25f2ac7981cf7ba : SUCCESS in 50m 37s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5a9f252ba22f423c8a54fbb78478d3d8 : SUCCESS in 51m 43s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/00065ad014454e31bd53daabe2d87842 : SUCCESS in 45m 47s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/0d0a8a3bc7814abaae289eb4d39cfc81 : SUCCESS in 39m 13s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9a0db8a8e8fa45b884e31ffa629713a6 : SUCCESS in 1h 01m 13s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c31528fce405437d9bc7b1b58f9966c4 : SUCCESS in 1h 06m 51s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/166ed4d2ac32456d8aee85c06c04d138 : SUCCESS in 1h 14m 10s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/fcaa63870f6c4f2493d70228d072750c : SUCCESS in 45m 16s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/8433692e4093414eb3b49f0ab16dabcf : SUCCESS in 20m 12s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/fc165de462fb429f8871960c359229c3 : FAILURE in 12m 26s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e3e278587d9d455ebb4953422850db7f : FAILURE in 6m 56s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5318c14232d84e8989381a46087411c3 : SUCCESS in 4m 28s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/dfe983f27ad342e38018185809bb90d8 : SUCCESS in 2h 30m 02s","accounts_in_message":[],"_revision_number":2},{"id":"50ffec25db2359e9c6090e82c1ec808b0a33cb13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 18:31:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9584ed7c8e4c4de66f780e55050f1b21345c92d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 20:53:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/962720a60fbf470d94088d82004d4e52 : SUCCESS in 5m 13s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1dcc9efcd85e46dbb71af5d15c2f26e8 : SUCCESS in 5m 02s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0c33f914c8594f6ea8d8149b19be7519 : SUCCESS in 4m 52s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/ddb2cef782b04657a862deaa01971fe4 : SUCCESS in 18m 35s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/07a58a879d9a46719608a61a231268d7 : SUCCESS in 18m 58s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/09d5b55077f5446989ac27e2ecb650f0 : SUCCESS in 16m 18s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e7645eaca2f846b5ab7561f72a090690 : SUCCESS in 54m 26s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f329625cf6b04ef7a0b81c45a5288e55 : SUCCESS in 1h 05m 37s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9d80eef8cb3b42438956dae8be87a954 : SUCCESS in 1h 13m 09s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c26b05a3f6364cebb4398b543d9272ca : SUCCESS in 43m 50s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/7de37b06cb3a48cebf0875c738797422 : SUCCESS in 20m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/238ed85305dd44c392e154456b8ea831 : SUCCESS in 3m 56s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3e1c1d3f9687451aa0234a8abb63f876 : SUCCESS in 2h 20m 50s","accounts_in_message":[],"_revision_number":2},{"id":"e425debb8d3a13933beb926713f785b50b3e24e7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 20:53:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"5cfb94a0d284c84a82adcae6d8caa6a02c035c41","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 20:55:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c214a61fc354a03bd9eff622564d5d2 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"dfeccd674988f34edea5ca7b95bdeea827bab000","revisions":{"c0e217ff1e9a7df72044f9ef33d98a0c1272678e":{"kind":"REWORK","_number":1,"created":"2020-04-21 07:33:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/39/721539/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/39/721539/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/39/721539/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/39/721539/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/39/721539/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/39/721539/1"}}},"commit":{"parents":[{"commit":"a641c9b051fbdd2b21124e72a297375f06b4076d","subject":"Merge \"Add support for resource provider bandwidth in Neutron config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/a641c9b051fbdd2b21124e72a297375f06b4076d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-21 07:29:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-21 07:29:57.000000000","tz":540},"subject":"Deprecate domain_id parameters","message":"Deprecate domain_id parameters\n\nCurrently we set both domain_name and domain_id for some configuration\nfiles, but it makes it difficult for operators to use their original\ndomain because they should update both parameters correctly.\n\nBecause it\u0027s redundant to required both name and id set, this patch\ndeprecates all domain_id parameters, so that operator needs to set\ntheir domain name, which is predictable, before creating actual domain.\n\nChange-Id: Id96ab9af7ac5776d5a0000889d26e494c68d9376\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c0e217ff1e9a7df72044f9ef33d98a0c1272678e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/c0e217ff1e9a7df72044f9ef33d98a0c1272678e"}]},"branch":"refs/heads/master"},"dfeccd674988f34edea5ca7b95bdeea827bab000":{"kind":"REWORK","_number":2,"created":"2020-04-21 13:07:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/39/721539/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/39/721539/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/39/721539/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/39/721539/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/39/721539/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/39/721539/2"}}},"commit":{"parents":[{"commit":"a641c9b051fbdd2b21124e72a297375f06b4076d","subject":"Merge \"Add support for resource provider bandwidth in Neutron config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/a641c9b051fbdd2b21124e72a297375f06b4076d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-21 07:29:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-21 13:07:07.000000000","tz":540},"subject":"Deprecate domain_id parameters","message":"Deprecate domain_id parameters\n\nCurrently we set both domain_name and domain_id for some configuration\nfiles, but it makes it difficult for operators to use their original\ndomain because they should update both parameters correctly.\n\nBecause it\u0027s redundant to required both name and id set, this patch\ndeprecates all domain_id parameters, so that operator needs to set\ntheir domain name, which is predictable, before creating actual domain.\n\nChange-Id: Id96ab9af7ac5776d5a0000889d26e494c68d9376\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/dfeccd674988f34edea5ca7b95bdeea827bab000"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/dfeccd674988f34edea5ca7b95bdeea827bab000"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
