)]}'
{"id":"openstack%2Fpuppet-neutron~724294","triplet_id":"openstack%2Fpuppet-neutron~stable%2Ftrain~I32ee1e9dac3aa97add161aa17c2e87584b6c17a2","project":"openstack/puppet-neutron","branch":"stable/train","topic":"bug/1861032-stable/train","hashtags":[],"change_id":"I32ee1e9dac3aa97add161aa17c2e87584b6c17a2","subject":"Add \u0027dnsmasq_enable_addr6_list\u0027 option","status":"MERGED","created":"2020-04-29 12:10:02.000000000","updated":"2020-04-30 10:48:29.000000000","submitted":"2020-04-30 10:47:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"724294-1588243624539-3c423189","meta_rev_id":"5d73879b2bd226764e28e6ee52ff10c885a00a19","_number":724294,"virtual_id_number":724294,"owner":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-30 10:47:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2020-04-30 08:28:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-04-30 07:31:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2020-04-30 08:28:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-30 07:31:24.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-04-30 08:28:00.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-04-30 10:47:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0e48d031c01749116841274fc4b250e6904b9325","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2020-04-29 12:10:02.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"aa36bb849ce9bdd4648727ddcf163806cae66e4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 14:13:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9f6f99e6da894c2c982925b7b309a732 : SUCCESS in 4m 46s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5f367e70362e49aeb27d914274d38124 : SUCCESS in 5m 13s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b04f327716454e2bafada37c45acb045 : SUCCESS in 4m 26s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/81a01480f8c44cba9b161595224c2bbc : SUCCESS in 32m 02s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/6c975fc4f45e44d6b5b62034caccd41b : SUCCESS in 19m 03s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/af4f818472e04591891988269e99a4de : SUCCESS in 19m 34s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9108ac452ec94e19ba627f28b4aa7ba1 : SUCCESS in 17m 43s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a583e77c7269493e86fd44c6a1bb9762 : FAILURE in 20m 51s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/055fb2802d3c45f9b4e10310fa97388c : SUCCESS in 54m 13s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/fd34da5d08254afb8d7d5e7dddd1c545 : SUCCESS in 56m 47s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/2b8948c613ab4b4aa6182e796a13096d : SUCCESS in 1h 03m 19s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/aa83af91ae224ddcbc3e6ef5150798f5 : SUCCESS in 46m 05s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/e545784a6655461eb7fb3e732e142b97 : SUCCESS in 43m 40s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2326e581a14941ef9b911d3fac583d93 : SUCCESS in 43m 53s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/025b31378d2f4f24b71de9cc8f35d8df : SUCCESS in 37m 12s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/0125f64256244e448d64120189aa8bdf : SUCCESS in 32m 06s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b0e1413010a34c3eaff5f70e4e494b48 : SUCCESS in 48m 23s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a7614d65c76a4ad8a24b7d312c4ba97d : SUCCESS in 54m 35s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3babc0b99d984551836e583ce7034413 : SUCCESS in 1h 06m 22s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/37948fe82a84449aa8d1b2c49ec7212c : SUCCESS in 1h 00m 30s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/24f844f6de01489490c358a3198bf0bd : SUCCESS in 18m 04s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a2bfc444e6cc4fb5a93f3f8066fcb6b7 : FAILURE in 17m 48s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5428a53213a147909692cc2052eeba43 : SUCCESS in 3m 18s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/83c05dd0255541b0821437b1b1c73811 : EXCEPTION","accounts_in_message":[],"_revision_number":1},{"id":"0c546f13a1db9719a6fae0e6c3bd59458a9541f0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-29 14:41:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f6698e0f7d86ff9825f0c217740486483cf8cca6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 17:39:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/87d18bcf5ea64d4dacf7ebc185a4c4a6 : SUCCESS in 6m 48s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/844a19645d7b4d108adbc758e83f6ae9 : SUCCESS in 4m 55s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/810a87dcc92b4bacaba4d76d1f818190 : SUCCESS in 6m 12s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/aadf59a3a91144a9ab28605e529e03f1 : SUCCESS in 32m 25s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/f43b4e2d6d7e4910a40ae83dd96e3226 : SUCCESS in 24m 56s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/b1970a38c4854caa9840ca9aa9326e8d : SUCCESS in 22m 17s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/06a082170b7d47f2956923832f63acff : SUCCESS in 21m 45s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b9540cb2fe0c46bca0a98864c11529e9 : FAILURE in 22m 57s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/496abf9a6b814604b1a65ea8c5a64bf2 : SUCCESS in 1h 03m 11s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/d28d8ea01f1b44a1b9e582018fa4773d : SUCCESS in 1h 01m 16s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/f2c2a83edcc64fa69a9f95dbcaf7b2e9 : SUCCESS in 56m 22s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/6f4c0032e5ea438a847c90f363c65af5 : SUCCESS in 49m 32s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/f283117a413b4907b9caf38132d8f9b5 : SUCCESS in 49m 07s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1c801a9d61d6479694b118b7b22d5491 : SUCCESS in 53m 26s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/64e9e96d87fc46b7a341cad1f49de99b : SUCCESS in 44m 14s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/b2d9aef931f64ef28a7e0a1ba042f2d1 : SUCCESS in 51m 02s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4a780ca03cde4ae0b26bdd3c8d6d0c79 : FAILURE in 58m 47s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/fb95b586c7e640ebb1d9542c0afd5e5f : SUCCESS in 1h 04m 16s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/cd7a86c72ea2411db648000a594ac430 : SUCCESS in 1h 03m 15s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6a8c5083ec004835b39567b005c72543 : SUCCESS in 46m 48s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/fda21643417442c18647de8c7e2d708d : SUCCESS in 19m 42s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ca37ee196641491380db4ac68935cedd : FAILURE in 21m 19s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf58d69311b64a78b47bd5637722b6e9 : SUCCESS in 5m 49s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4ffe6220852146548b1b06d9cb8c6a55 : SUCCESS in 2h 58m 15s","accounts_in_message":[],"_revision_number":1},{"id":"a391f63b4e53c764c993ad6b6fd1c312973320a5","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-04-30 07:31:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c5f619fe3638a076d9e7155d8f949e88c3bdec6e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-30 08:28:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nConfirmed that the master patch was merged before ussuri cut off, so we cat start backport from stable/train.","accounts_in_message":[],"_revision_number":1},{"id":"79c7a0387c7bca611be4cb68362e09002abb9b97","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 08:28:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f3b4ef5f24374714afe8c477e3c2c681ae13c73f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 10:47:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"44374d6111949071de68ddb704fe60c3588ccee8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 10:47:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/207726fa3df448f98415a8e50d1a4b54 : SUCCESS in 5m 28s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b7b3bcb24fdd43f4b80164918a53b423 : SUCCESS in 5m 51s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a6ed0ec69ced4ab690d37c03a4372499 : SUCCESS in 5m 00s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/dabec03d16b04d9b9fa62aff311388ee : SUCCESS in 26m 16s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/75bff48fe103448fa52c1c733c420c26 : SUCCESS in 15m 55s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/15188773a056432a90c79edc556528fa : SUCCESS in 50m 42s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/5929c4847cf0444aa312a6734760fa6d : SUCCESS in 57m 07s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/77684e634382487eb52eab01d07327d0 : SUCCESS in 59m 22s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/6c379ea3a6ca4bb39fd1547fd6084e13 : SUCCESS in 50m 38s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/99ef5d24722d44ed92b1d3ad51948d03 : SUCCESS in 18m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7793a48c04e403e80124e20b6292767 : SUCCESS in 4m 40s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9eea9adb33c24c1bb32a47652fd6431e : SUCCESS in 2h 15m 46s","accounts_in_message":[],"_revision_number":1},{"id":"62db698f6deefd5c2e1c6aaa70e7f3a6b0333947","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 10:48:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a61dda51c7f54a13a1395c8e42dff929 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"731a8fd7e51513fa5d2d3968cada7341070eb848","revisions":{"731a8fd7e51513fa5d2d3968cada7341070eb848":{"kind":"REWORK","_number":1,"created":"2020-04-29 12:10:02.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/94/724294/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/94/724294/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/94/724294/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/94/724294/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/94/724294/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/94/724294/1"}}},"commit":{"parents":[{"commit":"7001e6627e25e14d04f013c8a532aabbeda785ea","subject":"Refer keystone_authtoken instead of neutron to detect region","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/7001e6627e25e14d04f013c8a532aabbeda785ea"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2020-02-27 19:32:27.000000000","tz":60},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2020-04-29 12:10:02.000000000","tz":0},"subject":"Add \u0027dnsmasq_enable_addr6_list\u0027 option","message":"Add \u0027dnsmasq_enable_addr6_list\u0027 option\n\nAdd support to control the dhcp agent configuration\noption \u0027dnsmasq_enable_addr6_list\u0027.\n\nRelated-Bug: #1861032\nDepends-On: https://review.opendev.org/704436\nChange-Id: I32ee1e9dac3aa97add161aa17c2e87584b6c17a2\n(cherry picked from commit e3a932d02507e4a8f0e6b816669d3eea52813874)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/731a8fd7e51513fa5d2d3968cada7341070eb848"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/731a8fd7e51513fa5d2d3968cada7341070eb848"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
