)]}'
{"id":"openstack%2Fpuppet-neutron~724528","triplet_id":"openstack%2Fpuppet-neutron~master~I37c384924b30891df852a20633e2dede1bcaae85","project":"openstack/puppet-neutron","branch":"master","topic":"update-rabbitmq-version","hashtags":[],"change_id":"I37c384924b30891df852a20633e2dede1bcaae85","subject":"Fix rabbitmq version in metadata","status":"MERGED","created":"2020-04-30 06:05:25.000000000","updated":"2020-04-30 06:51:11.000000000","submitted":"2020-04-30 06:51:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"724528-1588229471891-3c423189","meta_rev_id":"812caa8185a55f16336d8798a90aebc4a485f7e2","_number":724528,"virtual_id_number":724528,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-30 06:51:11.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2020-04-30 06:29:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2020-04-30 06:29:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-30 06:51:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"82dc92bd98f86680d80bf3ac54c1c711f196cfe8","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-04-30 06:05:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e6e6998f288f27764be717d210887e7ec66347f","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-04-30 06:29:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fd3bf124ef6ef49fdbf03ac0e6041f58f553a3f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 06:38:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/c9223d50e284479ab22cc6a798ca6b36 : SUCCESS in 4m 59s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/77c3873104fe4ea295b2e3f150fb69d9 : SUCCESS in 6m 08s","accounts_in_message":[],"_revision_number":1},{"id":"cc9e6f7d2bf80d446ed31bb68ed9f69e231544e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 06:39:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"78ecf114e45a93675cc435362ea0215c80383076","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 06:51:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"458f8db3f965c46013200d81f4e9723f9416d4ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 06:51:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/aa90a8b907cd4436876b0b59ab05720f : SUCCESS in 5m 32s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/496bcdd5800941c09a2a56524ff1e95a : SUCCESS in 6m 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0f1f7fa98a0354cf1bd8a7fdded5b6ddda3f3921","revisions":{"0f1f7fa98a0354cf1bd8a7fdded5b6ddda3f3921":{"kind":"REWORK","_number":1,"created":"2020-04-30 06:05:25.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/28/724528/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/28/724528/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/28/724528/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/28/724528/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/28/724528/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/28/724528/1"}}},"commit":{"parents":[{"commit":"1b7c1874c897a1053c7770f87fb7729fca23a473","subject":"Change rpsec to rspec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/1b7c1874c897a1053c7770f87fb7729fca23a473"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-04-30 06:04:44.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-04-30 06:04:44.000000000","tz":480},"subject":"Fix rabbitmq version in metadata","message":"Fix rabbitmq version in metadata\n\nrabbitmq is now using the 10.0.1 version, so we should fix the\nrabbitmq version in metadata.\n\nChange-Id: I37c384924b30891df852a20633e2dede1bcaae85\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/0f1f7fa98a0354cf1bd8a7fdded5b6ddda3f3921"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/0f1f7fa98a0354cf1bd8a7fdded5b6ddda3f3921"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
