)]}'
{"id":"openstack%2Fpuppet-neutron~725065","triplet_id":"openstack%2Fpuppet-neutron~master~Id9d14c06abc12547ad2ca894b369baf864e6418b","project":"openstack/puppet-neutron","branch":"master","topic":"uwsg-config","hashtags":[],"change_id":"Id9d14c06abc12547ad2ca894b369baf864e6418b","subject":"configuring neutron-api-uwsgi.ini","status":"ABANDONED","created":"2020-05-02 21:20:36.000000000","updated":"2021-04-18 16:02:17.000000000","total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fd4e090b82d8510580f75267f4cae9996d680639","_number":725065,"virtual_id_number":725065,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-06-24 06:09:54.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"date":"2020-05-27 17:29:29.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"date":"2020-05-26 08:53:51.000000000","_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-05-25 12:59:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-11 22:15:44.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-05-25 12:59:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-27 17:29:29.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-06-24 06:09:54.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"0648236c5d6eaf17d34f78da009cddd10d403830","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-05-02 21:20:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"74f5d4850d78810548b52437440b6bdb6d1ce988","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-02 23:26:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d8e7f80ffcff4d08aa8eae1310ffb116 : SUCCESS in 4m 26s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8d39583e7c5241d7ab9c9296c7d57aa6 : SUCCESS in 4m 00s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2c559c6117304e4abfdb8c3e5bf1a50b : SUCCESS in 4m 14s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/f6ebac5f87d64b5d9c0c21b019ac073d : SUCCESS in 19m 22s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/60a67bf96f1647f09518b5d58a7be150 : SUCCESS in 20m 04s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e10641d0f7b94f08829ba3fc818aad59 : SUCCESS in 18m 24s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6909e408bc48420b8b7ab58365718455 : FAILURE in 18m 49s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/89cad662d748452fb3f1af6e0ca5087f : SUCCESS in 47m 06s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2ac141e9234e4242a9ecc12969711b25 : SUCCESS in 51m 08s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d0d4e705679b4a899a6317e7975644f4 : SUCCESS in 43m 58s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/b746272a7b47432ca69ca7cfb0f49e16 : SUCCESS in 35m 35s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/188b97944b2e42e0ae5c674885fa6738 : SUCCESS in 53m 25s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1a9a3c532a904b8f8e58bd543cfa653a : FAILURE in 1h 07m 14s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/66fc45c089d74b07b3a593e063260257 : SUCCESS in 1h 02m 45s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f3d783513bff4f30930c21ea1fd7794e : SUCCESS in 48m 22s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/1831fa760e7a499188bc55bcc3d389f2 : SUCCESS in 18m 42s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/3cf39bc23d174afb921eb446638d722b : FAILURE in 9m 35s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fe90210421584f948fa97b8716a987fd : FAILURE in 7m 31s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5b0607b77cda4885a745d686cb569310 : SUCCESS in 2h 00m 54s","accounts_in_message":[],"_revision_number":1},{"id":"c7889230ac2cbc2b9708c1cfc1929208d6fdb702","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-05-03 11:25:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"baea5150cce183c4df70cf3abd1010ab3ce10970","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 13:47:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/92aa4c547fb2473181d712f9698248dc : SUCCESS in 4m 59s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d0b1a41b711d428d9e52b54c7b3b8193 : SUCCESS in 4m 13s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b1920dde5c18453e8cec15e08ef28f26 : SUCCESS in 4m 20s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/b34741dcb831475da04b3d2951723a1a : SUCCESS in 18m 31s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/dbc4afd32a9b409180ae9c2ebff242c9 : SUCCESS in 18m 29s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c44c8e06c3244bc3bf94dd10c0c4dedc : SUCCESS in 17m 46s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1d844f91eed1432dbff77cd3b8315c3c : FAILURE in 18m 42s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/89cb76150b3444f9a2e064715162b339 : SUCCESS in 47m 40s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/977f5d2cfc36463a8c251886550b95f4 : SUCCESS in 49m 51s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f189595a6ff74bc3bb9e6a4638580719 : SUCCESS in 38m 46s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/3b7666d17efb423ba3a55aacc3af46e8 : SUCCESS in 34m 42s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/07cda237fc2645008346b40bfb7b3f0b : SUCCESS in 53m 30s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/df57d72c59fa427ab8bedae07cacc9ee : SUCCESS in 1h 02m 50s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f8418488666a4d07865042f66f3d469f : SUCCESS in 1h 06m 02s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/77fc607ee4e647f0b3c626699d0ae696 : SUCCESS in 45m 47s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/36dd11d120104850bdf5d81bba64e166 : SUCCESS in 18m 51s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/149792ba61ee4fcba7e8f113ae7f30da : FAILURE in 9m 29s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/68bcad4b188e4eeab03a99bcc3d5ca90 : FAILURE in 7m 02s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/fe88e6ebcca04981b498addbb39abc97 : SUCCESS in 2h 20m 41s","accounts_in_message":[],"_revision_number":1},{"id":"a8f569a232eddd3345996c37523e5ab8611ca63f","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-03 15:10:57.000000000","message":"Patch Set 1:\n\nThis is not purely OpenStack related though, this would mean uwsgi would become an official way to deploy API services in Puppet OpenStack and this should be fixed for all modules, is that something you\u0027d want to have done and how would you implement that?","accounts_in_message":[],"_revision_number":1},{"id":"9cce6277747fda691e160477f00862e684473e52","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-05-03 19:11:48.000000000","message":"Patch Set 1:\n\nHi Tobias. In Debian, uwsgi is already the standard, in the packages, for all of the OpenStack APIs. There\u0027s nothing to be done on the puppet side, just run the service. What I would like to do here, is just make it possible to configure the /etc/SERVICE/SERVICE-api-uwsgi.ini that comes with all OpenStack packages.\n\nYes, I would like to do it for all services. I just start with Neutron, to see how it goes. I will then make it so things like number of workers can be configured by default using this neutron_uwsgi_config thing, for example.","accounts_in_message":[],"_revision_number":1},{"id":"49ee97f3a28288a7413538ea18bcdb155c6cbce8","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-05-03 19:17:07.000000000","message":"Patch Set 1:\n\nTobias,\n\nI\u0027ve opened a mailing list thread, since it may be more easy to discuss the topic there.","accounts_in_message":[],"_revision_number":1},{"id":"b03e062cdff806293b1628c738c37b3237a9ab17","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-05-11 14:59:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dd800a888099720d3529c01fb776c11097cd06f7","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-05-11 15:23:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f99c37461efdb4b9c1a04067751814f27c98a922","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 17:05:59.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f02a8283a1744bc9965bf391c742a2e4 : SUCCESS in 8m 29s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f86d10f2706248e0abfc08c7c1ce762b : SUCCESS in 7m 41s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6fed12ecd0d94a1fa097b8a1b2b434d9 : SUCCESS in 5m 55s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/a8038c3a1b46484293af48d0ea830a30 : SUCCESS in 18m 05s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/18b638ed230241b5a8ea8763952fbf62 : SUCCESS in 24m 32s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a086897e1ac646ccae1e3c15df1f4d26 : SUCCESS in 21m 54s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ddb51063afcf4952b0673960f9412a5e : FAILURE in 18m 37s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/d485d3f40d754a318ea5f738d4397504 : SUCCESS in 50m 59s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3652c74f33d84d2681fc4230ecbc82fd : SUCCESS in 1h 00m 23s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8dd9d55622294eb0acf4468b556dd6a4 : SUCCESS in 1h 04m 06s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/bdb84953b1a24168b9ffdf7784450e32 : SUCCESS in 37m 19s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4db44baa71c1487b99458e39660796e7 : SUCCESS in 1h 04m 01s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b77321ebc6d644cb8dda2bd116c088c3 : FAILURE in 1h 09m 43s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d6ab9d4f3e1a473f8d67d8d8479b3659 : SUCCESS in 1h 04m 49s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7349463a6ff3404d83dec44dfb6aa01e : SUCCESS in 48m 21s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/45b5368d0309441d99a73d12647a69f6 : SUCCESS in 24m 06s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/ba6c6514fa9e4244998a47f709d55c8a : FAILURE in 9m 32s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5e90aa1d98f448fb9b74bd50013dbb0a : FAILURE in 7m 11s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5bcbd32246814ff38cc3078d38526623 : FAILURE in 1h 39m 56s","accounts_in_message":[],"_revision_number":3},{"id":"dff69b486664f07ace46a459e67446004cdbe9d4","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-05-11 19:42:36.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"4168c41dbf6cad1800a49beb0055b1cabf7bd365","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 21:58:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/161ca38fb6a34a7dad91d14295514d17 : SUCCESS in 5m 28s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/be566dccd0114b96b4eb3c33ff7cc9af : SUCCESS in 5m 17s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3c906b000c4246598209ac17fc0481af : SUCCESS in 5m 13s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/2a3d2270a7864bcfb604a67adcfe9aad : SUCCESS in 18m 04s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/65e26ba28a0b4581bba00d2e8a1e2803 : SUCCESS in 21m 42s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4fd04684d1db4b18b96e4ca53399f293 : SUCCESS in 18m 46s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/36262753b20b44d2888e080d3f1a40f8 : FAILURE in 21m 09s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/ad474a9fa6e141e5909c1c78b7ffc85a : SUCCESS in 46m 24s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4771e419c0a7413b871e1a53c6c46ebb : SUCCESS in 49m 02s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3ed6e943ecce4720a00a26047699f7b8 : SUCCESS in 49m 05s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/aa83065ffa54434ea774da8a87ff36c1 : SUCCESS in 35m 56s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/194a496ed980426693f76c563de617cf : SUCCESS in 57m 17s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c6160936b4f740a18e46ae32a57c64e2 : SUCCESS in 1h 05m 14s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/53cfc5c403ae4dfa90486dc1bafaf129 : SUCCESS in 1h 21m 41s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/31aa90f1348e4133aa1cb0d72b936682 : SUCCESS in 44m 03s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/6ca19154fac24d688fff590f0558f5a9 : SUCCESS in 23m 46s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/4bac2f66531d48dbabe540a33640c95a : FAILURE in 14m 55s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4ea79e74897341408b1e037f8a8c0dc6 : FAILURE in 10m 21s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c9db6905bd874cb2a8fe7424f6aa6ae9 : SUCCESS in 2h 12m 47s","accounts_in_message":[],"_revision_number":3},{"id":"a09805d57ceff6b46e928a0c7ac5087b8393a3aa","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-05-11 22:15:44.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"fe1e643932a5d39dca26e4ea227c76ece4fbc964","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-20 01:53:50.000000000","message":"Patch Set 3: Code-Review-1\n\n(4 comments)\n\nThe implementation almost looks good. The problem is mostly about test codes.\n\nAlso, can you add a release note because this is a new feature ?","accounts_in_message":[],"_revision_number":3},{"id":"bcfa47df60a7930e6f3ae381cd9d2a3ccf9eaee8","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-20 08:39:57.000000000","message":"Patch Set 3:\n\nNot sure what I think about this change, I was hoping we could introduce an more abstract layer to adding different WSGI serving applications like httpd is today.\n\nBut checking around I can\u0027t see that there is any good, well maintained uWSGI module for Puppet.","accounts_in_message":[],"_revision_number":3},{"id":"2ececbf2fa4827487b12ef4a31b527f12cfdc27a","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-05-20 12:15:34.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"242a2b191b117f02b486a3455be3f8204582d905","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-05-20 12:15:57.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3d5123c559999ed8a6e5aa0dd47c1c6df15d1c58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 14:22:30.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b016ba4121cf4d949ffa624719110fec : SUCCESS in 5m 47s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3c39b6c374bc4b3cbaa8116d67070905 : SUCCESS in 5m 17s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7df764571d924a81a9d82aad65b0764a : SUCCESS in 5m 34s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/5e36f1c62f1a4f9d8dfe1579cc6a3f1f : FAILURE in 21m 19s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/cebf3e64c93d4efb89fc138076a712a6 : SUCCESS in 22m 00s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/37936bd727f84e1e9e2423b13f0003df : SUCCESS in 23m 07s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d9c3bf6cc9ae47d9b97b94b673b95ce6 : FAILURE in 22m 47s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/1141155ac4c04647affcca0d318f1fc4 : FAILURE in 38m 11s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b3073785c9404906926710c683300f94 : FAILURE in 37m 10s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/00d7ad6b0b08468bb45179ce0d119b06 : FAILURE in 33m 33s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/ec602adcbfb747a7a8fcc0119734bfa3 : FAILURE in 28m 48s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e8c4f0fc3b0c4d78ae294169022b9c61 : SUCCESS in 1h 04m 39s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d49fb16631de4d4ea394d467e640f46e : SUCCESS in 1h 04m 39s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/95d4cdb0d3074aeb9b14a73f36560428 : SUCCESS in 1h 01m 58s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e971faf5484a4c94ae87baaa01fb3334 : SUCCESS in 48m 32s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/aa82d6266ad7486ba939a163d720c9e7 : SUCCESS in 21m 02s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/0d512c23aec54295a346229878dfa055 : FAILURE in 12m 13s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4ba7cbc294b04c7c8747f2d6a21204e5 : FAILURE in 7m 38s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0dab3c0323bf40b4a5ac09dd79cc426c : SUCCESS in 2h 05m 20s","accounts_in_message":[],"_revision_number":4},{"id":"e5793697931d4898ed82bc6ff2d906d094ec9e3b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-21 12:54:17.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6ae105fbf35a6550ec139ef2d7924fdf5c7d4e5d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-21 12:56:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1c25ffc3ea3c12d457d8366cb300128a1699fed1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-21 13:00:49.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0863fb6f2f2158d27848d5eeb3f4668909d88a85","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-05-25 10:49:25.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4eb7783f81a8c2e70490f96790cc5e07b7ceceb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 12:59:24.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6b65fadad0964288aed37126ced8c372 : SUCCESS in 4m 53s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/988c9494f4b64f16a75aac8719d4c0a5 : SUCCESS in 5m 39s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0b46c3efc3d2431c900535afca9a50eb : SUCCESS in 5m 20s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/647b29ebe0814756bf6fa81443e5f41f : SUCCESS in 23m 02s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/c9832a273b3e42b2948b7f660c32a403 : SUCCESS in 24m 24s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/46c0cb0f16d8445495bd8aaff9eaa908 : SUCCESS in 23m 17s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/df24bad90a9f4b91bb0a29dae1102481 : FAILURE in 22m 10s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/6c8d1b3373b049f9b33d12b63843fa4c : FAILURE in 38m 38s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6711820a6d45481da34bc01bced26d9c : FAILURE in 42m 48s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6805cc6e256e41bca8e7b859534cdfd7 : FAILURE in 30m 01s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/1ff72ed5f32c4762a60218394decec4b : FAILURE in 29m 28s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1f0331182d8f42f082406441a7195364 : FAILURE in 37m 19s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c7820a600d064460987a28dbbc0f3b9b : FAILURE in 36m 28s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/fcde506dc0f8423a87ab8e64e87b1788 : FAILURE in 40m 57s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/853ac684bb484cf4bd2b3c6a9e4fec70 : FAILURE in 31m 30s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/5ff85553257a448189ce412c82d878cc : SUCCESS in 25m 38s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/59984c45028f4bed8d7a4512f8d58494 : FAILURE in 10m 46s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f324b8c7dea348019449e15b8bb67865 : FAILURE in 10m 30s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ac03701ab37642b69af8c830dfe7fd93 : SUCCESS in 2h 08m 59s","accounts_in_message":[],"_revision_number":5},{"id":"07edada48ee7b454fd477950970c3962a2ba3d04","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-05-26 08:53:51.000000000","message":"Patch Set 5:\n\n(2 comments)\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"fcc27a89bc59b92545f69a5d1a25f5cb02c11caf","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-27 17:29:29.000000000","message":"Patch Set 5:\n\nStill think this needs more thinking over on how to implement it instead of just winging it and implementing new resource providers.\n\nWhat is the long term plan here? How can we integrate uWSGI more seamlessly, why should OpenStack related Puppet code configure uWSGI related things, it\u0027s like we would introduce resource providers to configure Apache, doesn\u0027t make sense.","accounts_in_message":[],"_revision_number":5},{"id":"13d0d841a3c175600fb952172a0b87d39ec15c44","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-06-24 06:09:54.000000000","message":"Patch Set 5:\n\nSorry for being late after review request.\n\nAfter taking some time, I agree with Tobias about his thought for some extent.\n\nHowever, oh the other hand, problem is that I can\u0027t find any popular puppet module for uwsgi, which is now actively maintained.\n\nSo I think what we can do would be to implement a provider and a base class in openstacklib instead of implementing provider codes in all puppet modules, and use the common class from each puppet modules as we do for openstacklib::wsgi::apache for example.","accounts_in_message":[],"_revision_number":5},{"id":"fd4e090b82d8510580f75267f4cae9996d680639","tag":"autogenerated:gerrit:abandon","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-18 16:02:17.000000000","message":"Abandoned\n\nA better patch has just been merged.","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"bdcfd1333bef73c875b8d5c2f8dedf2df0953f2f","revisions":{"4c1c80ebf40ad6d3693f665bc8abcdf5c7eb4d6c":{"kind":"REWORK","_number":1,"created":"2020-05-02 21:20:36.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/65/725065/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/65/725065/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/725065/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/725065/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/725065/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/65/725065/1"}}},"commit":{"parents":[{"commit":"aad8d248f82899943176bffbce928b0d2e62703d","subject":"Bump openstackdocstheme to 1.30.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/aad8d248f82899943176bffbce928b0d2e62703d"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-05-02 21:19:44.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-05-02 21:19:44.000000000","tz":120},"subject":"configuring neutron-api-uwsgi.ini","message":"configuring neutron-api-uwsgi.ini\n\nDebian uses uwsgi to run all of OpenStack API. This patch is a first\nattempt to allow puppet-openstack to configure it.\n\nChange-Id: Id9d14c06abc12547ad2ca894b369baf864e6418b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/4c1c80ebf40ad6d3693f665bc8abcdf5c7eb4d6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/4c1c80ebf40ad6d3693f665bc8abcdf5c7eb4d6c"}]},"branch":"refs/heads/master"},"338da1a2d911453d67671c6c6c2602fb725a8f26":{"kind":"REWORK","_number":2,"created":"2020-05-11 14:59:52.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/65/725065/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/65/725065/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/725065/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/725065/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/725065/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/65/725065/2"}}},"commit":{"parents":[{"commit":"aad8d248f82899943176bffbce928b0d2e62703d","subject":"Bump openstackdocstheme to 1.30.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/aad8d248f82899943176bffbce928b0d2e62703d"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-05-02 21:19:44.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-05-11 14:30:08.000000000","tz":120},"subject":"configuring neutron-api-uwsgi.ini","message":"configuring neutron-api-uwsgi.ini\n\nDebian uses uwsgi to run all of OpenStack API. This patch is a first\nattempt to allow puppet-openstack to configure it.\n\nChange-Id: Id9d14c06abc12547ad2ca894b369baf864e6418b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/338da1a2d911453d67671c6c6c2602fb725a8f26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/338da1a2d911453d67671c6c6c2602fb725a8f26"}]},"branch":"refs/heads/master"},"7b7043054e823188516ad3283b22670f848c24e3":{"kind":"REWORK","_number":3,"created":"2020-05-11 15:23:07.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/65/725065/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/65/725065/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/725065/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/725065/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/725065/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/65/725065/3"}}},"commit":{"parents":[{"commit":"aad8d248f82899943176bffbce928b0d2e62703d","subject":"Bump openstackdocstheme to 1.30.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/aad8d248f82899943176bffbce928b0d2e62703d"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-05-02 21:19:44.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-05-11 15:21:43.000000000","tz":120},"subject":"configuring neutron-api-uwsgi.ini","message":"configuring neutron-api-uwsgi.ini\n\nDebian uses uwsgi to run all of OpenStack API. This patch is a first\nattempt to allow puppet-openstack to configure it. This patch includes:\n- configuring of /etc/neutron/neutron-api-uwsgi.ini using a new provider\ncalled neutron_uwsgi_config\n- forward of $api_worker in neutron::server to neutron_uwsgi_config\n\nChange-Id: Id9d14c06abc12547ad2ca894b369baf864e6418b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/7b7043054e823188516ad3283b22670f848c24e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/7b7043054e823188516ad3283b22670f848c24e3"}]},"branch":"refs/heads/master"},"54ecd6caef165a12323b2c9213a3329d7e10edd4":{"kind":"REWORK","_number":4,"created":"2020-05-20 12:15:57.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/65/725065/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/65/725065/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/725065/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/725065/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/725065/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/65/725065/4"}}},"commit":{"parents":[{"commit":"aad8d248f82899943176bffbce928b0d2e62703d","subject":"Bump openstackdocstheme to 1.30.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/aad8d248f82899943176bffbce928b0d2e62703d"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-05-02 21:19:44.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-05-20 12:15:43.000000000","tz":120},"subject":"configuring neutron-api-uwsgi.ini","message":"configuring neutron-api-uwsgi.ini\n\nDebian uses uwsgi to run all of OpenStack API. This patch is a first\nattempt to allow puppet-openstack to configure it. This patch includes:\n- configuring of /etc/neutron/neutron-api-uwsgi.ini using a new provider\ncalled neutron_uwsgi_config\n- forward of $api_worker in neutron::server to neutron_uwsgi_config\n\nChange-Id: Id9d14c06abc12547ad2ca894b369baf864e6418b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/54ecd6caef165a12323b2c9213a3329d7e10edd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/54ecd6caef165a12323b2c9213a3329d7e10edd4"}]},"branch":"refs/heads/master"},"bdcfd1333bef73c875b8d5c2f8dedf2df0953f2f":{"kind":"REWORK","_number":5,"created":"2020-05-25 10:49:25.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/65/725065/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/65/725065/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/725065/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/725065/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/65/725065/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/65/725065/5"}}},"commit":{"parents":[{"commit":"aad8d248f82899943176bffbce928b0d2e62703d","subject":"Bump openstackdocstheme to 1.30.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/aad8d248f82899943176bffbce928b0d2e62703d"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-05-02 21:19:44.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-05-25 10:49:13.000000000","tz":120},"subject":"configuring neutron-api-uwsgi.ini","message":"configuring neutron-api-uwsgi.ini\n\nDebian uses uwsgi to run all of OpenStack API. This patch is a first\nattempt to allow puppet-openstack to configure it. This patch includes:\n- configuring of /etc/neutron/neutron-api-uwsgi.ini using a new provider\ncalled neutron_uwsgi_config\n- forward of $api_worker in neutron::server to neutron_uwsgi_config\n\nChange-Id: Id9d14c06abc12547ad2ca894b369baf864e6418b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/bdcfd1333bef73c875b8d5c2f8dedf2df0953f2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/bdcfd1333bef73c875b8d5c2f8dedf2df0953f2f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
