)]}'
{"id":"openstack%2Fpuppet-neutron~725832","triplet_id":"openstack%2Fpuppet-neutron~master~I4c982d6df9621e9ae385ed733336ec6c7410ec49","project":"openstack/puppet-neutron","branch":"master","topic":"plumgrid-removal","hashtags":[],"change_id":"I4c982d6df9621e9ae385ed733336ec6c7410ec49","subject":"Deprecate support for plumgrid plugin","status":"MERGED","created":"2020-05-06 11:44:23.000000000","updated":"2020-05-07 17:36:00.000000000","submitted":"2020-05-07 17:29:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725832-1588872553709-d95910ec","meta_rev_id":"43cf6f50c949a20c569da8c205bd4920bfd5ac09","_number":725832,"virtual_id_number":725832,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-07 17:29:13.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-05-07 14:17:53.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-05-07 14:40:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-05-07 14:40:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-07 14:17:53.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-05-07 14:40:02.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-05-07 17:29:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"838112d62a62b54c54c66a2964b6aa554122cd8e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-06 11:44:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"81fc3d5af04e2c5522a6e05fbae0d7117a6f5130","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 14:02:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f3d24f12a695439e81255b18fe2bbe96 : SUCCESS in 5m 18s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b8b0a5090b2a486c99d77d209e5d0841 : SUCCESS in 5m 24s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cb8e37cee059426ab2a6f976c08c6563 : SUCCESS in 4m 10s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/6b815dd9c96c4854a008739bcddfe28b : SUCCESS in 21m 22s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/d213f70b17424479bf30ebed03579682 : SUCCESS in 21m 02s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/883ececf2ae84f9f8841de87c8a3c576 : SUCCESS in 20m 03s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6fa731b520994bcbb5e7ee6d0dfe2b01 : FAILURE in 20m 53s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/99c93fb9c7384232809ece6107d59894 : SUCCESS in 44m 44s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e2ab8b90fbda4e54a3caf67989b8f937 : SUCCESS in 51m 00s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1c68b609f9a44403bdef91a888039e8b : SUCCESS in 42m 29s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/1cdfde2feb17412883045b0a4bd3731a : SUCCESS in 37m 33s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7398f074ec814557b38107a161b5aadb : SUCCESS in 52m 04s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/50ed5a276ce74d598951b192e77b0f2a : SUCCESS in 1h 13m 01s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/669308ec0b774b42991c435309688767 : SUCCESS in 1h 12m 50s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/715e436515994cb495599e7f1e81ad7d : SUCCESS in 50m 35s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/e2ebcdcbf0b3493da3da7082c18ec8b2 : SUCCESS in 20m 46s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/04c2184a8bd645aeab4e76569ff49388 : FAILURE in 9m 38s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/01e766971f03424587d0c7c9ffbefa8d : FAILURE in 9m 07s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7fb394983def4f4bba8e87483bd8ee1c : SUCCESS in 4m 08s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/eabeccfbea124717848cff7f6dee5492 : SUCCESS in 2h 14m 03s","accounts_in_message":[],"_revision_number":1},{"id":"fab22d459a70d8800bda507d117bc7ae3bac2dc0","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-05-07 14:17:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7bf5b3cd7e4b7036c2ab3b537b328e19f72fbac0","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-07 14:40:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"648c1c055eacdcc2472638124d3d76151c88a8a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 14:40:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"feba26c65562e82ea1d3d000e55e094be6cd34bd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 17:29:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d9b1a388243687eb7f7794ae51bef7ed5d65ba56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 17:29:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/38028545566346658a892add7394a2bc : SUCCESS in 6m 24s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/83c81737af794adda5edee1936a89320 : SUCCESS in 6m 47s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/db25876669bb48b3aaff9caaad1ece67 : SUCCESS in 7m 52s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/e5de1a46b74848a6896e3599b69a67b1 : SUCCESS in 22m 56s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/68fe67f3c65040dbbeeb9293e086511b : SUCCESS in 27m 56s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0ba30801ea52457296129779fd44c461 : SUCCESS in 22m 06s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a73ac23bd98c4aaea5312f769f27a085 : SUCCESS in 1h 00m 10s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c3dad4bca0644999a5f7b9073851268f : SUCCESS in 1h 07m 14s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/370546e5d5e4461ba51e5dce97f1030a : SUCCESS in 1h 13m 36s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ca4c99b13d384c60ac7c45cf48cce429 : SUCCESS in 50m 02s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/426eba1f6683401aa021b04c3516b775 : SUCCESS in 23m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a66ae37cc98249cea62dbc498b4464cf : SUCCESS in 5m 16s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5439a0f8ef4346d58d1c28dbb1d4e7e7 : SUCCESS in 2h 16m 04s","accounts_in_message":[],"_revision_number":1},{"id":"6d7ae8175ddbb55833650a66f8d739e4d8cc6020","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 17:36:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b3daf1ff180a4b289951f64c6a7dd793 : SUCCESS in 1m 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"24a5a9aed9bb8dd57adae1ae9315c9fb85e0b6f3","revisions":{"24a5a9aed9bb8dd57adae1ae9315c9fb85e0b6f3":{"kind":"REWORK","_number":1,"created":"2020-05-06 11:44:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/725832/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/32/725832/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/32/725832/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/32/725832/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/32/725832/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/32/725832/1"}}},"commit":{"parents":[{"commit":"aad8d248f82899943176bffbce928b0d2e62703d","subject":"Bump openstackdocstheme to 1.30.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/aad8d248f82899943176bffbce928b0d2e62703d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-06 11:42:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-06 11:42:48.000000000","tz":540},"subject":"Deprecate support for plumgrid plugin","message":"Deprecate support for plumgrid plugin\n\n... because networking-plumgrid[1] is no longer maintained.\n\n[1] https://opendev.org/x/networking-plumgrid\n\nChange-Id: I4c982d6df9621e9ae385ed733336ec6c7410ec49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/24a5a9aed9bb8dd57adae1ae9315c9fb85e0b6f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/24a5a9aed9bb8dd57adae1ae9315c9fb85e0b6f3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
