)]}'
{"id":"openstack%2Fpuppet-neutron~725845","triplet_id":"openstack%2Fpuppet-neutron~master~Ief13a16ae80deaeacdfa7ca9095a72216972c4f9","project":"openstack/puppet-neutron","branch":"master","topic":"arista-removal","hashtags":[],"change_id":"Ief13a16ae80deaeacdfa7ca9095a72216972c4f9","subject":"Deprecate support for Arista plugin","status":"MERGED","created":"2020-05-06 12:30:02.000000000","updated":"2020-05-07 17:33:44.000000000","submitted":"2020-05-07 17:29:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725845-1588872551184-a92907d4","meta_rev_id":"9669529cb09b2b64c752aa5486eb09b96491f07f","_number":725845,"virtual_id_number":725845,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2020-05-06 15:43:07.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-07 17:29:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-05-07 14:17:58.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-05-07 14:39:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2020-05-07 14:39:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-07 14:17:58.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-05-07 14:39:25.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-05-07 17:29:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"516713930c40ba4dfb5b711b12c074bb8c9051aa","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-06 12:30:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a55083f48306a2757162aeb0aeecd951c7387662","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 15:32:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a718ff207d8645048f4c4adc9554c7ed : SUCCESS in 4m 48s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/99fa0fd448ac4b81b69c381ba4c784ae : SUCCESS in 5m 08s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2987945268c04004b54b49530b4ce217 : SUCCESS in 5m 15s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/14372fa6c9cf49729070e8b70123750d : SUCCESS in 20m 09s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/16d11db8640244ba9c1591eef8b356dc : SUCCESS in 22m 23s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0a268a6e22b5497f913057b93053323d : SUCCESS in 20m 34s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a133abe296804991996c4f4be84f5a41 : FAILURE in 17m 49s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/dae980f447664f4fb4c246c5b941d4fa : SUCCESS in 51m 35s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9c7591cf026f4790a87397a8a978c330 : SUCCESS in 1h 02m 06s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d0a9217b6f2b4aa78df31743bb772f70 : SUCCESS in 53m 32s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/a615e4108bb1401b81746c844dbfc990 : SUCCESS in 41m 05s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/cb8c53881f0c4b3290b03e8ec39e7818 : SUCCESS in 55m 12s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3e6e543b9c714096acb1e360f9c45ec7 : SUCCESS in 1h 09m 38s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ac796c350d754e10afd27eb9fb08be16 : FAILURE in 1h 47m 08s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1e9912b7a617478f81994658bed3abf8 : SUCCESS in 1h 06m 05s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/f82052db61bb4fba81de82fc4ad6ed4b : SUCCESS in 25m 36s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/387fb3845f47430885266279402b9243 : FAILURE in 10m 59s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/926cce12975c4c498f5dce0f8af05325 : FAILURE in 8m 04s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9180d41e4664ba08eeabd4b05ddcaaf : SUCCESS in 5m 07s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d261d6d3e8624e43bb90bf5e6fc9e4b2 : SUCCESS in 2h 59m 41s","accounts_in_message":[],"_revision_number":1},{"id":"5234df44af0864ced18e0b8abe1018c779d81a1e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-06 15:43:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"993288ca5c84780c6c733e942389f5b795891e9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 18:20:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/06f885460e9b406fbceee1b881588a4c : SUCCESS in 6m 10s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/24cb44e771774c3bb4506fca375f870d : SUCCESS in 5m 11s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3270609ebb9f44708bfeeb72b71f2ae6 : SUCCESS in 5m 29s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/ee67d3041a1f4674b00dab20d47806f7 : SUCCESS in 21m 37s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/f65f3af4b85b4de09c84128e0d092be3 : SUCCESS in 24m 01s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dd46870b0055411cbdbff695e99df979 : SUCCESS in 26m 40s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c8df742be68a47c5833ccf247581db14 : FAILURE in 24m 40s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/19403dfbd0f3473ca0cc012a1c0767d5 : SUCCESS in 1h 01m 56s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/482b722315f0405fb60c597c04f7f1b0 : SUCCESS in 58m 49s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7a3730efe1e04485afa41a39a4f2b1e2 : SUCCESS in 50m 41s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/1e6a1508de7e4b31a01b310660f5c202 : SUCCESS in 52m 21s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d1e6b83ea449493f8120f8d9645445e5 : SUCCESS in 56m 04s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/253ba2518eb94457bbe35e085f50373b : SUCCESS in 1h 03m 12s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e0a6bca413b84cd2b96becb9d4291220 : SUCCESS in 1h 01m 42s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3e760ddced3341298d83893bfba3194c : SUCCESS in 48m 46s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/5e309bb23e1c4e1683f238e1e255522b : SUCCESS in 21m 44s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/578fef8a64f449d0bbe4cd23004244ad : FAILURE in 11m 11s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f52b48261ae1431ab031508ec8dabcb6 : FAILURE in 8m 27s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b644f8199fd545b1a136cd73aa0c7b70 : SUCCESS in 4m 20s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8f90dfc126fd4bef975059a4f214c248 : SUCCESS in 2h 30m 44s","accounts_in_message":[],"_revision_number":1},{"id":"4766135730dd996ddac5249e6e60b3f92952458a","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-05-07 14:17:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4ef4612ae90c372cfafada2d6eec7c5031d6b869","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-07 14:39:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9013601c3a3d17ccb27ab54d7480112e3c534f2e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 14:39:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a0c04bb7ce77cf66fe236b3767006bb8b48ff67f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 17:29:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c2f394646a814bd99ce89b6948dcb414 : SUCCESS in 6m 42s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a57af633bf534f8fb61a916071a06199 : SUCCESS in 6m 57s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9e932292f5db460d904a683f08772c56 : SUCCESS in 7m 15s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/2027422b55c446249b34c9855f804245 : SUCCESS in 25m 15s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/99b63271ef2f4a0e926a8ae951621fc1 : SUCCESS in 21m 34s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8e2d12f329f64d1784da0cd754eff7a4 : SUCCESS in 25m 16s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8125d53bffa742478e208aed55e28f69 : SUCCESS in 1h 12m 52s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6a3aa0731f674e8cbd65633926ee3e17 : SUCCESS in 1h 09m 44s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f3601dcbe6774ae7aba29177e04b2016 : SUCCESS in 1h 07m 06s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/90e6494416054969a82de26d98ece50a : SUCCESS in 48m 56s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/b2e3efe37526483688fba26405d5ff35 : SUCCESS in 19m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07fb874347b94075a6e5c3ed3bcc5c2c : SUCCESS in 4m 50s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/13d5dd7fb6a24cce83f6eec0fe6c6e65 : SUCCESS in 2h 06m 28s","accounts_in_message":[],"_revision_number":1},{"id":"444663889293322fa7f00fdc00e94f2483a09bd0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 17:29:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ff6928f1c2fe2b57c60e3c316691d05fbabb58fd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 17:33:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7fdc3521729404b9989ab4126572b5e : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5067451ddefe61902cf4b53734c040efe330da96","revisions":{"5067451ddefe61902cf4b53734c040efe330da96":{"kind":"REWORK","_number":1,"created":"2020-05-06 12:30:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/725845/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/45/725845/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/45/725845/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/45/725845/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/45/725845/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/45/725845/1"}}},"commit":{"parents":[{"commit":"aad8d248f82899943176bffbce928b0d2e62703d","subject":"Bump openstackdocstheme to 1.30.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/aad8d248f82899943176bffbce928b0d2e62703d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-06 12:28:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-06 12:29:48.000000000","tz":540},"subject":"Deprecate support for Arista plugin","message":"Deprecate support for Arista plugin\n\n... because networking-arista[1] is no longer maintained.\n\n[1] https://opendev.org/x/networking-arista/\n\nChange-Id: Ief13a16ae80deaeacdfa7ca9095a72216972c4f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/5067451ddefe61902cf4b53734c040efe330da96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/5067451ddefe61902cf4b53734c040efe330da96"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
