)]}'
{"id":"openstack%2Fpuppet-neutron~727783","triplet_id":"openstack%2Fpuppet-neutron~stable%2Fqueens~I0cf45085c6d9f278fffb6f55ea8d2159dd425746","project":"openstack/puppet-neutron","branch":"stable/queens","hashtags":[],"change_id":"I0cf45085c6d9f278fffb6f55ea8d2159dd425746","subject":"Add of_connect_timeout and of_request_timeout","status":"MERGED","created":"2020-05-13 14:31:49.000000000","updated":"2020-05-14 19:34:04.000000000","submitted":"2020-05-14 19:32:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"727783-1589484752160-8773121f","meta_rev_id":"083db26db6c48119c00c055413a3abc65828707f","_number":727783,"virtual_id_number":727783,"owner":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-14 19:32:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2020-05-14 07:08:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2020-05-13 17:50:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":2,"date":"2020-05-14 18:17:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":1,"date":"2020-05-14 18:17:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-13 17:50:09.000000000","updated_by":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"reviewer":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"state":"REVIEWER"},{"updated":"2020-05-14 07:08:43.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-05-14 18:17:48.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-05-14 19:32:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ea350ef23fb37328d8e404bf53486ad8b1262ffd","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-05-13 14:31:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b2768513870b3f43a1de077c1bbc88d7151640da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 15:49:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/7cc7ba8b4652478fa4e059134fd485de : SUCCESS in 5m 21s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/e68b132ef5a84202a1b4a386bf4ffe0e : SUCCESS in 5m 51s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/a555bb438d4441c9b92a154b72c592f1 : POST_FAILURE in 34m 56s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/3d9e0f6d80984908a503c23b0c1cd0a8 : SUCCESS in 31m 31s\n- puppet-openstack-integration-4-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/ddea63a77e4c4c51a38a9f91054b3ca9 : SUCCESS in 55m 01s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/8d87f31d0cb2430bab40eba2fb5c8f0f : SUCCESS in 1h 04m 44s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/66d764c55d2b4068a259a0a528e275cb : SUCCESS in 1h 02m 10s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/17c7235752474f569247a14bb8a9ce02 : SUCCESS in 46m 19s\n- puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial-luminous https://zuul.opendev.org/t/openstack/build/ad54ef9fc38247d4b00d5b63cda4d525 : FAILURE in 36m 43s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/c6e4a86559b942429617cf6ef00c7192 : FAILURE in 27m 48s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/0343f35c5b49498fa2d7bacd8377a12b : FAILURE in 24m 54s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial-luminous https://zuul.opendev.org/t/openstack/build/1f9a4ec4cf44430cafe54f1fa8f7f3bb : FAILURE in 23m 41s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/73273463f540420895307ceeeb5334b8 : SUCCESS in 18m 51s\n- puppet-openstack-beaker-xenial https://zuul.opendev.org/t/openstack/build/fcec89d20fb84078a21a4515fa6148da : SUCCESS in 19m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/163a6b7c88f34b51827096b81ce63284 : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":1},{"id":"b5169986fe7d6dfe0f0edae0ea4e095c50f3dc49","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2020-05-13 17:50:09.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fb02eed2483d9bcfd93dd44e55fde52e07373dc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 18:57:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/b592665710fd4b0dabf598b2a59926e2 : SUCCESS in 6m 27s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/e974fa8f6435407ba582b0ee13c311a8 : SUCCESS in 5m 39s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/9f5f3ccf1c1744b8825c2635895a87ee : SUCCESS in 31m 26s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/ecd8ce736c684fd785eec4bd6eb482d8 : SUCCESS in 28m 47s\n- puppet-openstack-integration-4-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/eb00575b5cd64c10940d2d87f26b2446 : SUCCESS in 59m 50s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/e28104d419064bfd8183635558ba0ba9 : SUCCESS in 58m 50s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/7893a6b9201d416296fd6d2556c236e4 : SUCCESS in 1h 01m 24s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/8887b5dc910347f5bcbeab00fa82b129 : SUCCESS in 46m 31s\n- puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial-luminous https://zuul.opendev.org/t/openstack/build/0698b37871d044b683f040498d6b3875 : FAILURE in 33m 43s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/3c0fa92779b449c39df2966503a53884 : FAILURE in 32m 30s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/726f361629f24f43924344a616166710 : FAILURE in 40m 03s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial-luminous https://zuul.opendev.org/t/openstack/build/853ca72999a940e88b5c9cf02b751970 : FAILURE in 34m 19s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/52208880dfb5477682196a9760a1f133 : SUCCESS in 25m 57s\n- puppet-openstack-beaker-xenial https://zuul.opendev.org/t/openstack/build/674ae51168944d77b73f08eaa576d4c8 : SUCCESS in 16m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42271210672947bdb4dc02ffe5bdb157 : SUCCESS in 7m 35s","accounts_in_message":[],"_revision_number":1},{"id":"b2b3210624657f712afb07df73d8a8c26b51f4c2","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-14 07:08:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2a18b06ad4cc749f54a5b830b9269fdf0f5875c1","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-14 18:17:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"edda3e9dd71c1b12aecad99d6b4caef526cb1d2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 18:18:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bb58975cb2faf78790b4b27d0cfcfcd00459b437","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 19:32:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/9780588a57864813b14e5d94c1d25156 : SUCCESS in 5m 01s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/a22ec346542b4e908f23f165eaaaa713 : SUCCESS in 5m 31s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/5c8eae0f50ee4b898c05f96832f60458 : SUCCESS in 25m 08s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/228f93cd58f64e7fbb6b2b8a5e315f9a : SUCCESS in 32m 44s\n- puppet-openstack-integration-4-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/4d5e0e5ed9684cf2917a985c363d2611 : SUCCESS in 1h 04m 19s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/e2294953dba54046be32dd11df1e6fe3 : SUCCESS in 57m 37s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/e7db6b73435344a488c8d4ec39387748 : SUCCESS in 1h 09m 04s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/827aa5c3676b42f192064efe49ef273e : SUCCESS in 44m 23s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/03f16ec9538740e68730ca7e8a1d9508 : SUCCESS in 18m 43s\n- puppet-openstack-beaker-xenial https://zuul.opendev.org/t/openstack/build/9f193b436fdd4f34845e1d3789edb82e : SUCCESS in 19m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2ff7ca46f76498d877626ed19b388ef : SUCCESS in 6m 39s","accounts_in_message":[],"_revision_number":1},{"id":"5353a4d3d61e25eadc109c2e3bddafd35d52adf0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 19:32:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"083db26db6c48119c00c055413a3abc65828707f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 19:34:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f5a85d5fe90141919e399e6d4bbefe0f : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f842cbf8d70f73699fb741bf6b19b4ea598b6bb3","revisions":{"f842cbf8d70f73699fb741bf6b19b4ea598b6bb3":{"kind":"REWORK","_number":1,"created":"2020-05-13 14:31:49.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/83/727783/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/83/727783/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/83/727783/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/83/727783/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/83/727783/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/83/727783/1"}}},"commit":{"parents":[{"commit":"7b19a78bc90167f34e968390c20857152cac9dee","subject":"Add OVS of_inactivity_probe parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/7b19a78bc90167f34e968390c20857152cac9dee"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-02-05 10:37:13.000000000","tz":60},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-05-13 14:31:07.000000000","tz":120},"subject":"Add of_connect_timeout and of_request_timeout","message":"Add of_connect_timeout and of_request_timeout\n\nAdds these two new parameters. [1]\n\n[1] https://github.com/openstack/neutron/blob/master/neutron/conf/plugins/ml2/drivers/ovs_conf.py#L119\n\nConflicts:\n\tspec/classes/neutron_agents_ml2_ovs_spec.rb\n\nChange-Id: I0cf45085c6d9f278fffb6f55ea8d2159dd425746\n(cherry picked from commit 95f442bd61b63a9a21fc999c3bb9ad0af4288ac5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/f842cbf8d70f73699fb741bf6b19b4ea598b6bb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/f842cbf8d70f73699fb741bf6b19b4ea598b6bb3"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
