)]}'
{"id":"openstack%2Fpuppet-neutron~728189","triplet_id":"openstack%2Fpuppet-neutron~stable%2Fussuri~I00ad87729e0b654c2fcb59b42374163a5674bfc9","project":"openstack/puppet-neutron","branch":"stable/ussuri","hashtags":[],"change_id":"I00ad87729e0b654c2fcb59b42374163a5674bfc9","subject":"Switch to Ussuri","status":"MERGED","created":"2020-05-14 18:03:11.000000000","updated":"2020-05-18 07:06:50.000000000","submitted":"2020-05-18 07:06:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"728189-1589785610939-7d7648ba","meta_rev_id":"a12d7fa5a4b081d5c0c03bf9bda0a180547eb23d","_number":728189,"virtual_id_number":728189,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-18 07:06:50.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2020-05-18 05:58:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2020-05-15 22:45:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2020-05-18 05:58:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-15 22:45:39.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-05-18 05:58:28.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2020-05-18 07:06:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"40f0db013105f406319001816c9a9bd5bc10b11a","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-14 18:03:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb00195400262285f002303508ea94f38ef5e6ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 19:26:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/98307ff336ee40d3b1bd66814b87e105 : SUCCESS in 5m 57s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/efbe63deefc94050b74a82d7eae9e905 : SUCCESS in 6m 34s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6411c66c76954dc1a85285be71655181 : SUCCESS in 7m 31s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/8864b6adcd164863a3d6dd2069c389fa : SUCCESS in 23m 13s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/9d6acb867c354fa3aef35f8fcddf67f9 : SUCCESS in 23m 20s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d4e5dd9fc5d849608a11b553e6e72d2b : SUCCESS in 22m 49s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7b5a9c003caa4d419869b686d0e7a007 : FAILURE in 21m 33s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/dcdc6f13a63b4ef2ba05d290b9a3d37b : SUCCESS in 45m 43s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d0ebc266b8a347a9a447b8f7d1736101 : SUCCESS in 49m 12s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ba48adf452aa4d68a17e2742653467bb : SUCCESS in 40m 51s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/d089f984f69e4dbe8f3945403d53af7f : SUCCESS in 49m 11s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5c502cc1a7084c689fc90366868550af : SUCCESS in 56m 42s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/bc1cd0b632e945eababcf67067c6fbe8 : SUCCESS in 1h 12m 07s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/73914e2cbe26404ea0b92540ed99a7e1 : SUCCESS in 1h 08m 33s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/89bb8a416f04427ea2bff3e00630c8e9 : SUCCESS in 46m 52s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/07edeb3a2d1e438eb99f7628eae371eb : SUCCESS in 21m 12s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/f87b7b4e340941b1b5f68e6832ae6326 : FAILURE in 11m 37s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f74fa79710cf4d18aaa0475663349daf : FAILURE in 9m 18s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/58337e1ed91f4f6da29bdbd20bd1c8ef : FAILURE in 30m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5e67ae724216cc5b54b5c0282f4fdfcdb32b6521","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-15 22:45:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"62773941c6f69f061d0e5acec1263045d9b0ec7e","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-05-18 05:58:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"96cea6de346e18d696e40f5dee876930550bff61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 05:58:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0761e8dc89f18d4b8d2c1c39f61b6239da73f7c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 07:06:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a12d7fa5a4b081d5c0c03bf9bda0a180547eb23d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 07:06:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8da1a52106da4846ada2a7fe3759bccc : SUCCESS in 6m 22s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4bb9bd3890d643a58bdfeb52b23f13f1 : SUCCESS in 7m 43s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/47a9b276cfd14cc3add7f8aa1ce796a3 : SUCCESS in 6m 46s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/8e11e20f65e14af2b0aaab65ce6eb7ee : SUCCESS in 23m 49s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/eddbf9db953c4ac2a161ed8cc0e59e73 : SUCCESS in 21m 15s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d99e730eebb743c78d49ee600c25370e : SUCCESS in 18m 47s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3423b8778f5f4d1aaad38ac5e51ca197 : SUCCESS in 58m 25s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1e680410c0724df5853e5a1687b18b19 : SUCCESS in 1h 02m 45s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e5a5a836ec6e4fe3b86615bbe1108beb : SUCCESS in 1h 06m 50s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c959f580de994208b4b8fbc362fadb8a : SUCCESS in 48m 23s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/c0ee9480638840efbaca53e2df2de765 : SUCCESS in 25m 24s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c7297ab0ae734c6ea8004f5e8eebf823 : FAILURE in 31m 12s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1d09d25149cf3eaf2491225bb42febead4c0eaf0","revisions":{"1d09d25149cf3eaf2491225bb42febead4c0eaf0":{"kind":"REWORK","_number":1,"created":"2020-05-14 18:03:11.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/89/728189/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/89/728189/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/89/728189/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/89/728189/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/89/728189/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/89/728189/1"}}},"commit":{"parents":[{"commit":"855c7dacee1881413b087823b9df8ce8941ab1f1","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/855c7dacee1881413b087823b9df8ce8941ab1f1"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-05-14 18:03:02.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-05-14 18:03:02.000000000","tz":120},"subject":"Switch to Ussuri","message":"Switch to Ussuri\n\nChange-Id: I00ad87729e0b654c2fcb59b42374163a5674bfc9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/1d09d25149cf3eaf2491225bb42febead4c0eaf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/1d09d25149cf3eaf2491225bb42febead4c0eaf0"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
