)]}'
{"id":"openstack%2Fpuppet-neutron~736757","triplet_id":"openstack%2Fpuppet-neutron~master~I58b79942dc49d27e3ef6a35f3970fc392849652e","project":"openstack/puppet-neutron","branch":"master","topic":"retire-neutron-fwaas","hashtags":[],"change_id":"I58b79942dc49d27e3ef6a35f3970fc392849652e","subject":"Deprecated neutron fwaas","status":"MERGED","created":"2020-06-18 15:44:35.000000000","updated":"2020-06-22 16:02:19.000000000","submitted":"2020-06-22 16:00:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"736757-1592841609473-4f349c0a","meta_rev_id":"2979c4e0293b89b2ca130c7567f0cb2882ade4de","_number":736757,"virtual_id_number":736757,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"date":"2020-06-20 10:44:55.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-22 16:00:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-06-22 12:43:57.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-06-21 06:40:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-06-22 12:43:57.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-21 06:40:19.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-06-22 12:43:57.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-06-22 16:00:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b7e387842f16b9e88cb98e7543955a6289fb6108","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-06-18 15:44:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"72f20ebf94e4ec3b4d930b374b75ef15f6e94d43","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-06-18 15:44:58.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"0bddaf3af8b0dc5714d86a17ca77359189338724","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 17:12:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7c713308caf245ed8caecb4c17dbce21 : SUCCESS in 6m 03s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/01816e4c4a0649bab5b5f6fbea544e64 : SUCCESS in 5m 47s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f31084c438094611b85313e9491bac7f : SUCCESS in 7m 36s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/c37437825d1b4bd9b1c8d9a8bb4cd1d8 : SUCCESS in 24m 27s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/df58c4c2ea77430e89a4aaefb94da860 : SUCCESS in 26m 01s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d034c6889baa42a990093edadc0deb7f : SUCCESS in 21m 14s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f7bc7b52305442bf97e9f18e65de8d25 : FAILURE in 24m 11s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/4c858f9bcab148d6aa6e318ee0140ce3 : SUCCESS in 49m 24s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ef5e57b286b34a2aab374b3f91c9864e : SUCCESS in 51m 18s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bc9bed6ea3dc4d1293c2d10eacf04e74 : SUCCESS in 47m 06s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/e0246770ea134d3fa370bbeb49918114 : SUCCESS in 39m 14s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/bebe62ae95664305ba4d68b029a82d0a : SUCCESS in 1h 02m 03s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7ee99b23069942dfb7fb50c2e7f7f833 : SUCCESS in 1h 02m 32s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b73ec2596da44920b29960d6528619b0 : SUCCESS in 1h 08m 24s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8a1ba6f6788b49158de6ce0bddb57620 : SUCCESS in 47m 25s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/a89a0ce264be452581842ca54dbe30e9 : SUCCESS in 22m 54s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/f61179e641de4fd5be27b721db0088f5 : FAILURE in 12m 17s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/636a1116459f4f82a3244ad06954b78b : FAILURE in 8m 32s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e712aada0de249dab1d82c7e491542a2 : SUCCESS in 5m 33s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ab9262119c98446082e2570b1a63bab8 : FAILURE in 1h 20m 18s","accounts_in_message":[],"_revision_number":2},{"id":"0bc2298678816b55e52438387d8f5b300e5d129f","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-06-20 10:44:55.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2555d45c8f2c9738b302206016b935365e57d9dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-20 14:22:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5365a40e40804a0bbd79453d48f19cc9 : SUCCESS in 4m 49s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4b74c6fecaaa4b5fbdc322a597eb10f8 : SUCCESS in 4m 54s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/36d00b68176e4a3da6629613d72dee63 : SUCCESS in 4m 32s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/3d50018232ee4dbd8b30b7ba3172c10d : SUCCESS in 23m 22s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/a6f0c4267b964eaba6f4c0ed5a6c8986 : SUCCESS in 19m 43s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2570f67610444703aef19cf1da0d5c81 : SUCCESS in 21m 19s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2572fbd387f04ac29697e38f4976e15a : FAILURE in 20m 15s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/e39c305a17924399a1f7d11ffe4f91fa : SUCCESS in 43m 42s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/25ef997fa4f0413f9a5ab90ed7d62d02 : SUCCESS in 51m 06s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a802357a24d24b47a0f7c20df6dc3124 : SUCCESS in 51m 00s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/af6a7a1935cc4be2b1f3dc28a55d9d01 : SUCCESS in 33m 15s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e574cbc699cc4108a40285241d1acc3f : SUCCESS in 57m 20s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/acef43b448dd45dfa9815e4afc65ce15 : SUCCESS in 43m 46s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/28d4ef98aafa41f68b60950c943563eb : SUCCESS in 42m 11s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/af37a6b0f9fa4af4b056aed2b16c9961 : SUCCESS in 40m 59s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/cd031153f0384081b477f4a200d9a9bf : SUCCESS in 18m 58s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/e0398436b70f46fdbe400db427aaf0b0 : FAILURE in 7m 50s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a9493aa1ffbc41ca909a9cc28a272ad1 : FAILURE in 6m 27s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f4c7d2ef39b7464a8bf135a19853e8f7 : SUCCESS in 4m 44s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e5e261cd15e64d1eab6cd800da1fd085 : SUCCESS in 2h 16m 53s","accounts_in_message":[],"_revision_number":2},{"id":"c53ea36cff4bf3fe3294f08d5019d6f1526ae676","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-06-21 06:40:19.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2ea487b19aed36ec353f1493c263830580783396","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-06-22 12:43:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5f113f01eaca906116e3f5243f3fac4d9a5d3a78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 12:44:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"394faa3857fde2232c7ed4a57be98adf3f041d27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 16:00:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bb70c610287f4afd89f90c72e58b5f31 : SUCCESS in 6m 09s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a41e7ac06c2642dd8d878fc17eb0d552 : SUCCESS in 5m 44s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5fea20d09b824d7496923f491d29b9e7 : SUCCESS in 6m 40s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/7f4bad4318de4310bd55c05f2ff704a2 : SUCCESS in 21m 56s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/cbeaa6456da94276b5a972d845618b4d : SUCCESS in 24m 14s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/11cb861cf2184654a73d78b1029c5c9f : SUCCESS in 19m 29s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e4f73bf8af1348d98914658e35318852 : SUCCESS in 59m 21s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/81e582dae8e044c0ba9906df4eab00d4 : SUCCESS in 1h 09m 14s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e00a5572762d410d9acbb96b850e9da4 : SUCCESS in 1h 14m 01s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/404e994b0ea24ef99ea26c459fcc41fe : SUCCESS in 45m 31s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/f33c2c439d5f403c979ba0f14ca333a4 : SUCCESS in 21m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35c651e1f9fe4d439cb6b6111a393b26 : SUCCESS in 5m 08s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4e8779d3e1e94ba5b0d1f54708e739d4 : SUCCESS in 2h 27m 49s","accounts_in_message":[],"_revision_number":2},{"id":"88d0b7e91760622c8696e268f8377aa639d4e537","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 16:00:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"71ea97b6078854e2dafda5cacfa9d25615e9f033","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 16:02:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/685810fec6ba4500b40f7b34ed6855a0 : SUCCESS in 1m 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"80f2b44bcb576a2eefca8639496bc04ec6fe3265","revisions":{"1f9b0cb17365628b5880456975d63847a92880e2":{"kind":"REWORK","_number":1,"created":"2020-06-18 15:44:35.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/57/736757/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/57/736757/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/57/736757/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/57/736757/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/57/736757/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/57/736757/1"}}},"commit":{"parents":[{"commit":"bd0fe20af9a969dac50b75bb35916d476f95d3dd","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/bd0fe20af9a969dac50b75bb35916d476f95d3dd"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-06-18 15:43:40.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-06-18 15:43:40.000000000","tz":120},"subject":"Deprecated neutron fwaas","message":"Deprecated neutron fwaas\n\nThe neutron fwaas project is being retired [1], this deprecates\nall related code so it can be removed next release.\n\n[1] https://review.opendev.org/#/c/735829/\n\nChange-Id: I58b79942dc49d27e3ef6a35f3970fc392849652e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/1f9b0cb17365628b5880456975d63847a92880e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/1f9b0cb17365628b5880456975d63847a92880e2"}]},"branch":"refs/heads/master"},"80f2b44bcb576a2eefca8639496bc04ec6fe3265":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-06-18 15:44:58.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/57/736757/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/57/736757/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/57/736757/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/57/736757/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/57/736757/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/57/736757/2"}}},"commit":{"parents":[{"commit":"bd0fe20af9a969dac50b75bb35916d476f95d3dd","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/bd0fe20af9a969dac50b75bb35916d476f95d3dd"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-06-18 15:43:40.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-06-18 15:44:56.000000000","tz":0},"subject":"Deprecated neutron fwaas","message":"Deprecated neutron fwaas\n\nThe neutron fwaas project is being retired [1], this deprecates\nall related code so it can be removed next release.\n\n[1] https://review.opendev.org/#/c/735829/\n\nDepends-On: https://review.opendev.org/#/c/736722/\nChange-Id: I58b79942dc49d27e3ef6a35f3970fc392849652e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/80f2b44bcb576a2eefca8639496bc04ec6fe3265"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/80f2b44bcb576a2eefca8639496bc04ec6fe3265"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
