)]}'
{"id":"openstack%2Fpuppet-neutron~739657","triplet_id":"openstack%2Fpuppet-neutron~master~Id6e609d0a556e23c484f38f4651d2709d898a76c","project":"openstack/puppet-neutron","branch":"master","hashtags":[],"change_id":"Id6e609d0a556e23c484f38f4651d2709d898a76c","subject":"Nuage: remove check on mechanism driver","status":"MERGED","created":"2020-07-07 05:09:19.000000000","updated":"2020-07-09 05:02:28.000000000","submitted":"2020-07-08 23:31:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"739657-1594251117764-86bf99d6","meta_rev_id":"58cde7db7ea87d9b6e8236d9afead6210e8918d9","_number":739657,"virtual_id_number":739657,"owner":{"_account_id":13988,"name":"Vlad Gridin","email":"vladyslav.gridin@nokia.com","username":"gridinv"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2020-07-08 18:26:49.000000000","_account_id":13988,"name":"Vlad Gridin","email":"vladyslav.gridin@nokia.com","username":"gridinv"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-08 23:31:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27069,"name":"Sai Ram Peesapati","email":"peesapatisairam@gmail.com","username":"sairamp"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-07-08 14:56:48.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":13988,"name":"Vlad Gridin","email":"vladyslav.gridin@nokia.com","username":"gridinv"},{"value":2,"date":"2020-07-08 12:14:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-07-08 15:26:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-08 14:19:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27069,"name":"Sai Ram Peesapati","email":"peesapatisairam@gmail.com","username":"sairamp"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-07-08 14:56:48.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":13988,"name":"Vlad Gridin","email":"vladyslav.gridin@nokia.com","username":"gridinv"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27069,"name":"Sai Ram Peesapati","email":"peesapatisairam@gmail.com","username":"sairamp"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":13988,"name":"Vlad Gridin","email":"vladyslav.gridin@nokia.com","username":"gridinv"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27069,"name":"Sai Ram Peesapati","email":"peesapatisairam@gmail.com","username":"sairamp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-08 12:14:48.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-07-08 14:19:02.000000000","updated_by":{"_account_id":27069,"name":"Sai Ram Peesapati","email":"peesapatisairam@gmail.com","username":"sairamp"},"reviewer":{"_account_id":27069,"name":"Sai Ram Peesapati","email":"peesapatisairam@gmail.com","username":"sairamp"},"state":"REVIEWER"},{"updated":"2020-07-08 14:56:48.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-07-08 15:26:08.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-07-08 23:31:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6d5e291beb672440748c475d1eab61d81312c6ed","author":{"_account_id":13988,"name":"Vlad Gridin","email":"vladyslav.gridin@nokia.com","username":"gridinv"},"date":"2020-07-07 05:09:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a7a15ba10e26d8ccf690438cf6c100941045adcd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 07:39:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b79ed0d63cdc4dfa90ae320eab86274d : SUCCESS in 4m 37s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6c3ea60580724fe98fcc633bba2fbdb3 : SUCCESS in 4m 15s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/83ddee2f3bca4a4490897fd8c83c85e6 : SUCCESS in 4m 03s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/fc39ecb8045f4ea78b278b8920690712 : FAILURE in 17m 43s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/052411df93414285a4bdd2c030b16b1e : FAILURE in 19m 36s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/647714514bfb493db516be9a6eb72654 : FAILURE in 13m 47s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9373765a6caa47f7bb183c5380b74b91 : FAILURE in 18m 07s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/644d6b24e61342079c94bd64468b3795 : SUCCESS in 53m 39s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/75a307195e60462d95a2e6e5e83a21e8 : SUCCESS in 58m 49s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/13cacaa99e89490c8dd042cf779fc988 : SUCCESS in 48m 30s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/ed8fb5caa8fc410abcffb2c5c5f8955c : SUCCESS in 46m 41s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a76bf1534bf24239acc5483ae0f43827 : SUCCESS in 39m 44s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/728958e801c1440da4f3b7f841a3e0aa : SUCCESS in 1h 05m 54s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f93fcf00f7a94eb49e7f7096c4349a8f : SUCCESS in 1h 08m 46s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7c7de779107348bb82fc3f7fbae97e68 : SUCCESS in 46m 10s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/c9fce4bf86534e36b2e44d0541ea9345 : SUCCESS in 23m 50s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/3460d31f1fc14014bff52a19b608d531 : FAILURE in 9m 12s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b4adfb754d854f28bb7fc95f0fab47e0 : FAILURE in 6m 29s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/bc649cbb4fda41a7a4cc0b7f8c6a0f4f : SUCCESS in 2h 24m 38s","accounts_in_message":[],"_revision_number":1},{"id":"d699f4752c2e48a80bdbd09e8159f736f67120db","author":{"_account_id":13988,"name":"Vlad Gridin","email":"vladyslav.gridin@nokia.com","username":"gridinv"},"date":"2020-07-07 09:41:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"03b58c83cc99350f22484f31458634a7393d2097","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 11:16:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/57fb34e3ccc1467aaa78a54bafb8dc09 : SUCCESS in 6m 52s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/68b60dca1c8c43ca8c53b0662f132ef4 : SUCCESS in 5m 38s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/67f28e71fb324228a1a3f08fa8dc8746 : SUCCESS in 6m 02s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/4b02c907980647eb8044349c6c87d822 : FAILURE in 20m 42s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/3bf007883d694aca95d8957fc08a3730 : FAILURE in 17m 10s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/681ee862f0f2410d84dbadb437731b8b : FAILURE in 22m 11s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1d53b88b572f4efaa60c4fe38e50df3e : FAILURE in 20m 57s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/8fcd5913793141c08c7fcab14722e966 : SUCCESS in 46m 48s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7facc0befa244f7cae8dc8db3a43dd2b : SUCCESS in 55m 10s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1966fa30a144459bb02d3ab6c2303d36 : SUCCESS in 49m 23s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/2af41ad2e5874109bb8ad7dc5f2a3542 : SUCCESS in 42m 45s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ecaf381a1bb444e5ab393431ea717c18 : SUCCESS in 58m 36s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/564cbc480aed4fecb2b8a75f5c309dcd : SUCCESS in 1h 10m 14s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/63647abb01d94a64b3c53e10b031813e : SUCCESS in 1h 10m 31s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/72ca70dd78a04e3fbe5784815ca79d92 : SUCCESS in 50m 02s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/86e4405e173a44e79558d0df27131d17 : SUCCESS in 24m 39s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/86382d297bd64fe5ba39e15f65bd8571 : FAILURE in 10m 18s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c4f9c8fae3994d55b91b65305151b53b : FAILURE in 9m 03s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f612262a1fc74dc0be628298baa68163 : SUCCESS in 1h 29m 58s","accounts_in_message":[],"_revision_number":2},{"id":"ae87e99fe85732984a3adb74fa25d526db979026","author":{"_account_id":13988,"name":"Vlad Gridin","email":"vladyslav.gridin@nokia.com","username":"gridinv"},"date":"2020-07-07 12:51:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"758cc16346df85b8a0287a5b6a5191736c026d0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 15:20:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d75ca160d14e4ddd87accbb61c3b65ae : SUCCESS in 5m 39s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8c7feda8ed8945cf951a6b423feafad8 : SUCCESS in 5m 45s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/56226b318eec4ea59da4ddbde7d71fe7 : SUCCESS in 5m 55s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/5bba47ec95d34ad2a91680da53789995 : SUCCESS in 18m 32s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/b93091ee96964a359ba36bac19520116 : SUCCESS in 23m 41s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3698c99d51584b328ecc2b0fa0d17bcc : SUCCESS in 21m 49s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d8fa1090521b4348bac946318cc73ec9 : FAILURE in 20m 34s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/81ad048ca9234b11a0d183108fc55869 : SUCCESS in 59m 57s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6528e0c6a639497c98375d5d724f27dd : SUCCESS in 55m 03s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3e618254049b48928b2eff3a4ad5cb89 : SUCCESS in 47m 41s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/4ee5ae2dfeca4539b1b4b1f342f5ca85 : SUCCESS in 42m 26s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f90a43a1433d4065a254c84a8c863b6f : SUCCESS in 57m 31s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c6ca59270f9441a8beb2e64b3767673b : SUCCESS in 1h 10m 14s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/752cb59b264c44e9a6ad2f8d4a9ebf73 : SUCCESS in 1h 12m 29s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6adb0f88e0764748a234b523bd066ec7 : SUCCESS in 51m 12s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/5119f008d7ef46ba9744660d91c67252 : SUCCESS in 20m 41s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/5abb0e45351c42fab4d4faa83b2174b8 : FAILURE in 14m 02s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/779b91394ffc4ce2947e10506b3c3721 : FAILURE in 8m 41s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/239d544e98e741eb988fc968b32a5a60 : SUCCESS in 2h 22m 06s","accounts_in_message":[],"_revision_number":3},{"id":"9ec707490f949722859a468f63f40706ba288dff","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-07-07 23:39:55.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"aecb9b1d30c6fb11539f5bb4ec6c1a9cfdeb66ee","author":{"_account_id":13988,"name":"Vlad Gridin","email":"vladyslav.gridin@nokia.com","username":"gridinv"},"date":"2020-07-08 06:04:34.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"38da66ccfedfd2f928b8ec889be532e516a0adf4","author":{"_account_id":13988,"name":"Vlad Gridin","email":"vladyslav.gridin@nokia.com","username":"gridinv"},"date":"2020-07-08 07:39:46.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4aecb18161184f8c426316fc7b8a89749b96264e","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-07-08 07:52:16.000000000","message":"Patch Set 4:\n\nI don\u0027t mind either way, actually personally I\u0027d rather see the check removed since it provides minimal help and restricts the ability to declare the classes in any order.","accounts_in_message":[],"_revision_number":4},{"id":"e985f187c8bbe4bfec62ee82367f10e08e4f467a","author":{"_account_id":13988,"name":"Vlad Gridin","email":"vladyslav.gridin@nokia.com","username":"gridinv"},"date":"2020-07-08 09:32:39.000000000","message":"Patch Set 4:\n\nHonestly, i don\u0027t see a great added value in validation either,\nbut if others feel it should be there, i am fine with it.","accounts_in_message":[],"_revision_number":4},{"id":"57f38917d8d2b664f0ce80b983344a8bcfdf9d1e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-07-08 09:44:59.000000000","message":"Patch Set 4:\n\nThe initial motivation why I asked the question was that I expected that the current validation logic could have accepted all \u0027nuage_\u003cfoo\u003e\u0027 mechanism drivers. However I noticed that it doesn\u0027t actually because it should assert the whole string.\n\nI agree that we can just remove that validation because it doesn\u0027t look very valuable. I confirmed that such validation is implemented  in only nuage manifests, so from perspective of standardization removing that validation looks better.\n\nSorry for making some confusion :-(","accounts_in_message":[],"_revision_number":4},{"id":"d0c19f23c5381581f3a2ea0e576502f7ea1cbb0c","author":{"_account_id":13988,"name":"Vlad Gridin","email":"vladyslav.gridin@nokia.com","username":"gridinv"},"date":"2020-07-08 10:04:49.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a93ed1759210af7515c5e03e3f3d5be35b33da88","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-07-08 12:14:48.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"351cb0278b1e3aff775ffc397f3bd3ff79e0bf8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 12:23:25.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/61fc03e4d00a41708ced3bb91d65b7f4 : SUCCESS in 6m 00s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/98450eee44ba42caa45cf9d8cef314a3 : SUCCESS in 7m 15s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0f63cc42aac44465a6b8338287d81d0f : SUCCESS in 6m 03s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/c9748180b277463c9f1781b0e327be09 : SUCCESS in 23m 19s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/cc7dca148e9d4e4494b5ce280f992045 : SUCCESS in 19m 56s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/20e6f4c1f2e7412c9647096c0f62e4d8 : SUCCESS in 16m 40s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6987acc77d37463480a71760b8ac75c0 : FAILURE in 21m 04s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/7bcb9d39e39146d4900c75c9ce0d0062 : FAILURE in 26m 10s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a39e66cdcedc413d86f3fe0c410fb9a4 : FAILURE in 56m 58s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9af4ca0868aa42fbaf456321b47cd5de : SUCCESS in 53m 19s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/27d0cd8cfbea4787a5f0b4997dc436ce : SUCCESS in 37m 36s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1ea6d003818a47a6b20369c96a7aca48 : SUCCESS in 1h 00m 09s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ff11368d47324a95bb088307b1de0ed6 : SUCCESS in 1h 05m 13s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/67f8438e9abd47498f8f392464ae97d0 : SUCCESS in 1h 25m 14s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d2655cb49e2446e187771bce3a4fdaa8 : SUCCESS in 43m 43s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/726fe01d169740fca8c670ad82d6ce3d : SUCCESS in 22m 22s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/033ec97dba5443f2a5ea46b08c7e2368 : FAILURE in 11m 47s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6c37133a68ca4cb8ab5e6151b0425b99 : FAILURE in 8m 42s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a9292fec7baa498db5417e07786d2a0a : SUCCESS in 2h 12m 16s","accounts_in_message":[],"_revision_number":5},{"id":"30c3d044c3e3493e552d61d2bc188f2d50090f6a","author":{"_account_id":27069,"name":"Sai Ram Peesapati","email":"peesapatisairam@gmail.com","username":"sairamp"},"date":"2020-07-08 14:19:02.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"249bde6cff3b1f6f62745c5619824b5cb85b53c2","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-07-08 14:56:48.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"c611d7af08eaa28f36165af7307bde8751865a90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 15:00:50.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"32faab83195baf95685d0a899ea36113786be166","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-07-08 15:26:08.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"56f1c40351e7d5c7b7f64b90ad48cae3194e7da7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 17:25:40.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8a300c91e9a44c57a3e043c3a03fc348 : SUCCESS in 7m 19s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1dd2aac4a91349978b409272d2a92fc5 : SUCCESS in 7m 55s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3b4648fcf8e34f7292c0e65253ca21cb : SUCCESS in 5m 30s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/719ded33c9d54d73a1fb74bfabfaade9 : SUCCESS in 28m 40s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/4869ebaedec74dd49cbd55c764643589 : SUCCESS in 23m 43s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/928ccc7a4d0a4931a8a8d3fb35f96097 : SUCCESS in 20m 10s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/09930203e1d74b5dbdfb853635685aaa : SUCCESS in 57m 52s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5379632d568244a6a077333b2ded61eb : FAILURE in 1h 04m 51s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6709325d46ee4564ad6eb51728ab26c4 : SUCCESS in 1h 11m 06s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/69cd36765a834882b61566e81204789a : SUCCESS in 46m 14s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/bfc85d5f66a94681a0fcff3ec4429e85 : SUCCESS in 22m 01s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/18d646ca41b7452ebd3d3aa9c13978c9 : SUCCESS in 2h 17m 45s","accounts_in_message":[],"_revision_number":5},{"id":"69c9d587fcfb33510afcbb7ddb97ecea082684fa","author":{"_account_id":13988,"name":"Vlad Gridin","email":"vladyslav.gridin@nokia.com","username":"gridinv"},"date":"2020-07-08 18:26:49.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"d1bcad4f1aef78efb2d95db8fd58356abdb77a95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 20:47:42.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/15d03d10ddd34fd6a59dc49142e4627a : SUCCESS in 6m 42s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/acccaf8743a441d5aaeb05d32bad54d9 : SUCCESS in 4m 30s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/19d67926f144452abe76500c6deccf11 : SUCCESS in 5m 19s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/a1680c2e39b54d078e5d3117e6ff3ab1 : SUCCESS in 25m 25s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/20e68a460fbf4e7badf6c0d40ffd5932 : SUCCESS in 22m 56s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a46d87290c114c139262e3b3043d80c2 : SUCCESS in 20m 18s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e279f177249f483ebac13ce2db357719 : FAILURE in 22m 24s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/1d39df511fc04b04bfac72643abc0da7 : SUCCESS in 54m 41s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d0ed375b5fc2426c81340109a66a654e : FAILURE in 53m 57s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/daba3f184cfb4fc5b655ac281aaf1bb4 : SUCCESS in 1h 11m 05s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/14211c4d9b674b368ab5d439063da8f8 : SUCCESS in 40m 27s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6b8a71b17dca4e2c93305066b18c6f71 : SUCCESS in 55m 30s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b17d4440621747de8bb7ff17b3a860dd : SUCCESS in 1h 04m 20s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/31ad70925700472cbc76712c248eb8bc : SUCCESS in 1h 04m 41s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/44daae4aed224cdfa9ad331281f92343 : SUCCESS in 50m 41s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/86fde87a4d9a4ba99862f1cb411b3dc1 : SUCCESS in 15m 48s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/34b716b8359e4775b8e5530745586c74 : FAILURE in 11m 18s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8f319aee028244faaf05316985a6a94d : FAILURE in 7m 07s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8f55e000b9174eb6bddb79b4824f9e5a : SUCCESS in 2h 18m 58s","accounts_in_message":[],"_revision_number":5},{"id":"2c9d86b5fe78c3db9d096c100800b8fe94e66a53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 20:47:57.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"0c4b0e4c384685b9084c7da4c45a8fc6a6a26050","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 23:31:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"601a1f3471557ce04f226ae39eca0b2b28497fd5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 23:31:57.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/58b42bcedd6343a58b253b620604621b : SUCCESS in 5m 42s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/43ac9dc8e16b466e86dc02a672236f9c : SUCCESS in 8m 06s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/58f1202687c34e4bbbfef5367a82eebc : SUCCESS in 5m 06s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/c5e846d23ddc4103be7a6b2ea836a568 : SUCCESS in 19m 34s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/14b8797d55734417898470fe47e41606 : SUCCESS in 20m 11s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/67373d60316f483a9ef00f38c0056fbe : SUCCESS in 28m 08s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/31af933292d14d3e890e2e0432b768e4 : SUCCESS in 57m 22s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b202a90cd43e454c906e1c40f71758ea : SUCCESS in 49m 45s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/dea7b5c8c2304f7980a96e8708c22044 : SUCCESS in 49m 56s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/21e040f3e73c4041b42fd844cb499af9 : SUCCESS in 45m 17s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/4c9202dd9a724e8283086df119923b4f : SUCCESS in 27m 30s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a8f35d47049e4e2ea98cb6914952a5e2 : SUCCESS in 2h 28m 23s","accounts_in_message":[],"_revision_number":5},{"id":"58cde7db7ea87d9b6e8236d9afead6210e8918d9","author":{"_account_id":13988,"name":"Vlad Gridin","email":"vladyslav.gridin@nokia.com","username":"gridinv"},"date":"2020-07-09 05:02:28.000000000","message":"Patch Set 5: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 7d6722d1ee31ce7876f9510c87d3fa98ad01d7f3","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"9fc323f48a466dfb5bc64c048a67ce878599ff2f","revisions":{"d472f5c54815bb1599f3bafc5cafd113447cb8ae":{"kind":"REWORK","_number":1,"created":"2020-07-07 05:09:19.000000000","uploader":{"_account_id":13988,"name":"Vlad Gridin","email":"vladyslav.gridin@nokia.com","username":"gridinv"},"ref":"refs/changes/57/739657/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/57/739657/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/57/739657/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/57/739657/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/57/739657/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/57/739657/1"}}},"commit":{"parents":[{"commit":"de2be7503caff6bb1bd2ab3865f2da8913fa0c5b","subject":"Merge \"Deprecated neutron fwaas\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/de2be7503caff6bb1bd2ab3865f2da8913fa0c5b"}]}],"author":{"name":"Vlad Gridin","email":"vladyslav.gridin@nokia.com","date":"2020-07-07 05:02:08.000000000","tz":120},"committer":{"name":"Vlad Gridin","email":"vladyslav.gridin@nokia.com","date":"2020-07-07 05:06:49.000000000","tz":120},"subject":"Nuage: remove check on mechanism driver","message":"Nuage: remove check on mechanism driver\n\nNuage neutron integration has a collection of mechanism\ndrivers sharing same configuration. This patchset removes check\non nuage ml2 mechanism driver which is not mandatory anymore.\n\nChange-Id: Id6e609d0a556e23c484f38f4651d2709d898a76c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/d472f5c54815bb1599f3bafc5cafd113447cb8ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/d472f5c54815bb1599f3bafc5cafd113447cb8ae"}]},"branch":"refs/heads/master"},"179bbd63ad23175d07858cacd9fc46dca0ed74f7":{"kind":"REWORK","_number":2,"created":"2020-07-07 09:41:48.000000000","uploader":{"_account_id":13988,"name":"Vlad Gridin","email":"vladyslav.gridin@nokia.com","username":"gridinv"},"ref":"refs/changes/57/739657/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/57/739657/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/57/739657/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/57/739657/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/57/739657/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/57/739657/2"}}},"commit":{"parents":[{"commit":"de2be7503caff6bb1bd2ab3865f2da8913fa0c5b","subject":"Merge \"Deprecated neutron fwaas\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/de2be7503caff6bb1bd2ab3865f2da8913fa0c5b"}]}],"author":{"name":"Vlad Gridin","email":"vladyslav.gridin@nokia.com","date":"2020-07-07 05:02:08.000000000","tz":120},"committer":{"name":"Vlad Gridin","email":"vladyslav.gridin@nokia.com","date":"2020-07-07 09:41:07.000000000","tz":120},"subject":"Nuage: remove check on mechanism driver","message":"Nuage: remove check on mechanism driver\n\nNuage neutron integration has a collection of mechanism\ndrivers sharing same configuration. This patchset removes check\non nuage ml2 mechanism driver which is not mandatory anymore.\n\nChange-Id: Id6e609d0a556e23c484f38f4651d2709d898a76c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/179bbd63ad23175d07858cacd9fc46dca0ed74f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/179bbd63ad23175d07858cacd9fc46dca0ed74f7"}]},"branch":"refs/heads/master"},"1c5d4cf463dcb05b4a36d397756edd6218dca736":{"kind":"REWORK","_number":3,"created":"2020-07-07 12:51:41.000000000","uploader":{"_account_id":13988,"name":"Vlad Gridin","email":"vladyslav.gridin@nokia.com","username":"gridinv"},"ref":"refs/changes/57/739657/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/57/739657/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/57/739657/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/57/739657/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/57/739657/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/57/739657/3"}}},"commit":{"parents":[{"commit":"de2be7503caff6bb1bd2ab3865f2da8913fa0c5b","subject":"Merge \"Deprecated neutron fwaas\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/de2be7503caff6bb1bd2ab3865f2da8913fa0c5b"}]}],"author":{"name":"Vlad Gridin","email":"vladyslav.gridin@nokia.com","date":"2020-07-07 05:02:08.000000000","tz":120},"committer":{"name":"Vlad Gridin","email":"vladyslav.gridin@nokia.com","date":"2020-07-07 12:50:34.000000000","tz":120},"subject":"Nuage: remove check on mechanism driver","message":"Nuage: remove check on mechanism driver\n\nNuage neutron integration has a collection of mechanism\ndrivers sharing same configuration. This patchset removes check\non nuage ml2 mechanism driver which is not mandatory anymore.\n\nChange-Id: Id6e609d0a556e23c484f38f4651d2709d898a76c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/1c5d4cf463dcb05b4a36d397756edd6218dca736"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/1c5d4cf463dcb05b4a36d397756edd6218dca736"}]},"branch":"refs/heads/master"},"6f342575c8f942eb8caa62b15a7bf839d3913177":{"kind":"REWORK","_number":4,"created":"2020-07-08 07:39:46.000000000","uploader":{"_account_id":13988,"name":"Vlad Gridin","email":"vladyslav.gridin@nokia.com","username":"gridinv"},"ref":"refs/changes/57/739657/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/57/739657/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/57/739657/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/57/739657/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/57/739657/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/57/739657/4"}}},"commit":{"parents":[{"commit":"de2be7503caff6bb1bd2ab3865f2da8913fa0c5b","subject":"Merge \"Deprecated neutron fwaas\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/de2be7503caff6bb1bd2ab3865f2da8913fa0c5b"}]}],"author":{"name":"Vlad Gridin","email":"vladyslav.gridin@nokia.com","date":"2020-07-08 07:35:45.000000000","tz":120},"committer":{"name":"Vlad Gridin","email":"vladyslav.gridin@nokia.com","date":"2020-07-08 07:37:59.000000000","tz":120},"subject":"Nuage: relax check on mechanism driver","message":"Nuage: relax check on mechanism driver\n\nNuage neutron integration has a collection of mechanism\ndrivers sharing same configuration. This patchset adapts\ncheck on nuage ml2 mechanism driver to use regex match.\n\nChange-Id: Id6e609d0a556e23c484f38f4651d2709d898a76c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/6f342575c8f942eb8caa62b15a7bf839d3913177"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/6f342575c8f942eb8caa62b15a7bf839d3913177"}]},"branch":"refs/heads/master"},"9fc323f48a466dfb5bc64c048a67ce878599ff2f":{"kind":"REWORK","_number":5,"created":"2020-07-08 10:04:49.000000000","uploader":{"_account_id":13988,"name":"Vlad Gridin","email":"vladyslav.gridin@nokia.com","username":"gridinv"},"ref":"refs/changes/57/739657/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/57/739657/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/57/739657/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/57/739657/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/57/739657/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/57/739657/5"}}},"commit":{"parents":[{"commit":"de2be7503caff6bb1bd2ab3865f2da8913fa0c5b","subject":"Merge \"Deprecated neutron fwaas\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/de2be7503caff6bb1bd2ab3865f2da8913fa0c5b"}]}],"author":{"name":"Vlad Gridin","email":"vladyslav.gridin@nokia.com","date":"2020-07-07 05:02:08.000000000","tz":120},"committer":{"name":"Vlad Gridin","email":"vladyslav.gridin@nokia.com","date":"2020-07-08 10:03:34.000000000","tz":120},"subject":"Nuage: remove check on mechanism driver","message":"Nuage: remove check on mechanism driver\n\nNuage neutron integration has a collection of mechanism\ndrivers sharing same configuration. This patchset removes check\non nuage ml2 mechanism driver which is not mandatory anymore.\n\nChange-Id: Id6e609d0a556e23c484f38f4651d2709d898a76c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/9fc323f48a466dfb5bc64c048a67ce878599ff2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/9fc323f48a466dfb5bc64c048a67ce878599ff2f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
