)]}'
{"id":"openstack%2Fpuppet-neutron~748331","triplet_id":"openstack%2Fpuppet-neutron~master~I0cffcd373d9882fe3b20643d589fdd6de73cc66c","project":"openstack/puppet-neutron","branch":"master","topic":"dnm","hashtags":[],"change_id":"I0cffcd373d9882fe3b20643d589fdd6de73cc66c","subject":"DNM: Test beaker job","status":"ABANDONED","created":"2020-08-27 00:51:09.000000000","updated":"2020-08-29 04:27:06.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"567db3fa72638e14d280ae3e75d8aa62c5a1939b","_number":748331,"virtual_id_number":748331,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-08-27 03:04:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-27 03:04:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9a1b5e38bd5e027348f0d519130202b91a46a57e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-27 00:51:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7372f38f316343e362eda6c0c5c0b690f5313e60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 03:04:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e8971f4e83ba4f88aaff4f3fc20b93c9 : SUCCESS in 2m 53s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/852d371c4b1c4996ae13cc01865b0aa7 : SUCCESS in 3m 28s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/226e719f104346a9b80ae9a98b068481 : SUCCESS in 4m 17s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/91ec1af0b8dc406e95c66d96d2d91d54 : SUCCESS in 17m 45s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/21a21618817d4cd78a35a7f90ef84651 : SUCCESS in 17m 38s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/60f5a5cfe3134bc5b5d780163341c6bc : SUCCESS in 16m 37s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/57ff9c58c9a4489290969c4a9f2e772e : FAILURE in 17m 44s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/0e9575c579984c0da84c7e38736aa96a : SUCCESS in 1h 29m 16s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/88afa9b2d14142a69e3186210bba63e1 : FAILURE in 1h 33m 08s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fce8728b9f05455598a7f6715a59527d : SUCCESS in 59m 31s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/c452d339a3c84179b64148e3a520f31d : SUCCESS in 49m 32s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/817ef1809cd441a1992b03fd22f61a97 : SUCCESS in 1h 24m 09s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/679d505c07594431b28145afde303e80 : SUCCESS in 1h 35m 13s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/902ab64b075348eea4931bbf9051614a : SUCCESS in 1h 30m 08s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/38ac9e022f5249e19f24b74e88930f55 : SUCCESS in 1h 02m 40s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/7766360239dc48a5a98dbdaae19f6425 : FAILURE in 24m 57s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fb7494ed3f664941b245d880227e1ec7 : FAILURE in 23m 06s (non-voting)\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d155f959023b4faa94a8d1b5cd241969 : SUCCESS in 2h 11m 54s","accounts_in_message":[],"_revision_number":1},{"id":"ebd1df64d9744ff93a5610d843c4e4113ae9724b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-29 04:27:06.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"14d2a144bcdb5d3852081575dd757091ad231da3","revisions":{"14d2a144bcdb5d3852081575dd757091ad231da3":{"kind":"REWORK","_number":1,"created":"2020-08-27 00:51:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/31/748331/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/31/748331/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/31/748331/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/31/748331/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/31/748331/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/31/748331/1"}}},"commit":{"parents":[{"commit":"5758b34e4b87d28b959811474ea45e718306c541","subject":"Remove deprecated firewall_driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/5758b34e4b87d28b959811474ea45e718306c541"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-27 00:50:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-27 00:50:50.000000000","tz":540},"subject":"DNM: Test beaker job","message":"DNM: Test beaker job\n\nChange-Id: I0cffcd373d9882fe3b20643d589fdd6de73cc66c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/14d2a144bcdb5d3852081575dd757091ad231da3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/14d2a144bcdb5d3852081575dd757091ad231da3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
