)]}'
{"id":"openstack%2Fpuppet-neutron~748468","triplet_id":"openstack%2Fpuppet-neutron~master~Ia6fb46e21b9483fdf26a90c7e44d6556c80af2b2","project":"openstack/puppet-neutron","branch":"master","topic":"selinux","hashtags":[],"change_id":"Ia6fb46e21b9483fdf26a90c7e44d6556c80af2b2","subject":"Test os_neutron_dac_override: on","status":"ABANDONED","created":"2020-08-27 15:24:30.000000000","updated":"2020-08-28 08:00:26.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2c361cb676929fc62a350be17a2fb16c43609958","_number":748468,"virtual_id_number":748468,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-08-28 05:45:04.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-08-27 17:16:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-27 17:16:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8809512b1fea1d721492ec4feeaac6ac18ca591a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-27 15:24:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"797ae935b9a573607d3e2c1965508cff90eea97d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 16:01:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/21f8035755164d159dcd029937fc896d : SUCCESS in 4m 39s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2b44e2d618c94df2b2f83e9b6d074130 : SUCCESS in 4m 01s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/46c246bac3ae4f9d883ba9758804ce4b : SUCCESS in 4m 11s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/bd49285f65ea411fa532845146a10353 : FAILURE in 26m 22s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/59fbe0952bb2451991d21c2808f68004 : FAILURE in 21m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"044371a2150c5ddf3f5a5ce0b17a90b514bf7e22","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-27 16:42:58.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"a4acf6938292ff2863bbb2e769107a8b4f00bf02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 17:16:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3271ea2d1d0f4271b384b003a773f21a : SUCCESS in 3m 46s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dfc44302ce404518b73ab67f471ac6cf : SUCCESS in 3m 15s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dbdb0877709f4b7fbfe35470eba0683a : SUCCESS in 3m 13s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/996b6e1d1ff74491ae3413ff84afc1f0 : SUCCESS in 23m 39s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/159d0eb11bcb48eab316299147ab965d : SUCCESS in 16m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"188dfcbf2c7818e0c8986f703277c7d73a789a36","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-28 05:45:04.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"125bb51d5c7dd705f7cee77737eb182217214454","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 06:07:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/173b7b477f424f52bc43e1db533a2533 : SUCCESS in 3m 00s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/448e118f1d0e4deabd1095ccdc2d2df3 : SUCCESS in 2m 29s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/33e1c2c667e749dbb605c8414e54b8f6 : SUCCESS in 3m 36s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/113ee09ede2a484f8ce56090abff11f9 : SUCCESS in 18m 06s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c212422eb29c4476a775c70968d4872d : SUCCESS in 14m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6f1118f5fcab56ca0d840633e13bf8b0ad297042","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-28 08:00:26.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4d756f213da25ad93c76d34848ca15c2e7ca457e","revisions":{"f479989b0a3d00f201185f4df2135b71380dfc76":{"kind":"REWORK","_number":1,"created":"2020-08-27 15:24:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/68/748468/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/68/748468/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/68/748468/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/68/748468/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/68/748468/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/68/748468/1"}}},"commit":{"parents":[{"commit":"5758b34e4b87d28b959811474ea45e718306c541","subject":"Remove deprecated firewall_driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/5758b34e4b87d28b959811474ea45e718306c541"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-27 14:20:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-27 15:24:01.000000000","tz":540},"subject":"Test os_neutron_dac_override: on","message":"Test os_neutron_dac_override: on\n\nDepends-on: https://review.opendev.org/#/c/748463\nChange-Id: Ia6fb46e21b9483fdf26a90c7e44d6556c80af2b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/f479989b0a3d00f201185f4df2135b71380dfc76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/f479989b0a3d00f201185f4df2135b71380dfc76"}]},"branch":"refs/heads/master"},"4d756f213da25ad93c76d34848ca15c2e7ca457e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-08-27 16:42:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/68/748468/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/68/748468/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/68/748468/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/68/748468/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/68/748468/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/68/748468/2"}}},"commit":{"parents":[{"commit":"945f61428790e52ca49bb99ab1d8f3a0f65c724b","subject":"Use common class to manage apache service in beaker job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/945f61428790e52ca49bb99ab1d8f3a0f65c724b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-27 14:20:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-27 16:42:58.000000000","tz":0},"subject":"Test os_neutron_dac_override: on","message":"Test os_neutron_dac_override: on\n\nDepends-on: https://review.opendev.org/#/c/748463\nChange-Id: Ia6fb46e21b9483fdf26a90c7e44d6556c80af2b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/4d756f213da25ad93c76d34848ca15c2e7ca457e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/4d756f213da25ad93c76d34848ca15c2e7ca457e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
