)]}'
{"id":"openstack%2Fpuppet-neutron~749124","triplet_id":"openstack%2Fpuppet-neutron~master~Id5f3ab3fdd9292de2c4ffe6cbca8e812273ca079","project":"openstack/puppet-neutron","branch":"master","topic":"749124","hashtags":[],"change_id":"Id5f3ab3fdd9292de2c4ffe6cbca8e812273ca079","subject":"Add Puppet Litmus","status":"MERGED","created":"2020-08-31 19:42:13.000000000","updated":"2020-09-10 03:08:31.000000000","submitted":"2020-09-10 03:08:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749124-1599707311980-06ebbb73","meta_rev_id":"37476a67e3871721e3a11b2a67fbd9e72b22e138","_number":749124,"virtual_id_number":749124,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"date":"2020-09-09 22:56:23.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-10 03:08:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-09-08 18:28:09.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-09-08 18:20:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-09-08 18:28:09.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-09-09 21:27:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-08 18:28:09.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-09-09 21:27:14.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-09-10 03:08:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a7fa9b6a0910ca31db3225d51b3478718437d499","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-08-31 19:42:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e3d62fbfd6c0a2f78302341e3a18176149fa33f","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-08-31 19:43:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"290aa30bc10509f812824151cdd8cdc19c727baf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 22:10:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e06c22c8791c47158def016080ea85b1 : SUCCESS in 3m 32s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ebfe9eea771d4b6b94d3208861bbf1d1 : SUCCESS in 3m 19s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/28eb3c251641420bad57dbe4610213ad : SUCCESS in 18m 35s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b082fb4cd7574cfeb6a37a01cfe7a89e : SUCCESS in 16m 19s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d0974b3c568f4f878339b6f6ae273062 : FAILURE in 18m 20s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/1fa37798a7ec49cd9a9fa00d3503f29a : SUCCESS in 54m 39s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5b67b4f6724b4c3cb2aa76ff813c1fc3 : FAILURE in 48m 53s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cbc2615a91e248ee88a98f8a2fa26b87 : SUCCESS in 47m 06s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/5fc75210e1bd4733a48f0107051d5716 : SUCCESS in 38m 51s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d75886e2261748699f1afa453dcd05b4 : SUCCESS in 1h 15m 12s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/133cec1c8e7e4d23aeada9f11b976c6d : SUCCESS in 1h 49m 18s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/317f72b305a54b49a4b468fbc690032f : SUCCESS in 1h 29m 41s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/93f20dd822a7426297e5fae974761dec : SUCCESS in 57m 28s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/5228a2b1f16d47f9bf2a54972d610588 : SUCCESS in 9m 20s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1455698bf76d40999666c007295e5100 : SUCCESS in 13m 51s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/531803b5a2824d1fb9861f43e88d9303 : SUCCESS in 2h 20m 33s","accounts_in_message":[],"_revision_number":2},{"id":"dca0fa8948403b46d58199bc15ce01a6c7483951","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-09-08 18:20:42.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"13b36ef3d48cae7bdaaff28efa2c7430da62d3ed","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-09-08 18:28:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2ff291cb27dfcb4a3346c0f449074c9ff67dd247","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-09-09 21:27:14.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5157e3187a8dcd5586c9ba8f126108e2dc4936c5","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-09-09 22:56:23.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c4b1f92669d6f52a2f5845c38c270f88b8621246","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 01:31:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/37cf22902ec54df981779d4110a7a7aa : SUCCESS in 4m 21s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/59fc9e84403f4ce6b1fd7b652bbef20f : SUCCESS in 3m 45s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/c726e598ebf342b2aad152218ac675d6 : SUCCESS in 20m 01s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b1b0da0d4c8c47d5b537a46fb95112d9 : SUCCESS in 17m 25s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/625b74a5ce79437991fc78f814e27181 : FAILURE in 20m 53s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/c56d0eb30a514a7facdf152ebce78ebe : SUCCESS in 53m 20s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/54a80a87c7ea4f0c98daa12fb08666ce : SUCCESS in 56m 04s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e9332c911db444eda1cbc975057965de : SUCCESS in 50m 37s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/383c14180e8a47e6b2f6b322f610dc75 : SUCCESS in 39m 24s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/525b8685b1924862924b3646ad30193a : SUCCESS in 1h 12m 24s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/befa1c73dd2c478e857a9d9c0ff6d373 : SUCCESS in 1h 28m 21s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f0fcf8a2a2754ec8855d5adbf47e7a24 : SUCCESS in 1h 33m 18s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/66b5664118d64623a5ff62eb9cdb775d : SUCCESS in 56m 42s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/d548ab93db3249beb8572e203286aaa2 : SUCCESS in 32m 42s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/12759eafb6714144b001154e033b0145 : SUCCESS in 18m 36s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6265b13251474a42b1d1bcea4e9295f5 : SUCCESS in 2h 01m 48s","accounts_in_message":[],"_revision_number":2},{"id":"0ea455673fefaa10f9e571b924f882c95737dcc7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 01:31:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"751e5025515f0e6df2640b4cde18abe44809fd48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 03:08:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b4e679eb5adc428cbf0ea68ba7f8457e : SUCCESS in 3m 57s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6c189ee42c634e21a6445ea7cfab1dfd : SUCCESS in 3m 58s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/88a81d6bb8f04ddf997ba6fe1d73be52 : SUCCESS in 18m 53s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eaf9db9bf6d248adbb921c7d945c68af : SUCCESS in 17m 20s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/eb422cf09a774619934f37d331917238 : SUCCESS in 1h 11m 42s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ab6243f587e0489299086ef7316b82c8 : SUCCESS in 1h 17m 03s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1572154243f74a08954614a7e016347c : SUCCESS in 1h 12m 46s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f0804288256c4052b6698f558e37ed68 : SUCCESS in 56m 28s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/b109ad39b34c4aacaa57b002083ac67d : SUCCESS in 19m 31s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6a0334edb06740f0bfaf82818fd4a383 : SUCCESS in 15m 37s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/eda7141895264af5b0bb13d33c7b607a : SUCCESS in 1h 32m 17s","accounts_in_message":[],"_revision_number":2},{"id":"5b26ea8aa93545f8a13325ca164bad7ec4c1f8ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 03:08:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f676b88e3223bfbfc08b6d6c95b6e9c69f53f67c","revisions":{"226aea0fa6d4597e170eb5fa5f768cc87b424540":{"kind":"REWORK","_number":1,"created":"2020-08-31 19:42:13.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/24/749124/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/24/749124/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/24/749124/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/24/749124/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/24/749124/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/24/749124/1"}}},"commit":{"parents":[{"commit":"d6d7add657c6f05ebad29a3a898266ca8c0a3e70","subject":"Merge \"Use openstack_spec_helper from zuul checkout\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/d6d7add657c6f05ebad29a3a898266ca8c0a3e70"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-08-31 19:42:04.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-08-31 19:42:04.000000000","tz":120},"subject":"Add Puppet Litmus","message":"Add Puppet Litmus\n\nDepends-On: https://review.opendev.org/#/c/740601/\nChange-Id: Id5f3ab3fdd9292de2c4ffe6cbca8e812273ca079\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/226aea0fa6d4597e170eb5fa5f768cc87b424540"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/226aea0fa6d4597e170eb5fa5f768cc87b424540"}]},"branch":"refs/heads/master"},"f676b88e3223bfbfc08b6d6c95b6e9c69f53f67c":{"kind":"REWORK","_number":2,"created":"2020-08-31 19:43:26.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/24/749124/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/24/749124/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/24/749124/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/24/749124/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/24/749124/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/24/749124/2"}}},"commit":{"parents":[{"commit":"d6d7add657c6f05ebad29a3a898266ca8c0a3e70","subject":"Merge \"Use openstack_spec_helper from zuul checkout\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/d6d7add657c6f05ebad29a3a898266ca8c0a3e70"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-08-31 19:42:04.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-08-31 19:43:20.000000000","tz":120},"subject":"Add Puppet Litmus","message":"Add Puppet Litmus\n\nDepends-On: https://review.opendev.org/#/c/740601/\nChange-Id: Id5f3ab3fdd9292de2c4ffe6cbca8e812273ca079\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/f676b88e3223bfbfc08b6d6c95b6e9c69f53f67c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/f676b88e3223bfbfc08b6d6c95b6e9c69f53f67c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
