)]}'
{"id":"openstack%2Fpuppet-neutron~749969","triplet_id":"openstack%2Fpuppet-neutron~stable%2Fstein~I6b9cc2aff14ce38a56c2757d9ad6c56898e27247","project":"openstack/puppet-neutron","branch":"stable/stein","hashtags":[],"change_id":"I6b9cc2aff14ce38a56c2757d9ad6c56898e27247","subject":"[OVN] Change the default metadata_workers to 2","status":"MERGED","created":"2020-09-04 14:31:38.000000000","updated":"2020-09-28 14:48:10.000000000","submitted":"2020-09-08 15:55:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749969-1599580515107-bb2e130b","meta_rev_id":"85c2d9c30f6f7065d80657906af250be068e0d41","_number":749969,"virtual_id_number":749969,"owner":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-08 15:55:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-09-08 13:29:00.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-09-08 12:24:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":2,"date":"2020-09-08 12:44:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-09-08 13:29:00.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-08 12:24:30.000000000","updated_by":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"reviewer":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"state":"REVIEWER"},{"updated":"2020-09-08 12:44:54.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-09-08 13:29:00.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-09-08 15:55:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"510d36a01dcf7b2e4a8a2c37c63e6718d923d49e","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-09-04 14:31:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c29036df9b3ec91f1f8fe8359cafda1caa33271b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 17:17:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cd64d4866f7f482094c9949f9b4064d7 : SUCCESS in 5m 13s\n- puppet-openstack-syntax-4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d2d31d147f5a411ab2047daa2c615b40 : SUCCESS in 5m 50s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0633e992af6548f1870cdc64815e6f82 : SUCCESS in 4m 21s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a9ea3ee67f314cbdbc9dc4c98df540e9 : SUCCESS in 4m 08s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/315f9ddc45314d63992b15d32bf70137 : SUCCESS in 27m 16s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/3d3d54503bc94bf99d6de37353ea9d61 : SUCCESS in 19m 54s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/13ff1e2fe3194906b55ab12d874da7a0 : SUCCESS in 19m 30s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0733e4084d5a49c192afd5648531e4ba : FAILURE in 35m 13s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/b7ba69edd6e247c8adef363b6a924455 : SUCCESS in 58m 46s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/fd94dd7a093d44a7ae10264f6a160278 : SUCCESS in 55m 53s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/4178933e9bb34f4e986cf17057dfab49 : SUCCESS in 40m 07s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/a93bf2fdbe7a475a90e44c942dbcae98 : SUCCESS in 33m 50s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/9a949b3b25634d06bb0b3e15d624064b : FAILURE in 26m 59s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/810763f7c66a49b0bc357edf49bd9f6e : FAILURE in 27m 44s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/411c49af91c3457eb71cc575cab36bb8 : FAILURE in 21m 23s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/0776b6faa80d45ab9ad9b1d45a0cf5ff : FAILURE in 18m 38s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/0dacb384c7264b75be0289a4b9fdf556 : SUCCESS in 24m 00s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6236f83ff2a2496eb5a439884f8c0c0a : FAILURE in 10m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22e22ff89c144514a09431bc581f3ac4 : SUCCESS in 7m 01s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/68d7a7d51d7a42ecb239e9516d55c4cc : SUCCESS in 2h 39m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b0d5e44ecccf11d3eb4e6ce38c8b404be53a0b81","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2020-09-08 12:24:30.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7112d0aa21d6f7b0f1185494e252e111df72e2a7","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-09-08 12:44:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"da9cd1e9b32e7aa0046cd73f723966f41d1b4d4b","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-09-08 13:29:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c477263672f921047908fb6266c01397c1efc593","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 13:29:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0fd89e06c260861cca2d9f6cafd7888c0dd8e8be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 15:55:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/acede11181644f6c9064fff649bccd79 : SUCCESS in 4m 51s\n- puppet-openstack-syntax-4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eb3e9515095640b3aae5ef56109e211a : SUCCESS in 4m 09s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/71b88d6718dc49c284972b2fa670f9af : SUCCESS in 7m 40s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4c15429dce0449f9a92882a7540e728c : SUCCESS in 5m 17s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/94c76e568b3d420786bb84baa2d6ce35 : SUCCESS in 31m 47s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/b8ca7da40d2149fd938b7de16d5e0152 : SUCCESS in 33m 14s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/67c90e5173fb48d7ba21a9c15f6d1a02 : SUCCESS in 27m 56s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/a104bc0b198f498e894dc143cc2d6780 : SUCCESS in 59m 42s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/81592f6dee514bf0a5abaf059c2f847a : SUCCESS in 1h 06m 34s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/be3a5563ae2e46ed8d1d3a6a5f9e662f : SUCCESS in 1h 02m 51s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/c6afa2fd92294e8a8f02d5afe4b3d399 : SUCCESS in 30m 48s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6991d75ca83e47349f1f9885e772f5b9 : FAILURE in 10m 40s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fbe44bfbdc604caca9b05a9f664cc820 : SUCCESS in 4m 55s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8664788ef6194e0383904c15e4ce310d : SUCCESS in 2h 19m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"850a4f5de23a4794acdd7a6cba386a7826828cf1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 15:55:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"891dd51d13b73ed3dbe28989b0bd9114e9d91fdb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 16:03:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fefdc49874b7476f86b5170dd1859f76 : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":1},{"id":"85c2d9c30f6f7065d80657906af250be068e0d41","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-09-28 14:48:10.000000000","message":"Created a revert of this change as I883d9573c6285d4a2480d68c6f3a56299f4c4108","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"080fd53ec4d013c8147578322c676568970cc4bc","revisions":{"080fd53ec4d013c8147578322c676568970cc4bc":{"kind":"REWORK","_number":1,"created":"2020-09-04 14:31:38.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/69/749969/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/69/749969/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/69/749969/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/69/749969/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/69/749969/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/69/749969/1"}}},"commit":{"parents":[{"commit":"4aae155e479f3c0e797bee239670143e9ed6b48c","subject":"Merge \"Fix incorrect openswan package in CentOS8\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/4aae155e479f3c0e797bee239670143e9ed6b48c"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-02-07 13:34:45.000000000","tz":0},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-09-04 14:31:36.000000000","tz":60},"subject":"[OVN] Change the default metadata_workers to 2","message":"[OVN] Change the default metadata_workers to 2\n\nPrior to this patch, the metadata_workers parameter defaults to\n$os_workers (calculated here [0]) but, even tho it\u0027s capped at 12 workers\nthis number still redundant.\n\nThere are a few differences of how the OVN metadata agent works compared\nwith the OVS metadata agents:\n\n1) The OVN metadata workers are deployed onto the compute nodes and not\n   on the controllers/gateways as the OVS ones are. This means it\u0027s\n   better distributed.\n\n2) Even tho we could have multiple workers, the OVN SB IDL in the OVN\n   metadata agents acquires a lock while listening to OVSDB events.\n   Therefore, even tho we could have dozens of agents only one of them\n   is listening to the events.\n\n3) Instantiating IDLs means maintaining in-memory replicas of the\n   OVSDB (or, the tables of that DB that is being watched). This puts a\n   burden on OVSDBs to maintain all these replicas updated, reducing\n   this number workers also reduces the burden on the OVSDBs.\n\nThis patch is reducing the default number of OVN metadata agents to two\n(per compute). Two because that keeps it HA, in case the active agente\ndies the lock moves to the other one and the service will keep working.\n\n[0]\nhttps://github.com/openstack/puppet-openstacklib/blob/74e0f5a3170b09632ad4ede7e20d3ebeb4514586/lib/facter/os_workers.rb#L37\n\nChange-Id: I6b9cc2aff14ce38a56c2757d9ad6c56898e27247\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n(cherry picked from commit 847f434140ee8435ee842801748a0deccdff8155)\n(cherry picked from commit 2ed6a1936ac71807d650cf26cfbcd9f4330331b6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/080fd53ec4d013c8147578322c676568970cc4bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/080fd53ec4d013c8147578322c676568970cc4bc"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
