)]}'
{"id":"openstack%2Fpuppet-neutron~749970","triplet_id":"openstack%2Fpuppet-neutron~stable%2Frocky~I6b9cc2aff14ce38a56c2757d9ad6c56898e27247","project":"openstack/puppet-neutron","branch":"stable/rocky","hashtags":[],"change_id":"I6b9cc2aff14ce38a56c2757d9ad6c56898e27247","subject":"[OVN] Change the default metadata_workers to 2","status":"MERGED","created":"2020-09-04 14:32:00.000000000","updated":"2020-09-09 11:58:16.000000000","submitted":"2020-09-09 11:55:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749970-1599652554236-a3cb6180","meta_rev_id":"c54bfe7e90cb24962bce6692554addea6e1f97a2","_number":749970,"virtual_id_number":749970,"owner":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-09-09 08:13:15.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-09 11:55:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":2,"date":"2020-09-08 15:08:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2020-09-08 18:20:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2020-09-08 18:20:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-08 15:08:52.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-09-08 18:20:53.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-09-09 11:55:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0329dc54a0e849e873f1079916584077435816d9","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-09-04 14:32:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b3b09750b3311a52cd50e77aa43f2a334f060d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 16:52:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/10b4c5d1fd014762a100403e0aa6ff85 : FAILURE in 5m 32s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/e7bb38fcf8f548dbbf7ece651492636b : SUCCESS in 4m 16s\n- puppet-openstack-syntax-5 https://zuul.opendev.org/t/openstack/build/2860893acf714061bda41e02f50d6c0b : SUCCESS in 4m 49s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/766d6f60bbba4e16b8a015c28ae1c3dd : SUCCESS in 29m 53s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/fefea6e921cb406783e722b6bad29499 : SUCCESS in 26m 03s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/bc291d0965aa4ccf81f293d1d0eaff59 : SUCCESS in 1h 00m 33s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/f785f2b0987b474787af3c7e4ec45910 : SUCCESS in 1h 00m 56s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/f52c19ffff12414a89c2a469cf866391 : SUCCESS in 59m 21s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic https://zuul.opendev.org/t/openstack/build/ad7ad437d73e4f27a438f6c8dae4d848 : SUCCESS in 59m 39s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/045c8751a80949cf85baafd4eaf19c46 : FAILURE in 29m 02s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6aac274f3edc41e881e0e534d14b431a : FAILURE in 32m 45s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a0902089c5454bddbb514a2d71eee1ba : FAILURE in 34m 50s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/2893fc6962ed4321930c697b201e1b34 : FAILURE in 19m 36s (non-voting)\n- puppet-openstack-integration-4-scenario001-tempest-debian-stable-luminous https://zuul.opendev.org/t/openstack/build/e9c491aa53ac415fbcbbf73ec10dfb8f : TIMED_OUT in 1h 31m 29s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-debian-stable https://zuul.opendev.org/t/openstack/build/bb9c07f1212a498cbbcc13c6e4f436bc : FAILURE in 51m 42s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-debian-stable https://zuul.opendev.org/t/openstack/build/47a569ca04a0420aaff27d38aac97d68 : TIMED_OUT in 1h 31m 15s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-debian-stable-luminous https://zuul.opendev.org/t/openstack/build/d2bd844a4b1c492e88c35f8366002e58 : TIMED_OUT in 1h 31m 11s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/6aac998764ca4fe29c40848f2ba5d453 : SUCCESS in 21m 40s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4cd4f0aa6737458895a02cfe01ffb48c : SUCCESS in 21m 35s\n- tripleo-puppet-ci-centos-7-undercloud-oooq https://zuul.opendev.org/t/openstack/build/3a59d6d6064549dda24238257806aa9b : SUCCESS in 1h 04m 26s\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/3bb1a79ebdfe4cf0a09b069454584825 : SUCCESS in 2h 05m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9de3871cda649cbbef320deff95dc9c : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":1},{"id":"1998bbbd9561159eb9c3214caaf914fbdf97ccf5","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-09-08 12:50:09.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"84a4b95c26bf168a0186d3a3743a7c65542db677","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 14:27:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/00e459dc22e64d1faef1b6b55bbfc862 : SUCCESS in 5m 19s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/408ec1211f2d4ed8971d664477fad0c7 : SUCCESS in 5m 58s\n- puppet-openstack-syntax-5 https://zuul.opendev.org/t/openstack/build/25cb10eba4d14a88a5d34cdaebd657ae : SUCCESS in 5m 14s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/10047b2d9a384dfc97af85f75d7bbede : SUCCESS in 25m 32s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/8c73c4a5f0ad4a6cb485b4a98e16ee33 : SUCCESS in 26m 04s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/36ab50d530f74c8e8f906f1574c753ba : SUCCESS in 50m 40s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/3e12583d50ed4a1e8c33d7548c43aa56 : SUCCESS in 55m 19s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/57cefde93c6d48bcbd757f6e3154b998 : SUCCESS in 1h 06m 49s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic https://zuul.opendev.org/t/openstack/build/c3f6f251265744938595c7a3b38e4f24 : SUCCESS in 51m 57s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/080d32a604654c1a8c30a9048f333e17 : FAILURE in 32m 41s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f9f74fc8267b47d195c0bed6fe46e67f : FAILURE in 48m 39s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ee693fafdd3146e191caf7d27004ed03 : FAILURE in 32m 23s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/0f512ffa661e48858fb2bfe70308025f : FAILURE in 21m 07s (non-voting)\n- puppet-openstack-integration-4-scenario001-tempest-debian-stable-luminous https://zuul.opendev.org/t/openstack/build/a8ef4ee8bdd7486babfdb6ad7af507c8 : TIMED_OUT in 1h 31m 00s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-debian-stable https://zuul.opendev.org/t/openstack/build/8c2955b6a7db447da7adf0ad4dac8aff : FAILURE in 52m 52s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-debian-stable https://zuul.opendev.org/t/openstack/build/43385b59d69740598848ec9c634cb0ce : TIMED_OUT in 1h 31m 00s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-debian-stable-luminous https://zuul.opendev.org/t/openstack/build/b93b83e1bcef4f7b87afd4b6bb08ad06 : TIMED_OUT in 1h 31m 03s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/68ce477078dd4773a512039362f02bc4 : SUCCESS in 20m 10s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fc170c21083f4b6392d9834d9e4d0954 : SUCCESS in 15m 55s\n- tripleo-puppet-ci-centos-7-undercloud-oooq https://zuul.opendev.org/t/openstack/build/45cc5d7b35374427925c82a98094fedf : SUCCESS in 1h 15m 52s\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/18a54207d7d5404b8ca6bc0b14f3e64b : SUCCESS in 1h 24m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/918d87a5a2ec4fc995f6c5d0e3689482 : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":2},{"id":"b997631dd3b17e0c29c5390992dbe584ae67d1a2","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-09-08 15:08:44.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"f60b536e60c8abaf4bd4f8b97d732c4536a53df4","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-09-08 15:08:52.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"1525438043825822595fa9733ee656baeab48010","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 17:01:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/1a25530d79654f67a1b939d650a9f697 : SUCCESS in 4m 39s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/0f775248b3d5488ab1636301871efa4f : SUCCESS in 4m 42s\n- puppet-openstack-syntax-5 https://zuul.opendev.org/t/openstack/build/51ee57c8fb6c45388b5b47950a458cd1 : SUCCESS in 5m 21s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/248a094062e5492888743a3de2499c5d : SUCCESS in 29m 39s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/340b849d94ad417c84d050970b316baf : SUCCESS in 25m 43s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/7773214e782c47e78581f138dae1d90c : SUCCESS in 51m 10s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/bee848e10a1d4dddaf60f79b5e9b253a : SUCCESS in 54m 12s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/c17ef4bbafe74943ae92328380474173 : SUCCESS in 57m 47s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic https://zuul.opendev.org/t/openstack/build/6035ceb1827143eeaf7adac35f92dc12 : SUCCESS in 45m 57s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/033337428cde4c04bc17b74a721281df : FAILURE in 33m 08s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/60de331d85994f97b445c5bd416df8b5 : FAILURE in 31m 35s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b1a320d3c4374212907feb660c2815b5 : FAILURE in 40m 11s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/28ea339b23864893b6d9921f7862dc1f : FAILURE in 24m 58s (non-voting)\n- puppet-openstack-integration-4-scenario001-tempest-debian-stable-luminous https://zuul.opendev.org/t/openstack/build/f92d8c519854404eb4b9d0c8eaef50c5 : TIMED_OUT in 1h 31m 09s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-debian-stable https://zuul.opendev.org/t/openstack/build/4f3d8c2eeea6419b81114a0ec8879b8f : FAILURE in 52m 38s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-debian-stable https://zuul.opendev.org/t/openstack/build/6f3cd0c74f0844d99563228d0aaf3b31 : TIMED_OUT in 1h 31m 03s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-debian-stable-luminous https://zuul.opendev.org/t/openstack/build/ee23fd46f1e24f809c3516bb9bafeb75 : TIMED_OUT in 1h 31m 43s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/2a1a013b16544333a5bdf334cfad739a : SUCCESS in 20m 05s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8dc75204b0124cb8a3191ed62732825b : SUCCESS in 20m 29s\n- tripleo-puppet-ci-centos-7-undercloud-oooq https://zuul.opendev.org/t/openstack/build/24a55b6338ce4161b95d660f5c816f17 : SUCCESS in 1h 10m 23s\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/d465d9719c9e43daa14ebc0265d8eff1 : SUCCESS in 1h 39m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0dc02269412945ffa4cec725bcbbfdc7 : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":3},{"id":"1368ff0540e03bdbc61f2ec4bf37372cc16c5ba9","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-09-08 18:20:53.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d3991ac378b9c547738787e51a45f851a5be159f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 18:23:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9d1c640652ff4d805a053ce55ac46c13c0950745","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 20:31:11.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/f7f216804c394935bc15186a40fbe2ae : SUCCESS in 4m 27s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/b6d5e21e72104b7b8914bf4292408802 : SUCCESS in 4m 07s\n- puppet-openstack-syntax-5 https://zuul.opendev.org/t/openstack/build/10c15eece1fa4fc8897d65ea0c333bae : SUCCESS in 4m 40s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/b8cd38f7c677420f8aaea6f7c8efc579 : SUCCESS in 31m 46s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/7a2b568dacb046c8994b2c6cdbae79aa : SUCCESS in 19m 48s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/bd5a65e4110b4575846a32c23ffe4562 : SUCCESS in 48m 06s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/b025744f9b144463a338b56328f58715 : SUCCESS in 49m 44s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/9e672f5d2cd44a698fde9050aaef61bb : FAILURE in 58m 13s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic https://zuul.opendev.org/t/openstack/build/7a6085e391c0439795c269b16abc2880 : SUCCESS in 41m 07s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/c5bf94ece0d049b6bc6f498ec291038b : SUCCESS in 17m 42s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0363507a07904720ae354a368d509d28 : SUCCESS in 17m 55s\n- tripleo-puppet-ci-centos-7-undercloud-oooq https://zuul.opendev.org/t/openstack/build/a0c08c268ebf4a30bb4a5be30d7a154d : SUCCESS in 1h 15m 37s\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/058d9ae079c148a3bc9636ae40c46ecf : SUCCESS in 1h 48m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61915a06b0754ca1b20857711e56843f : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":3},{"id":"4b6d09adf944eddc18b786eabb17ad943fbbefe4","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-09-08 22:34:23.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"da216dc2185647558bacd9f8fb48bf391cba6408","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 00:29:39.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/94e15f1667144715902bfe9290133857 : SUCCESS in 4m 35s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/1319e0d0a3b4406aa37a879f2459a25c : SUCCESS in 4m 07s\n- puppet-openstack-syntax-5 https://zuul.opendev.org/t/openstack/build/ab76c85769c44b05aa2738e244962061 : SUCCESS in 4m 27s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/dba1997c06004ff89149e80f964b45dc : SUCCESS in 22m 27s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/e1a820dc9bfa44e7bf3d56678a20160d : SUCCESS in 18m 16s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/a25327afac3c4d2b9924a849b7fc294c : SUCCESS in 52m 06s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/ac3f4b55714545e799e40c4e73449f9d : SUCCESS in 55m 29s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/d67ddf5a4af543b49414daa825761513 : SUCCESS in 56m 16s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic https://zuul.opendev.org/t/openstack/build/e5922bb27aa64f84bcb6787a94835afc : SUCCESS in 51m 27s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/4a7f8430458c4125b58f991a69d9a229 : FAILURE in 24m 50s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/847aaea2ab3a4b42b158da2470cfae94 : FAILURE in 28m 14s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/17bb27f6b6ed4c8a81c2087d177a80c3 : FAILURE in 32m 27s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/521b274f8fe84525ad12a55868eb840f : FAILURE in 20m 37s (non-voting)\n- puppet-openstack-integration-4-scenario001-tempest-debian-stable-luminous https://zuul.opendev.org/t/openstack/build/0b3723c9e7ca46da8bcace0a54c1a893 : TIMED_OUT in 1h 30m 43s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-debian-stable https://zuul.opendev.org/t/openstack/build/81637cf8de354fd39f1b454c2bb816a8 : FAILURE in 49m 12s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-debian-stable https://zuul.opendev.org/t/openstack/build/9b7fe58198ef4ae480ad3d50c1bfd6d3 : TIMED_OUT in 1h 30m 43s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-debian-stable-luminous https://zuul.opendev.org/t/openstack/build/abf3e9a19c7b469297d3a51407c6cc60 : TIMED_OUT in 1h 31m 01s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/1c421c4c1ee846ddbe023b97af37c5c3 : SUCCESS in 20m 23s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/224dbcc642d641ab82317225160becf0 : SUCCESS in 16m 14s\n- tripleo-puppet-ci-centos-7-undercloud-oooq https://zuul.opendev.org/t/openstack/build/dfcc5007b6e1469695e01d69125c9301 : SUCCESS in 1h 03m 54s\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/0c721e0d43dc4dfaa63d1205440056cd : FAILURE in 53m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d87a7de0588b4aceb585703c2a37ff5f : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":3},{"id":"8a87cdfed4777dce5f8044f87ae0cffc537014ae","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-09-09 08:13:15.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"06be8712aa446bd6282a9e5bf927296af0e1b8bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 09:55:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/6d5f50b17c804806b9b8e6ff0db49ac3 : SUCCESS in 4m 56s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/2168ff38e73f4095aae5f2499fa626e9 : SUCCESS in 5m 09s\n- puppet-openstack-syntax-5 https://zuul.opendev.org/t/openstack/build/0900172b250f4ed0afd90c863d9a9bd7 : SUCCESS in 4m 12s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/50cf944a8bf0428585e2875f47db76dc : SUCCESS in 24m 06s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/d667449ee80c41edbaa9075f28f0c648 : SUCCESS in 26m 42s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/25a3f80a6b0044709e119b532412b8c2 : SUCCESS in 51m 52s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/3bf2450d74d24506beeeffe69d2a9018 : SUCCESS in 55m 09s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/027b2f93a93147c6a36bbb230eb18792 : SUCCESS in 58m 04s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic https://zuul.opendev.org/t/openstack/build/09cb8c1f6e914861a7fcedaec5a4cb47 : SUCCESS in 30m 15s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/d6fe611e85444fe790b61c9ab4ddb470 : FAILURE in 23m 51s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6cfffaf3abfb43049602aa72d6aa81a3 : FAILURE in 29m 49s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a747ba960d70498cb1d4c17f2319b538 : FAILURE in 34m 39s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/b19b357ed08a4c50b80bb7de2f72a528 : FAILURE in 26m 46s (non-voting)\n- puppet-openstack-integration-4-scenario001-tempest-debian-stable-luminous https://zuul.opendev.org/t/openstack/build/100d37d8010c406681104d2b867ad47e : TIMED_OUT in 1h 31m 06s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-debian-stable https://zuul.opendev.org/t/openstack/build/2e8091281cf7416bbfd56da1296c3cdb : FAILURE in 49m 11s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-debian-stable https://zuul.opendev.org/t/openstack/build/bb3374b3698945a0b7a563a883089681 : TIMED_OUT in 1h 31m 17s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-debian-stable-luminous https://zuul.opendev.org/t/openstack/build/236c0dbce7c3434ba13aa804b14d5d9b : TIMED_OUT in 1h 30m 50s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/7a536988148640028cd1a2ae2541d5af : SUCCESS in 19m 01s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0307c638871a4e07ba2b4fd920625ebd : SUCCESS in 14m 39s\n- tripleo-puppet-ci-centos-7-undercloud-oooq https://zuul.opendev.org/t/openstack/build/bf919554b2704e18ad34de4862618c94 : SUCCESS in 1h 17m 09s\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/215d8a579fd24755922d74cd7da8e7ab : SUCCESS in 1h 24m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e050905851d45b58cbcbb976c636548 : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":3},{"id":"ba24d7d86d11e7f702c9c02a80a083fc0f31680b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 09:56:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"adfb950a2bb58ed97a4eac9739d39c4c298cf54d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 11:55:53.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/430ce31ac13441a09b6336737b5d88a6 : SUCCESS in 4m 14s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/76612a8a714b4109aecff5ace4526a9e : SUCCESS in 4m 05s\n- puppet-openstack-syntax-5 https://zuul.opendev.org/t/openstack/build/cdf1179e2f1345d29615018dca67a719 : SUCCESS in 4m 13s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/1f336a34cf2e4b088f9d3d2e7358b42f : SUCCESS in 17m 15s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/7a82e589629441b19bf9d6709c4e7ab4 : SUCCESS in 26m 31s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/2bb30baf632547c18ed10b3d133e10c8 : SUCCESS in 52m 23s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/f87fc9c7397c41b48701d9268d8bf7ec : SUCCESS in 54m 44s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/35bd08a58f4f44a081bf33fd326d6db8 : SUCCESS in 1h 27m 08s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic https://zuul.opendev.org/t/openstack/build/7ce5ad6fc7e84ff8826ad82b3190e16f : SUCCESS in 40m 56s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/8e113ca4c8ba49519e05602b30ddb356 : SUCCESS in 16m 43s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/718a11d31ffb402ca203a797e1aafb6d : SUCCESS in 20m 55s\n- tripleo-puppet-ci-centos-7-undercloud-oooq https://zuul.opendev.org/t/openstack/build/41f5c291082c41b3a84494bb07eb2c2a : SUCCESS in 1h 34m 11s\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/d565987214394d858a9c3716e54540e4 : SUCCESS in 1h 53m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7e12cdea8dc4ab09a6ccb0b83208172 : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":3},{"id":"146c3226b88ca207d65cf636cea379885ba7a71f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 11:55:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"c54bfe7e90cb24962bce6692554addea6e1f97a2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 11:58:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/40ff44c52d7d467198274be8a0499106 : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d839211929b1b94dd80026d176d9e3cb93de824b","revisions":{"133aedb00942a9178b18ec8404f507266c4ce787":{"kind":"REWORK","_number":1,"created":"2020-09-04 14:32:00.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/70/749970/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/70/749970/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/70/749970/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/70/749970/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/70/749970/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/70/749970/1"}}},"commit":{"parents":[{"commit":"37edd6ebd3ba413b5c29f91085d5e9577416eab8","subject":"Force cast bridge mappings as an array","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/37edd6ebd3ba413b5c29f91085d5e9577416eab8"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-02-07 13:34:45.000000000","tz":0},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-09-04 14:29:44.000000000","tz":60},"subject":"[OVN] Change the default metadata_workers to 2","message":"[OVN] Change the default metadata_workers to 2\n\nPrior to this patch, the metadata_workers parameter defaults to\n$os_workers (calculated here [0]) but, even tho it\u0027s capped at 12 workers\nthis number still redundant.\n\nThere are a few differences of how the OVN metadata agent works compared\nwith the OVS metadata agents:\n\n1) The OVN metadata workers are deployed onto the compute nodes and not\n   on the controllers/gateways as the OVS ones are. This means it\u0027s\n   better distributed.\n\n2) Even tho we could have multiple workers, the OVN SB IDL in the OVN\n   metadata agents acquires a lock while listening to OVSDB events.\n   Therefore, even tho we could have dozens of agents only one of them\n   is listening to the events.\n\n3) Instantiating IDLs means maintaining in-memory replicas of the\n   OVSDB (or, the tables of that DB that is being watched). This puts a\n   burden on OVSDBs to maintain all these replicas updated, reducing\n   this number workers also reduces the burden on the OVSDBs.\n\nThis patch is reducing the default number of OVN metadata agents to two\n(per compute). Two because that keeps it HA, in case the active agente\ndies the lock moves to the other one and the service will keep working.\n\n[0]\nhttps://github.com/openstack/puppet-openstacklib/blob/74e0f5a3170b09632ad4ede7e20d3ebeb4514586/lib/facter/os_workers.rb#L37\n\nChange-Id: I6b9cc2aff14ce38a56c2757d9ad6c56898e27247\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n(cherry picked from commit 847f434140ee8435ee842801748a0deccdff8155)\n(cherry picked from commit 2ed6a1936ac71807d650cf26cfbcd9f4330331b6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/133aedb00942a9178b18ec8404f507266c4ce787"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/133aedb00942a9178b18ec8404f507266c4ce787"}]},"branch":"refs/heads/stable/rocky"},"5d0132af080855878192ab1dc613587e394e7f3a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-09-08 12:50:09.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/70/749970/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/70/749970/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/70/749970/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/70/749970/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/70/749970/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/70/749970/2"}}},"commit":{"parents":[{"commit":"9f1924d54ad4f6b321c61a69cc913fefbeed4244","subject":"Fix lint issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/9f1924d54ad4f6b321c61a69cc913fefbeed4244"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-02-07 13:34:45.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-09-08 12:50:09.000000000","tz":0},"subject":"[OVN] Change the default metadata_workers to 2","message":"[OVN] Change the default metadata_workers to 2\n\nPrior to this patch, the metadata_workers parameter defaults to\n$os_workers (calculated here [0]) but, even tho it\u0027s capped at 12 workers\nthis number still redundant.\n\nThere are a few differences of how the OVN metadata agent works compared\nwith the OVS metadata agents:\n\n1) The OVN metadata workers are deployed onto the compute nodes and not\n   on the controllers/gateways as the OVS ones are. This means it\u0027s\n   better distributed.\n\n2) Even tho we could have multiple workers, the OVN SB IDL in the OVN\n   metadata agents acquires a lock while listening to OVSDB events.\n   Therefore, even tho we could have dozens of agents only one of them\n   is listening to the events.\n\n3) Instantiating IDLs means maintaining in-memory replicas of the\n   OVSDB (or, the tables of that DB that is being watched). This puts a\n   burden on OVSDBs to maintain all these replicas updated, reducing\n   this number workers also reduces the burden on the OVSDBs.\n\nThis patch is reducing the default number of OVN metadata agents to two\n(per compute). Two because that keeps it HA, in case the active agente\ndies the lock moves to the other one and the service will keep working.\n\n[0]\nhttps://github.com/openstack/puppet-openstacklib/blob/74e0f5a3170b09632ad4ede7e20d3ebeb4514586/lib/facter/os_workers.rb#L37\n\nChange-Id: I6b9cc2aff14ce38a56c2757d9ad6c56898e27247\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n(cherry picked from commit 847f434140ee8435ee842801748a0deccdff8155)\n(cherry picked from commit 2ed6a1936ac71807d650cf26cfbcd9f4330331b6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/5d0132af080855878192ab1dc613587e394e7f3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/5d0132af080855878192ab1dc613587e394e7f3a"}]},"branch":"refs/heads/stable/rocky"},"d839211929b1b94dd80026d176d9e3cb93de824b":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-09-08 15:08:44.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/70/749970/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/70/749970/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/70/749970/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/70/749970/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/70/749970/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/70/749970/3"}}},"commit":{"parents":[{"commit":"567d32427a57d0c5289d138f567b74e7a1625e70","subject":"Fix lint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/567d32427a57d0c5289d138f567b74e7a1625e70"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-02-07 13:34:45.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-09-08 15:08:44.000000000","tz":0},"subject":"[OVN] Change the default metadata_workers to 2","message":"[OVN] Change the default metadata_workers to 2\n\nPrior to this patch, the metadata_workers parameter defaults to\n$os_workers (calculated here [0]) but, even tho it\u0027s capped at 12 workers\nthis number still redundant.\n\nThere are a few differences of how the OVN metadata agent works compared\nwith the OVS metadata agents:\n\n1) The OVN metadata workers are deployed onto the compute nodes and not\n   on the controllers/gateways as the OVS ones are. This means it\u0027s\n   better distributed.\n\n2) Even tho we could have multiple workers, the OVN SB IDL in the OVN\n   metadata agents acquires a lock while listening to OVSDB events.\n   Therefore, even tho we could have dozens of agents only one of them\n   is listening to the events.\n\n3) Instantiating IDLs means maintaining in-memory replicas of the\n   OVSDB (or, the tables of that DB that is being watched). This puts a\n   burden on OVSDBs to maintain all these replicas updated, reducing\n   this number workers also reduces the burden on the OVSDBs.\n\nThis patch is reducing the default number of OVN metadata agents to two\n(per compute). Two because that keeps it HA, in case the active agente\ndies the lock moves to the other one and the service will keep working.\n\n[0]\nhttps://github.com/openstack/puppet-openstacklib/blob/74e0f5a3170b09632ad4ede7e20d3ebeb4514586/lib/facter/os_workers.rb#L37\n\nChange-Id: I6b9cc2aff14ce38a56c2757d9ad6c56898e27247\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n(cherry picked from commit 847f434140ee8435ee842801748a0deccdff8155)\n(cherry picked from commit 2ed6a1936ac71807d650cf26cfbcd9f4330331b6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/d839211929b1b94dd80026d176d9e3cb93de824b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/d839211929b1b94dd80026d176d9e3cb93de824b"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
