)]}'
{"id":"openstack%2Fpuppet-neutron~749972","triplet_id":"openstack%2Fpuppet-neutron~stable%2Fqueens~I6b9cc2aff14ce38a56c2757d9ad6c56898e27247","project":"openstack/puppet-neutron","branch":"stable/queens","hashtags":[],"change_id":"I6b9cc2aff14ce38a56c2757d9ad6c56898e27247","subject":"[OVN] Change the default metadata_workers to 2","status":"MERGED","created":"2020-09-04 14:32:21.000000000","updated":"2020-09-08 14:44:20.000000000","submitted":"2020-09-08 14:42:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749972-1599576141265-7f25b096","meta_rev_id":"fdeab64b5852b8dc37340b46147646e6dfc76bb9","_number":749972,"virtual_id_number":749972,"owner":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-08 14:42:21.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-09-08 13:28:57.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-09-08 12:44:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-09-08 13:28:57.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-08 12:44:46.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-09-08 13:28:57.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-09-08 14:42:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bfd888f8e6d161a8e83f879c18d552f124baab71","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-09-04 14:32:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b00e8510846cfeac574abc7a14d7c0863f1734a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 15:47:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/7249b88c66be4a4fb4bd528c2850bd43 : SUCCESS in 5m 58s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/7289cfe4e9c44a5c8fa997c147aa547d : SUCCESS in 5m 34s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/f00992458f224a22985ca64671031b57 : SUCCESS in 35m 50s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/0305c446356b4da2b1ddea0e1707357f : SUCCESS in 21m 16s\n- puppet-openstack-integration-4-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/96b8dff2a2544ec7b82ddb726150c4b7 : SUCCESS in 59m 10s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/279134a53900498a96d58de8cb159648 : SUCCESS in 56m 09s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/c32ed6ebf3a84d9abb9f53e7a4fc9eba : SUCCESS in 58m 17s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/117205c942a24743938d2d2d30933019 : SUCCESS in 47m 10s\n- puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial-luminous https://zuul.opendev.org/t/openstack/build/8707098fca424fc294be3d4ae88c3376 : FAILURE in 30m 02s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/6dd75574655e4e7dae14ca6b9c5708e8 : FAILURE in 34m 52s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/ce4cdbb61076472c99747b4ea8a448b4 : FAILURE in 27m 06s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial-luminous https://zuul.opendev.org/t/openstack/build/08233e80b0f14b61ad561d2c41251d59 : FAILURE in 21m 38s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/330910d244ee41e7bdf9ee10e8f6c165 : SUCCESS in 17m 40s\n- puppet-openstack-beaker-xenial https://zuul.opendev.org/t/openstack/build/c0eb550112b24c668dedb39d825b67db : SUCCESS in 24m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cfeb19be2c344f3ca38397f1c9b78c96 : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":1},{"id":"49cec92370a54bd3d27006175f4c7cbc178ba897","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-09-08 12:44:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0d88b9df3384a30ffcc77583dccc095f5b2d3d2b","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-09-08 13:28:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e3601047b4e2e973931627cd557389c5ea9c6c1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 13:29:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c28796cc77fae6491182ab3cbc46d89e1e971c89","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 14:42:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f36b5ed6dc21d49a51942385e0a38ba581df695f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 14:42:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/ec80ff32194948d39d38169e39fd8b67 : SUCCESS in 6m 15s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/240f9a44ba9e464f92b7a7d3407e39e2 : SUCCESS in 5m 04s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/77c55fcf354f4e30999ea1aee6b72846 : SUCCESS in 22m 07s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/af498b7475694cbe870f71891dafa637 : SUCCESS in 19m 14s\n- puppet-openstack-integration-4-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/8dcb9d2e43564f658cd44778868ac5e9 : SUCCESS in 58m 59s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/7cfc1ac02a3149c4a7d9cdf1fe03f486 : SUCCESS in 56m 10s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/b5d0bab0ac7840fd940cdc3ee1a2c592 : SUCCESS in 41m 31s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/a199205e37d8457e99472ed545d0ca60 : SUCCESS in 1h 00m 48s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/b3686c5f0ea047ef9c03ef96dc9ca4d7 : SUCCESS in 20m 34s\n- puppet-openstack-beaker-xenial https://zuul.opendev.org/t/openstack/build/aaa52afab0184cc5b9dfba1edc74ae97 : SUCCESS in 33m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd513917b58f4dcb8d8b557f8c3ccf77 : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":1},{"id":"fdeab64b5852b8dc37340b46147646e6dfc76bb9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 14:44:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ba1e417a2b94ffb838db8415d292c11 : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3987a1b53907543e9c730b5e62da2e498cc9065c","revisions":{"3987a1b53907543e9c730b5e62da2e498cc9065c":{"kind":"REWORK","_number":1,"created":"2020-09-04 14:32:21.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/72/749972/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/72/749972/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/72/749972/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/72/749972/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/72/749972/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/72/749972/1"}}},"commit":{"parents":[{"commit":"f842cbf8d70f73699fb741bf6b19b4ea598b6bb3","subject":"Add of_connect_timeout and of_request_timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/f842cbf8d70f73699fb741bf6b19b4ea598b6bb3"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-02-07 13:34:45.000000000","tz":0},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-09-04 14:30:58.000000000","tz":60},"subject":"[OVN] Change the default metadata_workers to 2","message":"[OVN] Change the default metadata_workers to 2\n\nPrior to this patch, the metadata_workers parameter defaults to\n$os_workers (calculated here [0]) but, even tho it\u0027s capped at 12 workers\nthis number still redundant.\n\nThere are a few differences of how the OVN metadata agent works compared\nwith the OVS metadata agents:\n\n1) The OVN metadata workers are deployed onto the compute nodes and not\n   on the controllers/gateways as the OVS ones are. This means it\u0027s\n   better distributed.\n\n2) Even tho we could have multiple workers, the OVN SB IDL in the OVN\n   metadata agents acquires a lock while listening to OVSDB events.\n   Therefore, even tho we could have dozens of agents only one of them\n   is listening to the events.\n\n3) Instantiating IDLs means maintaining in-memory replicas of the\n   OVSDB (or, the tables of that DB that is being watched). This puts a\n   burden on OVSDBs to maintain all these replicas updated, reducing\n   this number workers also reduces the burden on the OVSDBs.\n\nThis patch is reducing the default number of OVN metadata agents to two\n(per compute). Two because that keeps it HA, in case the active agente\ndies the lock moves to the other one and the service will keep working.\n\n[0]\nhttps://github.com/openstack/puppet-openstacklib/blob/74e0f5a3170b09632ad4ede7e20d3ebeb4514586/lib/facter/os_workers.rb#L37\n\nChange-Id: I6b9cc2aff14ce38a56c2757d9ad6c56898e27247\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n(cherry picked from commit 847f434140ee8435ee842801748a0deccdff8155)\n(cherry picked from commit 2ed6a1936ac71807d650cf26cfbcd9f4330331b6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/3987a1b53907543e9c730b5e62da2e498cc9065c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/3987a1b53907543e9c730b5e62da2e498cc9065c"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
