)]}'
{"id":"openstack%2Fpuppet-neutron~750100","triplet_id":"openstack%2Fpuppet-neutron~master~I4361ea53bd964a3a4861c4c28abb7063e19ebd94","project":"openstack/puppet-neutron","branch":"master","topic":"api_paste_ini","hashtags":[],"change_id":"I4361ea53bd964a3a4861c4c28abb7063e19ebd94","subject":"Deprecate neutron_api_config","status":"MERGED","created":"2020-09-06 15:05:31.000000000","updated":"2020-09-08 20:33:36.000000000","submitted":"2020-09-08 20:31:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"750100-1599597071380-06ebbb73","meta_rev_id":"cd1b994c17458f02b7a3ec421834a785d602a0b7","_number":750100,"virtual_id_number":750100,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-08 20:31:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-09-08 13:25:57.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-09-08 18:20:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2020-09-08 18:20:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-08 13:25:57.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-09-08 18:20:39.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-09-08 20:31:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1feda4044aadf5a4de12b84889165ba226bcacb9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-09-06 15:05:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b509c48e3e75ac57c54585e1ca2ce34b97206dca","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-09-06 15:07:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"80b2a087925ce7d137c3fa35d650d383b74d5f68","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-09-06 15:09:46.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cb48fa0be69941de729b1b296401e7f93d347c3a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-09-06 15:18:01.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c1c3cc4d230557943fae9d2bc2202ff0dec29b0d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-09-06 16:34:40.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"94c825d0df9aa9a512bc70ff02c8d2b63297e3dc","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-09-06 16:38:43.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"3e1ff1dcc6976fcfbbb1394cbfe83b89c0b05908","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-09-06 16:40:59.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"d9688a59542460fc00bee2b87ce2c562199a9d56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-06 18:54:07.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/74a02b8cf542469d9208c52708460931 : SUCCESS in 3m 04s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a90cc64b323649e28cfe1870a694202d : SUCCESS in 3m 39s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e152e2b6c6bc40fd8564dcdbd1551ef4 : SUCCESS in 2m 54s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/53f3663972324b288792bf8cc57f70f0 : SUCCESS in 16m 22s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/e651165511b04d2da4914f44fdee290d : SUCCESS in 16m 12s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/819a3ccb825d44809f7d8d89969775bf : SUCCESS in 13m 50s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ab4c55b4440c489ab7bce4e6859a5689 : FAILURE in 19m 07s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/d53748cf62a943719389d7af12b1775c : SUCCESS in 52m 17s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/15c592d009c541048781466ec0403d5b : FAILURE in 50m 58s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9ce285ab67fe4372be4d962aef653186 : SUCCESS in 31m 29s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/b2a4eb510c27416c88d73e767811bbba : SUCCESS in 38m 44s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8f7378d4f7134fcd9934c4d3c29c6138 : SUCCESS in 1h 28m 38s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/621ac523b969420c9555182a8fe02da4 : SUCCESS in 1h 37m 54s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8b6e8635d5754aa5af075ae11c1b7b70 : SUCCESS in 1h 26m 04s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/77bc93e053ec446682d037dc4463aef8 : SUCCESS in 1h 08m 59s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/9377e8c2d2da4664abbad99eefd6d9f4 : FAILURE in 19m 32s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3e40fcf53be74702a86025b812e08b65 : FAILURE in 23m 00s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/16765053e4764d0c9fccdef174e7d56f : SUCCESS in 3m 31s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d10f53441bca485fa717db9e4d493d9b : SUCCESS in 2h 10m 24s","accounts_in_message":[],"_revision_number":6},{"id":"dd3676b86bb63cc24e013ddd0270c7f34ff0ce59","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-09-06 23:05:42.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"5774b8e442ad607f00e92ce379b7a718600cd97b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 01:38:38.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/347f387c35b844eda8c402471bdc3db6 : SUCCESS in 2m 49s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/19ef6a79931144e292cd254354edd03b : SUCCESS in 2m 19s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3768442f0c56470192b437238d57ba13 : SUCCESS in 2m 37s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/41d834427c404b2380e233286742f163 : SUCCESS in 16m 34s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/f07649976406453e83ef4c84f060f203 : SUCCESS in 15m 53s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/465b3f1bec1949cf94f6c84dc28ffd78 : SUCCESS in 15m 16s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6c6467060b5f428ea9ddb5bff579d1d6 : FAILURE in 21m 21s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/51739bbedab44747bcde64a29683b80c : SUCCESS in 48m 18s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/51937f5eb2994fb488465c3cae85f3d1 : FAILURE in 52m 29s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/73b93051d7124d7e9d00e3112e801fe0 : SUCCESS in 39m 31s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/2d1ef66f56bf4010ac5da85d5baf4b55 : SUCCESS in 38m 21s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/55c52a2375934a4eab1d83fdff77b11b : SUCCESS in 1h 21m 23s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7fbc7136b62c47bfa3a3a8b92f0c08db : SUCCESS in 1h 40m 52s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4e33a6d81afa45738e2f49acc5077b37 : SUCCESS in 1h 29m 13s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/109dff2100084e28b9548c75219e194a : SUCCESS in 56m 28s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/fdbead1ce037488bb8a6986511d0d2f9 : SUCCESS in 19m 11s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9b0d05d321404da8b2e98775d9e5acfd : SUCCESS in 25m 32s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/126ac542feb040579df9b9e5728084ef : SUCCESS in 2m 45s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7465d42ad09d40199025b6fe66a85a67 : SUCCESS in 2h 25m 48s","accounts_in_message":[],"_revision_number":7},{"id":"da9fc52f86116b01e716557545d2a55ce329605f","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-09-08 13:25:57.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"93cff0085d99a8a91014e27c15175d50b7887235","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-09-08 18:20:39.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"a808789d199a42e423b3eb91c18fbbbf1444f760","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 18:23:05.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"1c38e3864aef2e048e68181ef87cc8f4f93dad5e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 20:31:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"71f30060b655883e8d341b2378159314c7e96762","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 20:31:11.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/61cc8223783d4211b9c856d4ba342002 : SUCCESS in 3m 26s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/737a629f55034a8a9368530e9808fc1b : SUCCESS in 3m 43s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/9e59aa7397484bc291e4f62e00f88c4b : SUCCESS in 18m 48s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7e96ed0e499c4abb82a298993a98488e : SUCCESS in 11m 27s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/065920b92a0243ad9663ff15e612a08b : SUCCESS in 1h 13m 13s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/963e7cc837cd4706a11ec99fa8cc533b : SUCCESS in 1h 10m 06s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/2f6ed8ebe18c4fcc8ab440a787484177 : SUCCESS in 1h 06m 01s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c891e5cfcb7848baa1ca75642f966dd5 : SUCCESS in 57m 42s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/7de18281acce404289ecd2b85873139d : SUCCESS in 21m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a9c856bd7102416393700983373f922a : SUCCESS in 3m 45s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d0961f5284ee491282828cf290d4e75e : SUCCESS in 1h 33m 36s","accounts_in_message":[],"_revision_number":7},{"id":"f71e3c7020403636553b86f451bbade871a4d441","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 20:33:36.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd7c278ee08d4903a938a7daee1199ec : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"2daa0b16444f2a2a865b618cbca76fd18c6cd7c6","revisions":{"82f9f3b8bc05c95d3b83385d369c301a2d752ae9":{"kind":"REWORK","_number":1,"created":"2020-09-06 15:05:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/750100/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/00/750100/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/00/750100/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/00/750100/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/00/750100/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/00/750100/1"}}},"commit":{"parents":[{"commit":"e656d11e783784c112574465d8e7888090160221","subject":"Merge \"Add mysql_enable_ndb option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/e656d11e783784c112574465d8e7888090160221"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-06 15:01:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-06 15:02:27.000000000","tz":540},"subject":"Deprecate neutron_api_config","message":"Deprecate neutron_api_config\n\nCurrently puppet-neutron provides 2 resource types, neutorn_api_config\nand neutron_api_paste_ini, to manage the same file.\n(/etc/neutron/api-paste.ini)\nThis patch deprecates neutron_api_config so that we can remove this\nduplicate resource type in a future release.\n\nChange-Id: I4361ea53bd964a3a4861c4c28abb7063e19ebd94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/82f9f3b8bc05c95d3b83385d369c301a2d752ae9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/82f9f3b8bc05c95d3b83385d369c301a2d752ae9"}]},"branch":"refs/heads/master"},"dd4f1b20d46d854592b1fcd0113d32d782f8c457":{"kind":"REWORK","_number":2,"created":"2020-09-06 15:07:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/750100/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/00/750100/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/00/750100/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/00/750100/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/00/750100/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/00/750100/2"}}},"commit":{"parents":[{"commit":"e656d11e783784c112574465d8e7888090160221","subject":"Merge \"Add mysql_enable_ndb option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/e656d11e783784c112574465d8e7888090160221"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-06 15:01:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-06 15:06:57.000000000","tz":540},"subject":"Deprecate neutron_api_config","message":"Deprecate neutron_api_config\n\nCurrently puppet-neutron provides 2 resource types, neutorn_api_config\nand neutron_api_paste_ini, to manage the same file.\n(/etc/neutron/api-paste.ini)\nThis patch deprecates neutron_api_config so that we can remove this\nduplicate resource type in a future release.\n\nChange-Id: I4361ea53bd964a3a4861c4c28abb7063e19ebd94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/dd4f1b20d46d854592b1fcd0113d32d782f8c457"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/dd4f1b20d46d854592b1fcd0113d32d782f8c457"}]},"branch":"refs/heads/master"},"5ba32fb3063870fc67b4f0e60041bcf6d3bdf4cf":{"kind":"REWORK","_number":3,"created":"2020-09-06 15:09:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/750100/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/00/750100/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/00/750100/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/00/750100/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/00/750100/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/00/750100/3"}}},"commit":{"parents":[{"commit":"e656d11e783784c112574465d8e7888090160221","subject":"Merge \"Add mysql_enable_ndb option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/e656d11e783784c112574465d8e7888090160221"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-06 15:01:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-06 15:09:25.000000000","tz":540},"subject":"Deprecate neutron_api_config","message":"Deprecate neutron_api_config\n\nCurrently puppet-neutron provides 2 resource types, neutorn_api_config\nand neutron_api_paste_ini, to manage the same file.\n(/etc/neutron/api-paste.ini)\nThis patch deprecates neutron_api_config so that we can remove this\nduplicate resource type in a future release.\n\nChange-Id: I4361ea53bd964a3a4861c4c28abb7063e19ebd94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/5ba32fb3063870fc67b4f0e60041bcf6d3bdf4cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/5ba32fb3063870fc67b4f0e60041bcf6d3bdf4cf"}]},"branch":"refs/heads/master"},"d377c9886e3b69277c6bde77f954250220bfecba":{"kind":"REWORK","_number":4,"created":"2020-09-06 15:18:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/750100/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/00/750100/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/00/750100/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/00/750100/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/00/750100/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/00/750100/4"}}},"commit":{"parents":[{"commit":"e656d11e783784c112574465d8e7888090160221","subject":"Merge \"Add mysql_enable_ndb option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/e656d11e783784c112574465d8e7888090160221"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-06 15:01:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-06 15:14:46.000000000","tz":540},"subject":"Deprecate neutron_api_config","message":"Deprecate neutron_api_config\n\nCurrently puppet-neutron provides 2 resource types, neutorn_api_config\nand neutron_api_paste_ini, to manage the same file.\n(/etc/neutron/api-paste.ini)\nThis patch deprecates neutron_api_config so that we can solve this\nduplication in a future release.\n\nNOTE:\nThis patch also removes the implementation to remove authtoken\nparameters from api-paste.ini, because that implementation was added\nto deal with parameter migration from api-paste.ini to neutron.conf,\nwhich happened a long ago.\n\nChange-Id: I4361ea53bd964a3a4861c4c28abb7063e19ebd94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/d377c9886e3b69277c6bde77f954250220bfecba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/d377c9886e3b69277c6bde77f954250220bfecba"}]},"branch":"refs/heads/master"},"95b7f9cc99efdb024f2d4fadbd20ddfcbb460e33":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2020-09-06 16:38:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/750100/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/00/750100/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/00/750100/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/00/750100/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/00/750100/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/00/750100/5"}}},"commit":{"parents":[{"commit":"e656d11e783784c112574465d8e7888090160221","subject":"Merge \"Add mysql_enable_ndb option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/e656d11e783784c112574465d8e7888090160221"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-06 15:01:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-06 16:37:03.000000000","tz":540},"subject":"Deprecate neutron_api_config","message":"Deprecate neutron_api_config\n\nCurrently puppet-neutron provides 2 resource types, neutorn_api_config\nand neutron_api_paste_ini, to manage the same file.\n(/etc/neutron/api-paste.ini)\nThis patch deprecates neutron_api_config so that we can solve this\nduplication in a future release. Historically neutron_api_config was\nadded before neutron_api_paste_ini, but in puppet-openstack modules\nwe use *_conf type to manage *.conf  file in general, so its module\nname is confusing.\n\nThis patch also removes the implementation to clean up authtoken\nparameters in api-paste.ini, because that implementation was added\nto deal with parameter migration from api-paste.ini to neutron.conf,\nwhich happened a long ago.\n\nChange-Id: I4361ea53bd964a3a4861c4c28abb7063e19ebd94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/95b7f9cc99efdb024f2d4fadbd20ddfcbb460e33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/95b7f9cc99efdb024f2d4fadbd20ddfcbb460e33"}]},"branch":"refs/heads/master"},"604430f3f4be9a70dbe78e62d2e4c717fecd7184":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2020-09-06 16:40:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/750100/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/00/750100/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/00/750100/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/00/750100/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/00/750100/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/00/750100/6"}}},"commit":{"parents":[{"commit":"e656d11e783784c112574465d8e7888090160221","subject":"Merge \"Add mysql_enable_ndb option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/e656d11e783784c112574465d8e7888090160221"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-06 15:01:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-06 16:40:36.000000000","tz":540},"subject":"Deprecate neutron_api_config","message":"Deprecate neutron_api_config\n\nCurrently puppet-neutron provides 2 resource types, neutorn_api_config\nand neutron_api_paste_ini, to manage the same file.\n(/etc/neutron/api-paste.ini)\nThis patch deprecates neutron_api_config so that we can solve this\nduplication in a future release. Historically neutron_api_config was\nadded before neutron_api_paste_ini, but in puppet-openstack modules\nwe use *_conf type to manage *.conf  file in general, so its name is\nconfusing.\n\nThis patch also removes the implementation to clean up authtoken\nparameters in api-paste.ini, because that implementation was added\nto deal with parameter migration from api-paste.ini to neutron.conf,\nwhich happened a long ago.\n\nChange-Id: I4361ea53bd964a3a4861c4c28abb7063e19ebd94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/604430f3f4be9a70dbe78e62d2e4c717fecd7184"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/604430f3f4be9a70dbe78e62d2e4c717fecd7184"}]},"branch":"refs/heads/master"},"2daa0b16444f2a2a865b618cbca76fd18c6cd7c6":{"kind":"REWORK","_number":7,"created":"2020-09-06 23:05:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/750100/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/00/750100/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/00/750100/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/00/750100/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/00/750100/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/00/750100/7"}}},"commit":{"parents":[{"commit":"e656d11e783784c112574465d8e7888090160221","subject":"Merge \"Add mysql_enable_ndb option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/e656d11e783784c112574465d8e7888090160221"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-06 15:01:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-06 23:05:26.000000000","tz":540},"subject":"Deprecate neutron_api_config","message":"Deprecate neutron_api_config\n\nCurrently puppet-neutron provides 2 resource types, neutorn_api_config\nand neutron_api_paste_ini, to manage the same file.\n(/etc/neutron/api-paste.ini)\nThis patch deprecates neutron_api_config so that we can solve this\nduplication in a future release. Historically neutron_api_config was\nadded before neutron_api_paste_ini, but in puppet-openstack modules\nwe use *_conf type to manage *.conf  file in general, so its name is\nconfusing.\n\nThis patch also removes the implementation to clean up authtoken\nparameters in api-paste.ini, because that implementation was added\nto deal with parameter migration from api-paste.ini to neutron.conf,\nwhich happened a long ago.\n\nChange-Id: I4361ea53bd964a3a4861c4c28abb7063e19ebd94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/2daa0b16444f2a2a865b618cbca76fd18c6cd7c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/2daa0b16444f2a2a865b618cbca76fd18c6cd7c6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
