)]}'
{"id":"openstack%2Fpuppet-neutron~754725","triplet_id":"openstack%2Fpuppet-neutron~stable%2Fstein~I883d9573c6285d4a2480d68c6f3a56299f4c4108","project":"openstack/puppet-neutron","branch":"stable/stein","hashtags":[],"change_id":"I883d9573c6285d4a2480d68c6f3a56299f4c4108","subject":"Revert \"[OVN] Change the default metadata_workers to 2\"","status":"ABANDONED","created":"2020-09-28 14:48:10.000000000","updated":"2020-10-01 19:02:09.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"918e3945f10013bc36ebdc787122db9492626c4e","_number":754725,"virtual_id_number":754725,"owner":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-09-28 17:39:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"date":"2020-09-28 14:48:10.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"date":"2020-09-28 14:48:10.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"date":"2020-09-28 14:48:10.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"date":"2020-09-28 14:48:10.000000000","_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-28 14:48:10.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-09-28 14:48:10.000000000","updated_by":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"reviewer":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"state":"REVIEWER"},{"updated":"2020-09-28 14:48:10.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2020-09-28 14:48:10.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-09-28 17:39:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"22206f5017051e0029a271e19ce87f6676ed092a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-09-28 14:48:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"012c0762bebae72a94915657b6895e03a4488ae0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 17:39:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6b986500d6b4492889a07a159a52517e : SUCCESS in 5m 12s\n- puppet-openstack-syntax-4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0567043d5877403f98c0dded23ce3979 : SUCCESS in 4m 29s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ee45938a75264b52a80805851300c993 : SUCCESS in 4m 10s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/59faab76e527480c834b8682060b5649 : SUCCESS in 5m 09s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/6554c08336ed4269853fb0ddeee17297 : SUCCESS in 22m 57s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/a8bc29248bab4f9687881e7b73af81aa : SUCCESS in 22m 41s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5ba6766118b445bf9ddabb2acb90d672 : SUCCESS in 20m 13s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d6e6671feecd4146a3d92d54e0ad5948 : FAILURE in 32m 21s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/fc5de8b150ea480aa36336367881fcbe : SUCCESS in 52m 51s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/adc5ee94f515422fb450fbb7c129e819 : SUCCESS in 51m 37s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/811a9d6c68dc40428b22bb9e49dc911b : SUCCESS in 56m 52s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/d5471b92aab341a19d225c355c27c98c : SUCCESS in 46m 40s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/17338307104a46ac9923cd1b1b05c7de : FAILURE in 28m 14s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4cd7501e591a40ff8c798f1e8d95eaa2 : FAILURE in 24m 54s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4d91bf85f4314d7a83c00a9f5ca5733f : FAILURE in 24m 01s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/8ff24e1bd1824e17b9860af67aa3fa7c : FAILURE in 20m 56s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/63552c1153be41919d3fbaf795c4d77b : SUCCESS in 24m 43s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e9bcd515bcf746cf8f84ef85a04dae7b : FAILURE in 6m 46s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce31fa8d5b21449f95c7364f0e0e3c4c : SUCCESS in 3m 23s\n- puppet-neutron-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7d42b81f11654c3b9aacec0ff98c0901 : SUCCESS in 2h 35m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"918e3945f10013bc36ebdc787122db9492626c4e","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-10-01 19:02:09.000000000","message":"Abandoned\n\ndoesn\u0027t look related necessarily","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"65c9c805191e3067c92b297ea8b79a4b6c1fc445","revisions":{"65c9c805191e3067c92b297ea8b79a4b6c1fc445":{"kind":"REWORK","_number":1,"created":"2020-09-28 14:48:10.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/25/754725/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/25/754725/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/25/754725/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/25/754725/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/25/754725/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/25/754725/1"}}},"commit":{"parents":[{"commit":"080fd53ec4d013c8147578322c676568970cc4bc","subject":"[OVN] Change the default metadata_workers to 2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/080fd53ec4d013c8147578322c676568970cc4bc"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2020-09-28 14:48:10.000000000","tz":0},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2020-09-28 14:48:10.000000000","tz":0},"subject":"Revert \"[OVN] Change the default metadata_workers to 2\"","message":"Revert \"[OVN] Change the default metadata_workers to 2\"\n\nThis reverts commit 080fd53ec4d013c8147578322c676568970cc4bc.\n\nTesting some upgrade issues.\n\nChange-Id: I883d9573c6285d4a2480d68c6f3a56299f4c4108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/65c9c805191e3067c92b297ea8b79a4b6c1fc445"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/65c9c805191e3067c92b297ea8b79a4b6c1fc445"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
