)]}'
{"id":"openstack%2Fpuppet-neutron~756246","triplet_id":"openstack%2Fpuppet-neutron~stable%2Ftrain~Iff50d2d086c687ef9ec1ba878e24c06dfc8a7d99","project":"openstack/puppet-neutron","branch":"stable/train","hashtags":[],"change_id":"Iff50d2d086c687ef9ec1ba878e24c06dfc8a7d99","subject":"Add agent/explicitly_egress_direct support","status":"MERGED","created":"2020-10-06 10:37:49.000000000","updated":"2020-10-23 16:14:55.000000000","submitted":"2020-10-23 16:13:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756246-1603469604457-12e0dd0d","meta_rev_id":"e4941073f473b2c88b0ce45d3e9fee20a2aa60be","_number":756246,"virtual_id_number":756246,"owner":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-10-21 12:55:12.000000000","_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-23 16:13:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-10-23 15:00:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-10-22 14:15:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2020-10-23 15:00:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-22 14:15:45.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-10-23 15:00:24.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-10-23 16:13:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aa077a0ab5c85d6591c583eef8ba35a41c69cc20","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-10-06 10:37:49.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"7d825b7672f9d12b8a3cc368d84f866c19cc9b56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 12:01:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7131c9884ec44bb98389f133ad30e5f3 : FAILURE in 4m 56s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8297cfbd834a48beae65e5a92c03f759 : SUCCESS in 4m 37s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/68125cba740e4d3e94d21e05bcefd5e0 : SUCCESS in 4m 10s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/4ca9d28724c740c6876172e9e5c5e725 : SUCCESS in 51m 02s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/89b1d51e067d4195a658d0f32ac43e5a : SUCCESS in 27m 24s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/db72af5af53c4863be4c6ea0857494b9 : SUCCESS in 20m 06s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9baf0dd317314e68aa7f281f607da71c : SUCCESS in 18m 58s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9d1b8245d2ad4fe4bc537d79ccf7c834 : FAILURE in 30m 01s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/ed906e2c3b074d1882ed39fb20ef95a0 : SUCCESS in 56m 26s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/21c93d7645c74bcdb89f0ebe48e24fa1 : SUCCESS in 1h 01m 26s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/72e6688937e34a888b42bc65f5b48526 : SUCCESS in 1h 00m 49s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/4b7776fff23c4252a0730997bb35dece : SUCCESS in 45m 52s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/1728e32336374cdeac165448f360c921 : SUCCESS in 48m 34s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/941291b429a2444eb1b4a8d6a9bc6442 : SUCCESS in 1h 03m 23s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/356d76fd962a42619f07eca968e3756a : SUCCESS in 47m 52s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/d5e506dd6b0f42c38b20d60eacbe6713 : SUCCESS in 39m 11s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/89ad4a1b921c4ff3b26f055f6bba6d95 : FAILURE in 22m 09s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/461169e9a1ec49618f385bb64ed5c4ae : FAILURE in 9m 14s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5429b07a3b4247cfa0e3ed47154e9507 : FAILURE in 9m 25s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3accaf2d127246e58661e3af6a4dc757 : SUCCESS in 46m 06s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/835f4846b60a457e9396d308028d22ef : SUCCESS in 19m 02s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/70809daea7214394ac9af937a6afd62e : FAILURE in 9m 36s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/313a7d7985464080a3e57c98afcc131a : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":1},{"id":"67b2273cc5b96a8d2d023a0897e80ea21abbd117","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-10-21 12:55:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9383ca2c45b2bc66f0368f4190dc0a285ec5794d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 14:12:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d97adffa579245b39acf49856ef493d8 : SUCCESS in 6m 11s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cf2821e238624b46b381ccd0ee0edeea : SUCCESS in 5m 40s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8ebf95f458fc43a39c0c01261465582b : SUCCESS in 5m 57s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/9b12edbf69a44042961011227990b616 : SUCCESS in 24m 59s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/55b48b896f3a46669edceb9fae6aa31c : SUCCESS in 19m 05s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/fc7373821f784688ba6b39f37ec390af : SUCCESS in 21m 02s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b812e02edb64452398a5bd0620a4dc86 : SUCCESS in 17m 27s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ae05877619a34c82b6f1703993416367 : FAILURE in 32m 19s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/06907803d0774bd693646a2b32c0070b : SUCCESS in 48m 33s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/818bee0468fd445f94052fb037f13a54 : SUCCESS in 1h 02m 59s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/92a59c734c6b48bb933c894bd641b916 : SUCCESS in 57m 59s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/83034e55bb6d4cbeacf2f7b15897a5ca : SUCCESS in 46m 23s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/fbb98ef1996840a9907fcd791215511a : SUCCESS in 54m 41s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/be3d6f358553471ca2f869652ed11ff2 : SUCCESS in 47m 15s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6463355e713948a4959ad674110da052 : SUCCESS in 41m 56s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/1641af12ff5940f998100dc7426990a0 : SUCCESS in 37m 12s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/29eca82e38014553bbc101f46108ba2b : SUCCESS in 51m 30s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3276b9028695440d863f072088d65d09 : SUCCESS in 52m 33s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/179315d4a4314f27a105ee69334f05b9 : SUCCESS in 1h 09m 21s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/797ffa7842fb458e8991e8fa1b3ef64c : SUCCESS in 42m 35s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/c05d86ebdfd34bbbbe3dfcf348027123 : SUCCESS in 21m 18s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ada2ed5727d749a486cbf1a45eb602e1 : FAILURE in 9m 15s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/40ab0ccbc7cf47659b444210525c4a74 : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":1},{"id":"96bf06821bba5604648e88cef4993f10ff7f0184","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-10-22 14:15:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f86cb86ac4f13203318ec29b13bb2894cc2aae39","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-23 15:00:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nClean backport. LGTM.","accounts_in_message":[],"_revision_number":1},{"id":"0e82cc91c3bb3eeac8442a17ed137077c45bf3c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 15:00:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e17a8722d04045249341eeae3b28bca994b37929","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 16:13:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/666697f77e4f40229ba0d643fb65e013 : SUCCESS in 8m 23s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e9479df9694b433dafdde34e0654b599 : SUCCESS in 7m 39s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3b4529ce5c83453f8a2092c72a1a85c2 : SUCCESS in 7m 54s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/b46e843043734c3daa6261a8038956d6 : SUCCESS in 33m 28s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/95094817a8884b3f825134bedee1748e : SUCCESS in 17m 20s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/fc9f13f9e933435f9e43b8419c66abca : SUCCESS in 51m 20s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/8f901b2f99864c8f90631a4b3e446915 : SUCCESS in 1h 02m 17s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/59eb5ee916c34949bfa23f8c2eee9f5b : SUCCESS in 1h 05m 27s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/863e41c18f3940409e26f50d159cd0f1 : SUCCESS in 46m 15s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/da7d2d1fb1a947d0b28619771d404112 : SUCCESS in 26m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fdd9bc3b06d3419e993853ea2a8bed22 : SUCCESS in 7m 14s","accounts_in_message":[],"_revision_number":1},{"id":"fa60be63ce504ca0d8c0a89130abab0dad69d8c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 16:13:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e4941073f473b2c88b0ce45d3e9fee20a2aa60be","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 16:14:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88ad4d0c7dd34f25a72c86c626f73919 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8ea35656957b2009edf084ff6b5abe4d91df868f","revisions":{"8ea35656957b2009edf084ff6b5abe4d91df868f":{"kind":"REWORK","_number":1,"created":"2020-10-06 10:37:49.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/46/756246/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/46/756246/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/46/756246/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/46/756246/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/46/756246/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/46/756246/1"}}},"commit":{"parents":[{"commit":"7d0406b1a2bb054f42b198e9494ddc54372e7285","subject":"Nuage: remove check on mechanism driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/7d0406b1a2bb054f42b198e9494ddc54372e7285"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-09-29 15:40:16.000000000","tz":120},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-10-06 10:37:49.000000000","tz":0},"subject":"Add agent/explicitly_egress_direct support","message":"Add agent/explicitly_egress_direct support\n\nThis configuration parameter for the ML2/OVS agent was added in\nI14fefe289a19b718b247bf0740ca9bc47f8903f4 and backported up to\nstable/queens, preventing flooding on integration bridge in some cases.\n\nIt needs to be enabled on specific node types, so add option to\nconfigure it.\n\nRelated-Bug: #1732067\nChange-Id: Iff50d2d086c687ef9ec1ba878e24c06dfc8a7d99\n(cherry picked from commit d0bc82d8427c43dac07a8118d2fe35ba5a9d9dce)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/8ea35656957b2009edf084ff6b5abe4d91df868f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/8ea35656957b2009edf084ff6b5abe4d91df868f"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
