)]}'
{"id":"openstack%2Fpuppet-neutron~756256","triplet_id":"openstack%2Fpuppet-neutron~stable%2Fqueens~Iff50d2d086c687ef9ec1ba878e24c06dfc8a7d99","project":"openstack/puppet-neutron","branch":"stable/queens","hashtags":[],"change_id":"Iff50d2d086c687ef9ec1ba878e24c06dfc8a7d99","subject":"Add agent/explicitly_egress_direct support","status":"MERGED","created":"2020-10-06 10:49:30.000000000","updated":"2021-03-12 16:14:29.000000000","submitted":"2021-03-12 16:12:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756256","meta_rev_id":"4fb11e4aa0de9044067c0ca5a147f4af855ba585","_number":756256,"virtual_id_number":756256,"owner":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-12 16:12:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2021-03-10 12:35:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-03-12 15:12:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2021-03-12 15:12:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-06 12:02:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-30 15:08:07.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2021-03-10 12:35:10.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-03-12 15:12:42.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"2e0b193ae040005dc22379dacd04c2860e1ea056","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-10-06 10:49:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b21038ac49b51b086b4a54377609fccd610ddb1c","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-10-06 10:52:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3de7db99fb721032b9fb54aa2b0e3fe62a931570","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 12:02:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/6f76188397c248a2bf50ec27141399ee : FAILURE in 5m 02s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/12d538411f644a5c80f338d8a7bb90bf : SUCCESS in 4m 32s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/bc22d231a9f7457cb4373b424953ce9f : SUCCESS in 31m 26s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/80ac4ddf8f19441daddf3788c2739a87 : SUCCESS in 26m 30s\n- puppet-openstack-integration-4-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/9793e0c2a8144daf9bbb4ecaccd35599 : SUCCESS in 54m 16s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/025222699c2a4ffb9873dc20f5d5ab38 : SUCCESS in 59m 48s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/eda36586930e47fab22725e9fa250129 : SUCCESS in 57m 58s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/127521163456421482b7ef7163999dc4 : SUCCESS in 44m 05s\n- puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial-luminous https://zuul.opendev.org/t/openstack/build/e7c437632f8140a48ec098be0dd9763c : FAILURE in 25m 50s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/d4620e6a5ae1464d82f6533a7a19ce01 : FAILURE in 28m 27s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/46931ea83cf04835a01f6bf9590fc6ca : FAILURE in 24m 16s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial-luminous https://zuul.opendev.org/t/openstack/build/012ce9a52dbd45fc948097f16732c1d9 : FAILURE in 22m 29s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/a31f0dca75d04f5ab4dbcf662c85edaa : SUCCESS in 18m 36s\n- puppet-openstack-beaker-xenial https://zuul.opendev.org/t/openstack/build/91221b40888442988506891949073647 : SUCCESS in 12m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/68160f7fdc824a5b86dd7e63cc4aa76d : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":2},{"id":"c4b6a2616b808d39bbc0cd59b5b1599e3f2b27e3","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-10-23 11:13:14.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"42069144d49b90616d51c2edff86f64aaad3fa47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 12:40:04.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/1ddc766503eb43eaaad710b4700b2cc8 : SUCCESS in 6m 43s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/9729cb143d544b9fb353a9b8f0fb61de : SUCCESS in 5m 31s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/8eaa132576b14e4ca7d701d85f96e479 : SUCCESS in 24m 57s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/50142abfa75f4b4484cf6d3f04b85c5f : RETRY_LIMIT in 5m 39s\n- puppet-openstack-integration-4-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/f30763e7af33480bac2e5c17f439720c : SUCCESS in 57m 45s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/a038c2ec9f8b478daf09a6a4c018bb59 : SUCCESS in 59m 08s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/0dac175a19cd44ce8e29ca3286435b40 : FAILURE in 58m 47s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/cacbc0380ff844caa2e781091e70abc4 : SUCCESS in 48m 45s\n- puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial-luminous https://zuul.opendev.org/t/openstack/build/bd9612d431d741fc8c8f0a29e69e5332 : FAILURE in 33m 08s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/5581d293f2a64d91bfc964e2afe1f96e : FAILURE in 27m 06s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/71156a49ddf7419a899e17ee553cb5aa : FAILURE in 27m 26s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial-luminous https://zuul.opendev.org/t/openstack/build/c730b687d95b4e3698954d7e97bcb87c : FAILURE in 23m 15s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/509628a7e7d243c0b6abd9f822b2d53a : SUCCESS in 23m 41s\n- puppet-openstack-beaker-xenial https://zuul.opendev.org/t/openstack/build/116b7c3e4cdb4ef7a74d4fb24a812cfb : SUCCESS in 15m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac477c99a33642b494573ddfcf599887 : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":2},{"id":"2e6741e2e6a4a86bdf8ec31f286eb72be14b4ecf","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-11-30 15:08:07.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ca5b87530b902930d31db79826907ccc776b7950","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 16:10:40.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/d8bc955682f34a4bbc952c241b758624 : SUCCESS in 4m 11s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/93b99be58fcd4d7c97ec7bcbbea67c13 : SUCCESS in 3m 51s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/13d3c40c60db4cfe929aa1a6779498c8 : SUCCESS in 19m 54s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/daaf05e849414c56a13ff1a64263e0d4 : SUCCESS in 22m 12s\n- puppet-openstack-integration-4-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/b8041955b9884be4940a3eadf6f4491a : SUCCESS in 54m 50s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/28e771db23f747bca4beb94d7c46efad : FAILURE in 55m 27s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/cd486ce33402453aa6983a74a0eae8fe : SUCCESS in 55m 50s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/6912b72e6e1a49faa03f3f985b7edc24 : SUCCESS in 44m 50s\n- puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial-luminous https://zuul.opendev.org/t/openstack/build/e9ecc05cf53d4e4e9e493dfb060e82ad : FAILURE in 4m 50s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/a02e5dfbd3ef4400b3208e28679552c8 : FAILURE in 6m 43s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/69be31f646274f32a9fd643a1a13d100 : FAILURE in 5m 03s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial-luminous https://zuul.opendev.org/t/openstack/build/ad2514c7567940a69e56f706c80698de : FAILURE in 5m 00s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/e43b5a0f9e924460bfab7c2fcb555afe : SUCCESS in 18m 08s\n- puppet-openstack-beaker-xenial https://zuul.opendev.org/t/openstack/build/4621b3e1207d4489b4fd6e42145ec652 : FAILURE in 13m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c17219602cd44f3086c8b9f298695dae : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":2},{"id":"1da9fd5e935428fd0f8b25209201edaa88a74f0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-03-10 10:57:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"25953863901338f7ad11c85a2ccb188425800b9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 12:09:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/c905652ca3554bcdaf59b8cf5d110a81 : SUCCESS in 5m 13s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/3570959ab31f4bb98b417daeacaee23a : SUCCESS in 5m 11s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/12f5ee1b01ed4408964059b8f342407a : SUCCESS in 19m 53s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/1a2ff0930b214484b7ec9d4f0a695d5c : SUCCESS in 21m 16s\n- puppet-openstack-integration-4-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/70f63f0c611d4b3da059f5643d4112ac : SUCCESS in 59m 10s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/36b3d48a60d146c5a8852410c63efec1 : SUCCESS in 58m 00s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/6a0e964dbfa44e149be9b68a1511bfe8 : SUCCESS in 1h 05m 04s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/ea73459d6ca241e0817c48a54ea0b17c : SUCCESS in 46m 04s\n- puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial-luminous https://zuul.opendev.org/t/openstack/build/ebc3bfdd8378464f88de9c80cd484dae : FAILURE in 6m 31s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/e7714cc30f74453bbcef9db87e5b5b98 : FAILURE in 5m 51s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/cfba5c3c6eea499cbbd4b231b78eed06 : FAILURE in 5m 18s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial-luminous https://zuul.opendev.org/t/openstack/build/b0fa455355074ba59a68ed02576dc8ee : FAILURE in 6m 26s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/a239c5627df8476a851b1afd4c8ddc48 : SUCCESS in 19m 46s\n- puppet-openstack-beaker-xenial https://zuul.opendev.org/t/openstack/build/910a3a4803f14d6dbafdaa9fa407372a : FAILURE in 16m 09s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cda5c2f805644cc4bb23e79d4ccaec50 : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":3},{"id":"ed924202c06d41e4ac8471aa3d166d2acaad022b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-10 12:32:40.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"1e5b7bb40fe4dbf023b386dff892cc6e4d7fff26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-10 12:34:06.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"c01259712444243c29e43101f86eab27362ac211","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-10 12:35:10.000000000","message":"Patch Set 5: Code-Review+2\n\nI updated the place where we insert these changes to make it consistent with current implementation in Rocky and later.","accounts_in_message":[],"_revision_number":5},{"id":"c6a59675d6e2eadae210fa4aef166f1aecbc3670","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 13:49:13.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/3032741a41db4095970a81da1cde2771 : SUCCESS in 7m 08s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/13f46cdd984c4dbf86cec20a44b7b47c : SUCCESS in 5m 24s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/726f753abfe745098c654d0f6658b4f9 : SUCCESS in 28m 23s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/a7a2787204fe4b1f9046b5e6c7d13d56 : SUCCESS in 25m 01s\n- puppet-openstack-integration-4-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/a836fdd4ffa840699d9eab4eae3b12f8 : SUCCESS in 54m 27s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/d2addf2a33e74c5daa4448449b5b66f4 : SUCCESS in 53m 00s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/042b6dd337e841c79bda18bac503a811 : SUCCESS in 55m 52s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/fafb6921af5e4d038ce3fcaaa4ef51a9 : SUCCESS in 42m 23s\n- puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial-luminous https://zuul.opendev.org/t/openstack/build/07c6424e7d1a41b4999c6082e9a279b5 : FAILURE in 9m 42s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/03ac7577b4164fefa222baea9ae064cc : FAILURE in 8m 02s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/f9ab0b8674e54f80a248dd4feebfbe5d : FAILURE in 10m 07s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial-luminous https://zuul.opendev.org/t/openstack/build/0634951f3ce443f2bcc8473d3d780d24 : FAILURE in 6m 45s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/02a6bd7590b94891882546d2f9b8ede2 : SUCCESS in 15m 35s\n- puppet-openstack-beaker-xenial https://zuul.opendev.org/t/openstack/build/cd00412d1b954749b27dd6ee2131502d : FAILURE in 13m 12s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31d9612b28df404a9676e5ad81d4e63f : SUCCESS in 5m 58s","accounts_in_message":[],"_revision_number":5},{"id":"ba8a8a982b5fb3aefa79d60bd35f40268635ffec","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-03-10 13:58:42.000000000","message":"Patch Set 5:\n\nThanks for the rebase update! I hesitated between both options when fixing the conflict, but yes having the files closer to rocky and newer is better than \"last added option goes at the bottom\" style","accounts_in_message":[],"_revision_number":5},{"id":"d01f50d9567e7a85bdf00371de2a9f7b70b79d08","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-03-12 15:12:42.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"35139b39758bf76ca36b1746fe3d4ba3c8c0c946","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 15:12:56.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"df8855e3fa84806ebda0b7d2d8e6c201f5935094","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 16:12:43.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/6c695a2405814eda8f0e76fc250224f3 : SUCCESS in 4m 57s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/2c294899e1ca4c0da3e84488a75357f8 : SUCCESS in 4m 46s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/63742a1278244b77a12eba2b7d32d860 : SUCCESS in 19m 36s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/d0032d2949454d949c68dbe19eca66a1 : SUCCESS in 18m 01s\n- puppet-openstack-integration-4-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/385e8379418640b0b895ca9b8d782602 : SUCCESS in 53m 24s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/4966dbe5bc6942fba2e7c5e18e8671de : SUCCESS in 53m 25s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/405a6148fa6e484c8136384c120546ce : SUCCESS in 56m 26s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/83808c5766794a9c8263e96caacf4709 : SUCCESS in 41m 55s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/ef699418382445319d38a443de55f9d5 : SUCCESS in 19m 42s\n- puppet-openstack-beaker-xenial https://zuul.opendev.org/t/openstack/build/08ba8550c6dd4fa1a20f07803857a1d9 : FAILURE in 14m 24s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4cfc1f846534b13a1ba43d22e86c056 : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":5},{"id":"5aa44c281d76ae3b2af485504c9ac85c73755364","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 16:12:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"4fb11e4aa0de9044067c0ca5a147f4af855ba585","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 16:14:29.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8baac27a7b2f49c7b49a4e6f35ce908f : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"9e4b885595bf4dfab66f3c601484d0e8b4b5e7d3","revisions":{"9a31ffd7ed92bab95ff4736d3caad27eb7be1595":{"kind":"REWORK","_number":1,"created":"2020-10-06 10:49:30.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/56/756256/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/56/756256/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/56/756256/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/56/756256/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/56/756256/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/56/756256/1"}}},"commit":{"parents":[{"commit":"3987a1b53907543e9c730b5e62da2e498cc9065c","subject":"[OVN] Change the default metadata_workers to 2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/3987a1b53907543e9c730b5e62da2e498cc9065c"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-09-29 15:40:16.000000000","tz":120},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-10-06 10:49:18.000000000","tz":120},"subject":"Add agent/explicitly_egress_direct support","message":"Add agent/explicitly_egress_direct support\n\nThis configuration parameter for the ML2/OVS agent was added in\nI14fefe289a19b718b247bf0740ca9bc47f8903f4 and backported up to\nstable/queens, preventing flooding on integration bridge in some cases.\n\nIt needs to be enabled on specific node types, so add option to\nconfigure it.\n\nConflicts:\n\tmanifests/agents/ml2/ovs.pp\n\tspec/classes/neutron_agents_ml2_ovs_spec.rb\n\nRelated-Bug: #1732067\nChange-Id: Iff50d2d086c687ef9ec1ba878e24c06dfc8a7d99\n(cherry picked from commit d0bc82d8427c43dac07a8118d2fe35ba5a9d9dce)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/9a31ffd7ed92bab95ff4736d3caad27eb7be1595"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/9a31ffd7ed92bab95ff4736d3caad27eb7be1595"}]},"branch":"refs/heads/stable/queens"},"63061d2a4c700a10b5f741b96bb528221bdf94b8":{"kind":"REWORK","_number":2,"created":"2020-10-06 10:52:20.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/56/756256/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/56/756256/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/56/756256/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/56/756256/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/56/756256/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/56/756256/2"}}},"commit":{"parents":[{"commit":"3987a1b53907543e9c730b5e62da2e498cc9065c","subject":"[OVN] Change the default metadata_workers to 2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/3987a1b53907543e9c730b5e62da2e498cc9065c"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-09-29 15:40:16.000000000","tz":120},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-10-06 10:52:14.000000000","tz":120},"subject":"Add agent/explicitly_egress_direct support","message":"Add agent/explicitly_egress_direct support\n\nThis configuration parameter for the ML2/OVS agent was added in\nI14fefe289a19b718b247bf0740ca9bc47f8903f4 and backported up to\nstable/queens, preventing flooding on integration bridge in some cases.\n\nIt needs to be enabled on specific node types, so add option to\nconfigure it.\n\nConflicts:\n\tmanifests/agents/ml2/ovs.pp\n\tspec/classes/neutron_agents_ml2_ovs_spec.rb\n\nRelated-Bug: #1732067\nChange-Id: Iff50d2d086c687ef9ec1ba878e24c06dfc8a7d99\n(cherry picked from commit d0bc82d8427c43dac07a8118d2fe35ba5a9d9dce)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/63061d2a4c700a10b5f741b96bb528221bdf94b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/63061d2a4c700a10b5f741b96bb528221bdf94b8"}]},"branch":"refs/heads/stable/queens"},"e3a21dcbff889951f368de1a7748b69c4f6a89b4":{"kind":"REWORK","_number":3,"created":"2021-03-10 10:57:01.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/56/756256/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/56/756256/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/56/756256/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/56/756256/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/56/756256/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/56/756256/3"}}},"commit":{"parents":[{"commit":"b34137a93ea046b50bae76e10fd00a52c6f74256","subject":"Add ovs/igmp_snooping_enable support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b34137a93ea046b50bae76e10fd00a52c6f74256"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-09-29 15:40:16.000000000","tz":120},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2021-03-10 10:55:15.000000000","tz":60},"subject":"Add agent/explicitly_egress_direct support","message":"Add agent/explicitly_egress_direct support\n\nThis configuration parameter for the ML2/OVS agent was added in\nI14fefe289a19b718b247bf0740ca9bc47f8903f4 and backported up to\nstable/queens, preventing flooding on integration bridge in some cases.\n\nIt needs to be enabled on specific node types, so add option to\nconfigure it.\n\nConflicts:\n\tmanifests/agents/ml2/ovs.pp\n\tspec/classes/neutron_agents_ml2_ovs_spec.rb\n\nRelated-Bug: #1732067\nChange-Id: Iff50d2d086c687ef9ec1ba878e24c06dfc8a7d99\n(cherry picked from commit d0bc82d8427c43dac07a8118d2fe35ba5a9d9dce)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/e3a21dcbff889951f368de1a7748b69c4f6a89b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/e3a21dcbff889951f368de1a7748b69c4f6a89b4"}]},"branch":"refs/heads/stable/queens"},"d8246fa4c329b1e8075c3afa44eb9195c611cab0":{"kind":"REWORK","_number":4,"created":"2021-03-10 12:32:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/56/756256/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/56/756256/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/56/756256/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/56/756256/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/56/756256/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/56/756256/4"}}},"commit":{"parents":[{"commit":"b34137a93ea046b50bae76e10fd00a52c6f74256","subject":"Add ovs/igmp_snooping_enable support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b34137a93ea046b50bae76e10fd00a52c6f74256"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-09-29 15:40:16.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-10 12:32:10.000000000","tz":0},"subject":"Add agent/explicitly_egress_direct support","message":"Add agent/explicitly_egress_direct support\n\nThis configuration parameter for the ML2/OVS agent was added in\nI14fefe289a19b718b247bf0740ca9bc47f8903f4 and backported up to\nstable/queens, preventing flooding on integration bridge in some cases.\n\nIt needs to be enabled on specific node types, so add option to\nconfigure it.\n\nConflicts:\n\tmanifests/agents/ml2/ovs.pp\n\tspec/classes/neutron_agents_ml2_ovs_spec.rb\n\nRelated-Bug: #1732067\nChange-Id: Iff50d2d086c687ef9ec1ba878e24c06dfc8a7d99\n(cherry picked from commit d0bc82d8427c43dac07a8118d2fe35ba5a9d9dce)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/d8246fa4c329b1e8075c3afa44eb9195c611cab0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/d8246fa4c329b1e8075c3afa44eb9195c611cab0"}]},"branch":"refs/heads/stable/queens"},"9e4b885595bf4dfab66f3c601484d0e8b4b5e7d3":{"kind":"REWORK","_number":5,"created":"2021-03-10 12:34:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/56/756256/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-neutron","ref":"refs/changes/56/756256/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/56/756256/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/56/756256/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-neutron refs/changes/56/756256/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-neutron refs/changes/56/756256/5"}}},"commit":{"parents":[{"commit":"b34137a93ea046b50bae76e10fd00a52c6f74256","subject":"Add ovs/igmp_snooping_enable support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/b34137a93ea046b50bae76e10fd00a52c6f74256"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-09-29 15:40:16.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-10 12:34:01.000000000","tz":0},"subject":"Add agent/explicitly_egress_direct support","message":"Add agent/explicitly_egress_direct support\n\nThis configuration parameter for the ML2/OVS agent was added in\nI14fefe289a19b718b247bf0740ca9bc47f8903f4 and backported up to\nstable/queens, preventing flooding on integration bridge in some cases.\n\nIt needs to be enabled on specific node types, so add option to\nconfigure it.\n\nConflicts:\n\tmanifests/agents/ml2/ovs.pp\n\tspec/classes/neutron_agents_ml2_ovs_spec.rb\n\nRelated-Bug: #1732067\nChange-Id: Iff50d2d086c687ef9ec1ba878e24c06dfc8a7d99\n(cherry picked from commit d0bc82d8427c43dac07a8118d2fe35ba5a9d9dce)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/9e4b885595bf4dfab66f3c601484d0e8b4b5e7d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-neutron/commit/9e4b885595bf4dfab66f3c601484d0e8b4b5e7d3"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
